)]}'
{"id":"openstack%2Frequirements~698681","triplet_id":"openstack%2Frequirements~master~I4b4277cb649ba1399cb30019fbcab9c9c3574fcd","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4b4277cb649ba1399cb30019fbcab9c9c3574fcd","subject":"update constraint for kuryr-lib to new release 2.0.0","status":"MERGED","created":"2019-12-12 10:17:02.000000000","updated":"2019-12-12 17:13:54.000000000","submitted":"2019-12-12 17:09:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698681-1576170576316-91269b4c","meta_rev_id":"fa4e10de9f282fe4206beaf81113949845971a87","_number":698681,"virtual_id_number":698681,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 17:09:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-12 13:53:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-12 13:53:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 13:53:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-12 17:09:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"802cb7b5e6d6fc414c42028c52a7ae271ff124b1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-12 10:17:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebe89e7a4965fa3f5ab17b7b948809603003947f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 12:19:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d8947e8293e4012961ea51db274b122 : SUCCESS in 4m 23s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5dacaaac92cf46efa77ce17ac985c703 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b06915b01b24ce0860b343beb4a9e49 : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8fd7fe9204a04cc88b9ab78e634fd964 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6da93deee8274e98980baa0a522f2d4d : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2007c2dabdc448f1ac1bbd601bac0393 : SUCCESS in 6m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c64ffca34a6423c9651bf117696ac0c : SUCCESS in 6m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eb6633c5cc5a43e99f1122f0c1b6ea2a : SUCCESS in 7m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2ac9e806b7654693a1f7b21efa9556dc : SUCCESS in 8m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ad4028dbc4bf4b54966c82ee3cb8ce1b : SUCCESS in 12m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ca2d1720be38403a806433bb8f54ea65 : SUCCESS in 10m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/888c9f5e9267454ba38d6e1e6a3d1da5 : SUCCESS in 6m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/15faeaf7423c491585acb76e3b27f3b3 : SUCCESS in 9m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e901be09774544e0a66eaa81c9ee4a19 : SUCCESS in 8m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ef331eccedb04d5bbfb5726dc9072978 : SUCCESS in 27m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/99c1567fbe8f4c568863367c1bb54885 : SUCCESS in 24m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bb4cc1fbdaa24c95ab98660a1c2c767a : SUCCESS in 14m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/df8da07827994e6f9178ff790e20810b : SUCCESS in 14m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/880c288a47d8406183968ffe95822711 : SUCCESS in 19m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/57404c10f2474587aae2b951a87c510d : SUCCESS in 6m 49s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/599e6620d344482794e1891dab4c4b9f : SUCCESS in 8m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c32371fc8a7445082c346a533a39fc7 : SUCCESS in 8m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ce8d280c3ece4df4a649f628821a4866 : SUCCESS in 14m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/df330cefa3d64568a124baa8e051ec57 : SUCCESS in 13m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/284d717180d94b9792a13fa8c66887f7 : SUCCESS in 1h 17m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4119cfa3bb584278ad2e2712c2d97f51 : SUCCESS in 1h 52m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b2e08afd85b04d57be9d2586fe51fc1d : SUCCESS in 1h 03m 45s","accounts_in_message":[],"_revision_number":1},{"id":"2a5361232e5dcdfb82d0564fe903a36c4d81e804","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-12 13:53:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b89715cb0aac4cc9cb0201e092f432d44fc6f19a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 13:53:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"82f6428ff1b947cfc3b59e243ee422791179c006","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 17:09:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f38e6489ebf349e5a99dfa4416a7d2c3 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d52abd278cee45f7b49cc8254f0fd94e : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5950a01cafc34d24bba12de6f8bf9993 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eca12d0069c542a3b1a0a7315512e272 : SUCCESS in 7m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/534bfbfa8c5b42fb971bb128fe179234 : SUCCESS in 6m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9eaf6c80428471a863cc92d47bfeb12 : SUCCESS in 6m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1fbb4a0d69b14f2cb6d5acf1bacb3a2f : SUCCESS in 5m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c3b92b6d08aa411d86d7a9d6ccd9a911 : SUCCESS in 11m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e9f26ba955834910b0f5290336e99b4a : SUCCESS in 8m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ea51a95e2d324d85ae7177d9447708b5 : SUCCESS in 12m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/470766a3c10941adacd51d14f8ac53c3 : SUCCESS in 7m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7370c39e40da4445b71e285560e8398f : SUCCESS in 12m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/87970ac688d54ba990a1976c16105a2f : SUCCESS in 8m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1b7f5cb2d43148d5a174b67c0cc685f9 : SUCCESS in 21m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/36d587c894a446cdac53ce00e1ee4a6f : SUCCESS in 22m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e9ee31a879df4924916c07ff74e4e52a : SUCCESS in 17m 47s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/47f8bdaf80654b7f9151206413786e2e : SUCCESS in 15m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3be8635bdc1f43968a6a3510eb2f236c : SUCCESS in 20m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8a48c897a7ef439c84163d972da490e2 : SUCCESS in 6m 41s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/75c839ec13eb4a6a9c08fc694e7ea4ed : SUCCESS in 6m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/101183b700d14b489fd3bb31478bd3d0 : SUCCESS in 9m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/25402c2d7d794e71908ec55f06bda1fe : SUCCESS in 11m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e5d1adfa3619414587c55ef508b31e4d : SUCCESS in 10m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/35e448c5d3d041dabeb4492b513e76c6 : SUCCESS in 1h 12m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fdd7cc77ce454a9cb92f68eef4685e23 : SUCCESS in 1h 32m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5a57a6a1fd304b279d01a57b295e43e2 : SUCCESS in 57m 28s","accounts_in_message":[],"_revision_number":1},{"id":"0e561128af5592d21175dfbf103734fcab54af50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 17:09:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3d60e144f84c60ee0d329c9cc06faea53455c368","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 17:13:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/322fb818f60a4131add2ad6e822cd7b4 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"49a481e60f34f8cf08ee8a0c4aed0e4f424ec210","revisions":{"49a481e60f34f8cf08ee8a0c4aed0e4f424ec210":{"kind":"REWORK","_number":1,"created":"2019-12-12 10:17:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/81/698681/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/698681/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/698681/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/698681/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/698681/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/698681/1"}}},"commit":{"parents":[{"commit":"0dd1c8f446912eebb668ed8946e3dc071865e8a1","subject":"Merge \"update constraint for openstacksdk to new release 0.39.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0dd1c8f446912eebb668ed8946e3dc071865e8a1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:17:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:17:00.000000000","tz":0},"subject":"update constraint for kuryr-lib to new release 2.0.0","message":"update constraint for kuryr-lib to new release 2.0.0\n\nChange-Id: I4b4277cb649ba1399cb30019fbcab9c9c3574fcd\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ide89d1d975e94586f8f58c85baf97d53f0c713de\nmeta:release:Code-Review+1: Michał Dulko \u003cmdulko@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49a481e60f34f8cf08ee8a0c4aed0e4f424ec210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49a481e60f34f8cf08ee8a0c4aed0e4f424ec210"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
