)]}'
{"id":"openstack%2Frequirements~698683","triplet_id":"openstack%2Frequirements~master~I6dd09c4aa6f857aa1fdd59fac9a66838d7ec5c19","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6dd09c4aa6f857aa1fdd59fac9a66838d7ec5c19","subject":"update constraint for monasca-statsd to new release 2.0.0","status":"MERGED","created":"2019-12-12 10:18:29.000000000","updated":"2019-12-12 18:56:25.000000000","submitted":"2019-12-12 18:52:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698683-1576176770712-8a699a53","meta_rev_id":"a529910cc03e61b77802eb5468b2ffe4caad971b","_number":698683,"virtual_id_number":698683,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 18:52:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-12 13:53:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-12 13:53:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 13:53:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-12 18:52:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"041538d2650c61547753d3f366201e656a0c570d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-12 10:18:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b10acc074f656487d2dd0e9841c936008cf78abf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 11:51:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d58e4675f584a7c8c98e5d86adcfd4b : SUCCESS in 5m 47s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/58bbbc5b8fa64e8ebbe6721c16a751fd : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62754db893d44f6d9b3bb4249e965460 : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d92a1d13af9a476a8a4166ff91528900 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a33bc25d230c44638d9a52e5be9438c2 : SUCCESS in 8m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c5595ca961ac4c4ea8a48cb07dda32a1 : SUCCESS in 5m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/84e5aa7f5fc34622a258d72a15d6a184 : SUCCESS in 6m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8c058539302c4be4b4dca03d5ce3de69 : SUCCESS in 5m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8dc5b39d4e60452d8b12b8fc68819780 : SUCCESS in 9m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c253eee84da647e88c323bcdae5905d8 : SUCCESS in 7m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/20f51cf2bb2e4f6a8573c448dc4bfeaa : SUCCESS in 10m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d90102a946ce4ff6a911107f25595b5b : SUCCESS in 7m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b3595f6b1caf44f1910979ff56f0f6f5 : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ec630927494e4e3c8991765fb292e9ca : SUCCESS in 10m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d6c9fb00ace1461baa43b2f8cb58b2c4 : SUCCESS in 14m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/aa207ccbfbc740ba8cc264b1083fb579 : SUCCESS in 23m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/51d2a3919ca1468a8b881ebdd07fc833 : SUCCESS in 15m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/eaaf451156e3432fadc71ac4a3b24e81 : SUCCESS in 20m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/06e4013c65274e81a7eb95d558143630 : SUCCESS in 25m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8ec7befe85d340d0a5c16a479b4b813d : SUCCESS in 6m 12s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/52646da1871c413d81ced49e33e3e2ae : SUCCESS in 7m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/da8aaf0a84434f65851e1b69c080ec31 : SUCCESS in 7m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1293fb849d6d459f92c37e09666720ff : SUCCESS in 10m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ba444879146c471aa47c3a23cbd6b8fe : SUCCESS in 10m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/efe2ff5a03c54da78f8557dd6175a56a : SUCCESS in 1h 13m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/279f3b1880b74ac5909eaf1b73068915 : SUCCESS in 1h 25m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d38ef9eee891417c960d3f8293cc700c : SUCCESS in 57m 56s","accounts_in_message":[],"_revision_number":1},{"id":"560783ff30c4ea6faba1fb846d414ad5555da295","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-12 13:53:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"619a9435ae66519c7ce1895d6cf077d542a42633","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 13:57:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26bb06aff688020df7f1cf3efa67cea1aed38346","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 18:52:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c34153462ccd4350aeea18c5de8f6027 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/501ceb77341d4954b65c853a070be36f : SUCCESS in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e120d647e60471e893912075bb16e1c : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1478ca28c354c2d83fdca06075be6e9 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dd533411b01e410a9cb08857f973188b : SUCCESS in 5m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bb4827617d2b4a6fac21daa85a94a2eb : SUCCESS in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a0f4477243b24304a0a54ddcd64449ef : SUCCESS in 5m 16s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/126df109de0849359edb947c5453f25c : SUCCESS in 8m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c665a94a6ec94c179c4a9b900e811e4a : SUCCESS in 11m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e641402aa16a46379537294097d7477a : SUCCESS in 11m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/38b0379ec03a486ca1b2cfd5caf0f6a6 : SUCCESS in 6m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fd5fa58ba2744d158377f21bb1e65d06 : SUCCESS in 12m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e6f8415b463a4c7596fca789e6c57950 : SUCCESS in 11m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4b7b17cd2ccf4217991f1941cfae320b : SUCCESS in 14m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c2a91d03add24741840d437f43eb2e8f : SUCCESS in 24m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/38f569bd6d144d63a0e5cb2eb0290dbc : SUCCESS in 14m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/21884a3fccb940c38b9c80aad4be9683 : SUCCESS in 17m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/78435e0958dc45e5b1193a6877333f30 : SUCCESS in 20m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/938602a3afc94c9d959be4f7abb56d33 : SUCCESS in 7m 09s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/711a99322e7b4cecacdedd168317858a : SUCCESS in 7m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f969669135a44e37891f48b2e4e38f75 : SUCCESS in 7m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bdb181f08b90408e9e5b1e64426b6bf6 : SUCCESS in 10m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0f596f748af24c17a17d445124b1bb47 : SUCCESS in 11m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7bcfc720748f4c2583aee158d5fc2b4a : SUCCESS in 1h 19m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f12b32a0f9784ffbae7276a123992e03 : SUCCESS in 1h 54m 29s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8bd256417d1b447ab963f81ebab38bb6 : SUCCESS in 1h 06m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4367af5caf31166234a708a7ec95478077420998","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 18:52:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cfc61a41e1114e47d69c33f50ba248be54507b67","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 18:56:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05f6edf42f3e4baf82bfb1b51201caf8 : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"692e718ca1fa157d0e6fc10a73a7ce169ecc9f9c","revisions":{"692e718ca1fa157d0e6fc10a73a7ce169ecc9f9c":{"kind":"REWORK","_number":1,"created":"2019-12-12 10:18:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/698683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/698683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/698683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/698683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/698683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/698683/1"}}},"commit":{"parents":[{"commit":"0dd1c8f446912eebb668ed8946e3dc071865e8a1","subject":"Merge \"update constraint for openstacksdk to new release 0.39.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0dd1c8f446912eebb668ed8946e3dc071865e8a1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:18:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:18:28.000000000","tz":0},"subject":"update constraint for monasca-statsd to new release 2.0.0","message":"update constraint for monasca-statsd to new release 2.0.0\n\nChange-Id: I6dd09c4aa6f857aa1fdd59fac9a66838d7ec5c19\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I2092e5b5b6d646c4b2e5dfc1ede65b7790cea39f\nmeta:release:Code-Review+1: Witold Bedyk \u003cwitold.bedyk@suse.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/692e718ca1fa157d0e6fc10a73a7ce169ecc9f9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/692e718ca1fa157d0e6fc10a73a7ce169ecc9f9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
