)]}'
{"id":"openstack%2Frequirements~698684","triplet_id":"openstack%2Frequirements~master~I420d57bdb6db970bdf33bd566a35fb3561305aec","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I420d57bdb6db970bdf33bd566a35fb3561305aec","subject":"update constraint for oslo.service to new release 1.41.0","status":"MERGED","created":"2019-12-12 10:19:40.000000000","updated":"2019-12-12 18:42:01.000000000","submitted":"2019-12-12 18:39:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698684-1576175986891-8b70566f","meta_rev_id":"e4541c8923a6a0fa2a4788f4c8ee61583a2c9b03","_number":698684,"virtual_id_number":698684,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 18:39:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-12 13:53:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-12 13:53:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 13:53:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-12 18:39:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"492354420ca9c2297764b6b0b5f5a9fc7524522c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-12 10:19:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d67dcbc0c953db3cc94cba9d65494d8f045d264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 12:10:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f81be9734f34794b9b1a0197dbd21ff : SUCCESS in 5m 51s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/665ee14f87d34880982817b5d991f790 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08fadb88cf6e47838970d4e814e09df2 : SUCCESS in 6m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c0f43fa9ec4455996dc97f95b64db87 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b19ec3e86842416c803a36e17dcf5f37 : SUCCESS in 6m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee11775902f9497d905e5be5c5724f6a : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/03fa44c163a943e183c74ea22408f7a4 : SUCCESS in 5m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d696cce6ce3245069330320a3aa014ff : SUCCESS in 5m 00s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/daed129380324cd5baee83f291f6d36d : SUCCESS in 9m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f6c87db3dc074ae59f0f79b38767251b : SUCCESS in 10m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/557fc5733ed541b7a1618b57039985f3 : SUCCESS in 11m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6269e4c08cc74922b48ca20c642fba76 : SUCCESS in 6m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/84d8e662b7324e0d83e6182f75062477 : SUCCESS in 12m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b71c380498a94bc694c94aebca748bef : SUCCESS in 9m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9450796fd1eb412a93b1d60beea2b131 : SUCCESS in 20m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/14417a4fab054714a0f79e937928e31e : SUCCESS in 22m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/32f8b440a6d94feaa6baf1e461c64059 : SUCCESS in 13m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/194497e8c871413386c63fa2e5186ba6 : SUCCESS in 16m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6a4dd15e47a04cbc9fc8a54f35e08bf8 : SUCCESS in 20m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8aaa7e7d1a0c4865a72cb6dc107c00b1 : SUCCESS in 8m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7b6ef35fead644a3961435611e33cea0 : SUCCESS in 9m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fb4e04682b5741a186ec68fe06403cda : SUCCESS in 8m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e7dcabbfc65544b1b9b1240079f1f063 : SUCCESS in 12m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1a5d36cf7faf47cfbfdd4297973590cf : SUCCESS in 12m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2dcd898a24f44397a19701c2aefa2e9f : SUCCESS in 1h 13m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c3ef7e52fcb84ba1b3467f97b4030e0d : SUCCESS in 1h 40m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9b568a9317b041e7a6e8e9981e49d6c7 : SUCCESS in 57m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e8028a341fa1424b2295e1fc63d94d00f23f0f97","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-12 13:53:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"99caa963dcd73b2c08bba85434a1a198497083eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 13:57:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d1c867ee6918bd8f451ab7c04604b95e6e9de21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 18:39:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07a61e97b61e4e9cbca0e435902868f0 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6223acb0370d4b218205fb55bfb36709 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7059675546a4a1290c216e1e89ee049 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bccb124b0ac04a668594accf11bb4065 : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/079a437f6baf4aa6aa0b23845b600180 : SUCCESS in 6m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb9a55db9ff14010a64bd4baf826b3da : SUCCESS in 5m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/05210ab0fb494ee1be6c6b88c0e99070 : SUCCESS in 6m 05s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/18f39d30993f47cea7800cd03e8367c4 : SUCCESS in 13m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4dfee8ba36d44e0a8963ac04c6562fe7 : SUCCESS in 12m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/81f43dc18bfa4f158d37f388726f3edc : SUCCESS in 15m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b1b1f3ddbc9f43a89b4277693c8bb0e4 : SUCCESS in 9m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/170da98c2e254e28a4bbf455d5fb5084 : SUCCESS in 13m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e0a114bdd26740e9b69a2d4814862a59 : SUCCESS in 12m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2b70e012898847fb87d4729afbd3208f : SUCCESS in 20m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0bf6d98e35f14aa1b86dbbf1bd4188ae : SUCCESS in 30m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bf8c746033884a14b856d2546ec2ac55 : SUCCESS in 17m 17s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9b3ccfcc6bd546eb8e97e2592d235e5d : SUCCESS in 18m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/90fb5066e75049b0a7c29adb8d5497a4 : SUCCESS in 20m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1c2d7cb991d8488c88e3ef7656fc31b8 : SUCCESS in 7m 46s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f833a84fe1794a1fa40f9b5c0d24efff : SUCCESS in 8m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aa08c7468b7b46db93328149b8d75aca : SUCCESS in 8m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/049d99baa55544f5aef760a85e80fedd : SUCCESS in 12m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4eb8a7fb497941919db3b91a32b28b9c : SUCCESS in 11m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/34f48ecbbc884c44a1c3c38516e1bcbc : SUCCESS in 1h 07m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4166156ad7c443dd966be3da13f367de : SUCCESS in 1h 43m 09s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ccd2a393240c4ceaa117928e656423bd : SUCCESS in 1h 08m 26s","accounts_in_message":[],"_revision_number":1},{"id":"439de57a67a94d4bbcd98fe56528236ac3807e2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 18:39:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2f408e37221daefd57bc64d61e0e58f782ff8e70","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 18:42:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d64a832e4104225846642e2ed098323 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8f41197c7704b2bd43cfc51d4a3ac49030c6dff3","revisions":{"8f41197c7704b2bd43cfc51d4a3ac49030c6dff3":{"kind":"REWORK","_number":1,"created":"2019-12-12 10:19:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/698684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/698684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/698684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/698684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/698684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/698684/1"}}},"commit":{"parents":[{"commit":"0dd1c8f446912eebb668ed8946e3dc071865e8a1","subject":"Merge \"update constraint for openstacksdk to new release 0.39.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0dd1c8f446912eebb668ed8946e3dc071865e8a1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:19:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:19:39.000000000","tz":0},"subject":"update constraint for oslo.service to new release 1.41.0","message":"update constraint for oslo.service to new release 1.41.0\n\nChange-Id: I420d57bdb6db970bdf33bd566a35fb3561305aec\nmeta:version: 1.41.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I679e0c544092d2de07658c6484c062449d6cb7f1\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8f41197c7704b2bd43cfc51d4a3ac49030c6dff3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8f41197c7704b2bd43cfc51d4a3ac49030c6dff3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
