)]}'
{"id":"openstack%2Frequirements~698685","triplet_id":"openstack%2Frequirements~master~I6e3169233ecded59b608771e667e3f86e9e52165","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6e3169233ecded59b608771e667e3f86e9e52165","subject":"update constraint for oslo.db to new release 5.1.0","status":"MERGED","created":"2019-12-12 10:20:18.000000000","updated":"2019-12-13 05:36:31.000000000","submitted":"2019-12-13 05:35:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698685-1576215308139-75a8b836","meta_rev_id":"d269617de299164bfb7034a10b09df20dac742dd","_number":698685,"virtual_id_number":698685,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-13 05:35:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-12 13:53:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-12 13:53:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 13:53:34.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-13 05:35:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e770c28a1f42d6329ffc587baa4604138fcec19","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-12 10:20:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb3a540c310840a1857ed606301c625b8ff5d6e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 11:51:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66565646c92a4a3abf46422e01f0350d : SUCCESS in 5m 41s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9be8522d06a040a0aeee8b029b31a833 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d90671d50bb4f13b791469a47a56a07 : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e5efc2142867409b9301c729c05ec2aa : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b87c5bacd6143f7ad6207f8ec6ccd8f : SUCCESS in 7m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab93ff22090141529ca44d609f102deb : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ef9fdb81da840ab9dcd44b9241e21cc : SUCCESS in 5m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d9ac7045359c4b9da12267da0b6ff38e : SUCCESS in 4m 59s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/713ffac69faa4d14b6680e288f7d83e1 : SUCCESS in 8m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/046aabf79e2a42e2ae4bb603920f98c3 : SUCCESS in 8m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0a40496315f54d3ca3f885e632346503 : SUCCESS in 10m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e1535ab530864f73854fe1d5474a996a : SUCCESS in 6m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d65d81d285274251ae0f167919a7bba5 : SUCCESS in 10m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d5a6c9e5386c4d12b2a67ed1af5bdda6 : SUCCESS in 11m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1cbc3be557d2445186c95eb7ee0ff096 : SUCCESS in 22m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/890c2fbc7cc14da582349dc94a59664a : SUCCESS in 21m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f02fa2f1bbc844a2a86d31e556d073f5 : SUCCESS in 16m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c91e061af1084e6aa4b3bdad7b0b238a : SUCCESS in 15m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5ea159ca9231498bb26fe78ee298f458 : SUCCESS in 22m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/991da4d52722410c9db512abe4e59750 : SUCCESS in 8m 40s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/72da69bf45e1491989afa2ec505a8052 : SUCCESS in 6m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/50afdfde0314413fbfc3a14a1845413f : SUCCESS in 7m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4105d6b84862478bb3e9b14a457c219d : SUCCESS in 11m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/faa30e858baa4533b6ba96eee4220e2a : SUCCESS in 11m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c46e58e467f54d9bb4554ff88fa859b4 : SUCCESS in 1h 11m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8bd02eaa01b8485e93313c0872d9aec7 : SUCCESS in 1h 22m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8a86b8b9842e45f0905fad488a4bef3a : SUCCESS in 1h 05m 39s","accounts_in_message":[],"_revision_number":1},{"id":"5ba97a09c788f3b12131d10841245aa4dccc174f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-12 13:53:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5dfd8a0202aaf37f1ea47f3ee31c4e68042ab5a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 13:57:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ce0940235073ac154de50d83e64ceed6db35459","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 18:52:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66ac1459f05f4d69861f0d9080c39fa6 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/517387fb0c0f49cf82349347005e30b9 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a3473fd25664478b53fba1c939aee6b : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c7076e118b9417a911fc3c4c094b807 : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef0510c298b44962bc9c76562aad0ca4 : SUCCESS in 6m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e3b2e794e43e46afa45f2f0f7c66d6b0 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/112c1964133346c084dcb4d26e321695 : SUCCESS in 4m 53s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1f911940dbe9475785882c12fa48c79e : SUCCESS in 8m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b332005e039b4fd5bf28f1ab34a20e4d : SUCCESS in 8m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6d4892095f9c41a08ddddaab326e6df4 : SUCCESS in 10m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/317141e227f54a37ac9143c5fc9c9a02 : SUCCESS in 6m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e206aadd5d98408a90fa7504f9ef971e : SUCCESS in 11m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/95294fa96dff4f889571d7f3fb23e941 : SUCCESS in 7m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/be5dba0ee6834b16b474e756e45a8631 : SUCCESS in 13m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/712ca41b97cb4195bdabd9a715afaae6 : SUCCESS in 30m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/43edd6c4c9b4499c9bca3f25c9a2a703 : SUCCESS in 19m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fa4ebbd3fcf6479d98f2f0b40fd23ce9 : SUCCESS in 20m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ed1e0777225a4e7993f8b6711d2e4f50 : SUCCESS in 22m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6e66afdcbcfa4f21b396befd5cc0f2ce : SUCCESS in 7m 19s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0c789b394f934758babb3886dedc70c8 : SUCCESS in 8m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8135cac819974c3c9ca452aa9dcbb746 : SUCCESS in 9m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ca20d988975b441f8fe792937d095c4a : SUCCESS in 14m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/22c5a8d23dd546d4990743f21983349a : TIMED_OUT in 41m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b480f5e555404233950da9ed7d08344f : SUCCESS in 1h 26m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/43079c5111d1425ba696070c41dbc45f : SUCCESS in 1h 31m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/359d48e4e49d42a4b35a130f4341b732 : SUCCESS in 57m 39s","accounts_in_message":[],"_revision_number":1},{"id":"0d059b9274b5b61ea1e713332aacedf1588abf93","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-12 22:01:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0ca230166f484e4a9e7eaa318b41189ec1f986a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 23:52:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5756bbaa6724fb49979f5dafac53879 : SUCCESS in 5m 59s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f9bc406d5dcd4e2d917cc57449373ae6 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd8dbfd819a945d1bd5d2f8d16bef9ea : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f7dfe4120eb4858a50561af8bcb23b5 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53bdd687eea0436fab00cc714d3402c1 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf18926b3c8a4feb9c20caf961857248 : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/261ef8f32996413c86d77f138fe947ba : SUCCESS in 6m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/544c3775ec54492e99b099682000d655 : SUCCESS in 4m 12s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e51aa8573ab1449a8b80cd22f1374024 : SUCCESS in 8m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/362af858ffd140d5a475a7fce875cd3d : SUCCESS in 8m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c705388913744547bbfe30e447e9cf3d : SUCCESS in 9m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/08c9b4c06c3446358f91c146964c3780 : SUCCESS in 5m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9ddc9606d8ff4d5eaaf6bb689ef9c2fd : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a03785f1fae64b12a46d740dc0c21b8b : SUCCESS in 10m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/275dce8dadb649fe85a10c376794e577 : SUCCESS in 15m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/83c06de8e33d4499a0e0019f5ad6256c : SUCCESS in 21m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/aa3bb89849f24312813999c2343a5bb7 : SUCCESS in 13m 04s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7619f4f7952140e4895c91c2a9316601 : SUCCESS in 14m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6cb8a4ca1e9242e8a083198af4bd6811 : SUCCESS in 20m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f581665bd8414f71aeb9f595fea4ee2f : SUCCESS in 5m 27s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7d7bb20c5ae44ec0a090a477d7afa286 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5a8190d1326c4a46a7cde95d40fc0689 : SUCCESS in 9m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4aa1461b6b794ec1898787dd7a4a07fc : SUCCESS in 10m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/037015b7db3c42dc8b2a02bf3b9b199f : SUCCESS in 12m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dbf5586e8fc340159498ba3d1b1114e2 : SUCCESS in 1h 03m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bbab662901254440a51ce44de66704f6 : SUCCESS in 1h 42m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cd55e2353b3c4f2cafc20303cdb032b2 : SUCCESS in 1h 05m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a8f927d90ba4b0c21a363d8bdf8a9e434378e50f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 23:53:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"778484e7389777eb05c1c3f5a6262085b197db8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 05:35:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cd8d9416f30454ebba980dd0cb33ba4 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cae246aafbbe43e192d44e8d3cbcdc42 : SUCCESS in 3m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5714f36e90343359b2b269e6b4a6b96 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3ed3b2a16874b91a5c77c395ad2820f : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee787305decd42a8800bf5a50965cb7c : SUCCESS in 4m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/81bdbf3c5b7d4e3e97bf9cf5cc2aa286 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/133319875e244d0694e1e9c1ced859dc : SUCCESS in 4m 13s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d825eafa2f3249a48db66f9d40827c0c : SUCCESS in 8m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/96dc27a98e2344d7a1716cdacf263a2b : SUCCESS in 7m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d2d0ba3e203d468b9a03efc308364edc : SUCCESS in 8m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b14a165d86484881a6a35d833345ba74 : SUCCESS in 4m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f0119bc7fd2643c0a9842edbfa692286 : SUCCESS in 8m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b65dff54fb4a4f329d6989cfe433a3cc : SUCCESS in 7m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dfe834b2f51c4941bd7516377f8c1d5f : SUCCESS in 27m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/147253312935412eababc94338803373 : SUCCESS in 25m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/20f6ddb5430f41a9b274d36a02436175 : SUCCESS in 12m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/97dbee5d5642401683f895c0c30fe9fd : SUCCESS in 25m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/186beccd44bc407aa71971005df220c1 : SUCCESS in 14m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3c1f7f71a00e413d9f0ffee86f12da3a : SUCCESS in 6m 39s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/772bf22be9a140d0a3347a1f1f8b0a5b : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cf702e6a2d5f4fba84217f012b098d7e : SUCCESS in 6m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/97c5b379a11e4b2c841ec13179350c4c : SUCCESS in 10m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2b6645246a7b4862855350d2e39ecff9 : SUCCESS in 12m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e043544cc8b472796cc0df1cfb6cb3a : SUCCESS in 1h 20m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c1ed8eb23b664966bcbe8a229b342b0b : SUCCESS in 1h 48m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/eca067dea225476ebc8472d316e55966 : SUCCESS in 54m 11s","accounts_in_message":[],"_revision_number":1},{"id":"1ad95903200091915663c94bad0a2a8694993274","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 05:35:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"41281ee0805a60f371b0789413d7c4fdb1b8efca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 05:36:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f19a682814845d39c6570888eaf13f0 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b9664afb7a1578ae315a61ebea41d1008a88f0c","revisions":{"9b9664afb7a1578ae315a61ebea41d1008a88f0c":{"kind":"REWORK","_number":1,"created":"2019-12-12 10:20:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/698685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/698685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/698685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/698685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/698685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/698685/1"}}},"commit":{"parents":[{"commit":"0dd1c8f446912eebb668ed8946e3dc071865e8a1","subject":"Merge \"update constraint for openstacksdk to new release 0.39.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0dd1c8f446912eebb668ed8946e3dc071865e8a1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:20:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:20:15.000000000","tz":0},"subject":"update constraint for oslo.db to new release 5.1.0","message":"update constraint for oslo.db to new release 5.1.0\n\nChange-Id: I6e3169233ecded59b608771e667e3f86e9e52165\nmeta:version: 5.1.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I52c555d2b967f67e14768dba5c5e5ae9e53bd5ad\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b9664afb7a1578ae315a61ebea41d1008a88f0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b9664afb7a1578ae315a61ebea41d1008a88f0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
