)]}'
{"id":"openstack%2Frequirements~698686","triplet_id":"openstack%2Frequirements~stable%2Fstein~Ieebd532b78806e38ce092c3cc00a4094b73a624a","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"Ieebd532b78806e38ce092c3cc00a4094b73a624a","subject":"update constraint for os-net-config to new release 10.4.2","status":"MERGED","created":"2019-12-12 10:23:28.000000000","updated":"2019-12-12 23:42:26.000000000","submitted":"2019-12-12 23:42:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698686-1576194146565-7edb6769","meta_rev_id":"544a4cecd85f22b86c6513fcc57e579084f7a488","_number":698686,"virtual_id_number":698686,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 23:42:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-12 13:53:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-12-12 18:11:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-12-12 18:11:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 13:53:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-12 18:11:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-12 23:42:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12df50289cd4759dbe9a9be5fac3f86698381558","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-12 10:23:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"404c8cc9c919d149235aea3748a5c12df706b6b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 12:05:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7c8605656c314b1db54f616939ab1cab : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f654429cf184567bfc9187bcc9a54de : SUCCESS in 6m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/089da949ca1348588f0e38da3a2567d4 : SUCCESS in 6m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b48dbb4014134652b8f7fed917c66dd4 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/761562025b574203853faf7d96989935 : SUCCESS in 7m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/170fd40b6cbf4acfad35df8eb26684d6 : SUCCESS in 7m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bb2a4f27607746999fca73af91f8d52b : SUCCESS in 5m 32s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/387fcda99bbb41c28cab08696e6549af : SUCCESS in 10m 30s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/8d136175975d445bbb23f9069dc51a63 : SUCCESS in 11m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/940edd8d5f79489fa695df21a19285fc : SUCCESS in 10m 01s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e073485ec94e49ae90c4da74b38073bf : SUCCESS in 11m 02s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/ac0fc5c1c8f641de84857292b7e4c5e1 : SUCCESS in 11m 06s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ac215dfb451242e0a6c79e65849dfe89 : SUCCESS in 7m 48s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/8e262fe71f934fa588dac43a92fd8d2a : SUCCESS in 5m 38s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ef8265d95c7a43d79d7dc3d861bb6ed2 : SUCCESS in 12m 14s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/d86eb6b5fb2b466396667fd5b991af25 : SUCCESS in 11m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cfd538c482674306a8c3517b786c5433 : SUCCESS in 9m 28s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/475dca1411544de2b0af83da48c6debb : SUCCESS in 26m 15s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/b89667c7577e45bd894287cf9fe4f955 : SUCCESS in 30m 38s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/edd7d0d6be85456ab18e6f1276e051b1 : SUCCESS in 24m 56s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/c2b3559dfeaa44c5ad039f410a148051 : SUCCESS in 26m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/dd6a3077ea234c5c9c1ec2397197b004 : SUCCESS in 14m 33s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/15e41ea7a3e44e5c94b6374a7d0ae92d : SUCCESS in 15m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/187d52a80d2c4f5db74ead09e476e95e : SUCCESS in 19m 29s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/376513a4b5d64c5c937e7bbed64cf4ea : SUCCESS in 7m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c16ae85491f64ee58078c897fe585719 : SUCCESS in 12m 20s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/8e8fdc6c7fa44162a4cf8d3ff3a4b86d : SUCCESS in 12m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/900d597eb73e4589bb062d26f86c3317 : SUCCESS in 1h 00m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/acacab197f87428197ba2bb71986e0b3 : SUCCESS in 1h 31m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/74c61f0d715e4802b74128d8d26fb82e : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":1},{"id":"0467fb2d83a4469c9ba8e845c1a7c1250f8e493a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-12 13:53:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e5aece63ba028a0d1cf45266c20e6a67dec877a6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-12 18:11:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks good from a stable perspective.","accounts_in_message":[],"_revision_number":1},{"id":"f7fa07f85b8f30ce630dab35950c310cd08385dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 18:15:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7757104d3bac6e2fca375c5cfda8068c73b3e84b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 23:42:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c0db5002c8a72f0fc88053dcdafb1484bbda4687","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 23:42:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/619d5732ca824b9db05c5805b7b94a2f : SUCCESS in 6m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e96ea1bc45af427c870978c854ff4666 : SUCCESS in 7m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0117a6ef6e6e421084900cf107e36dd0 : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e4b7e7184b24c0da84c488066cdd213 : SUCCESS in 8m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/67d33dc0247f449da352c83e6cb8eb87 : SUCCESS in 6m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9386fb887bab498d987da608054a1072 : SUCCESS in 5m 59s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/94eb90476cb04100bde1bd3173118b61 : SUCCESS in 9m 38s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/f7afd9a9b59a4d879119bb6357dad66e : SUCCESS in 8m 15s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0166d53ebb1f436ca358cb5e12905ef1 : SUCCESS in 11m 48s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/c17f9f250f8d48f3b97d59fa75d77757 : SUCCESS in 12m 38s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ade20090941a4c0c9f6bf2b9d6cf30af : SUCCESS in 6m 58s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/6b8c831f4d1d492aa84ac3e6f8d9cd0c : SUCCESS in 6m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/220a3a9d711548fc8512f5d0517e0803 : SUCCESS in 13m 13s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/a3d7abb2323440b1ba657e52c05f5063 : SUCCESS in 11m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/01730228e60243cfbb4e9a9800811eef : SUCCESS in 11m 31s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d42ddf42b329446caeba4bfe9ac885cc : SUCCESS in 21m 00s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/41449e7e370241dabf0aee469401bc99 : SUCCESS in 30m 02s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/59fd2ba9dfd946cb98f51cc4b0910009 : SUCCESS in 24m 23s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/5781a22b84f2409daadb397877caf18c : SUCCESS in 26m 06s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3d48af25e56d4478a897ab92e100a269 : SUCCESS in 19m 31s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/aad74b7cfec14f2c95be3bbc73584ee9 : SUCCESS in 14m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/167534dc17bc4426a13197d5c55e9bcd : SUCCESS in 22m 16s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/0553c4402c1143f7a5a961ad89243964 : SUCCESS in 8m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9e9ea631fd94461d86cf631c7e74af00 : SUCCESS in 10m 02s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/9f97a11ee2fe44e986691c0ec6285c95 : SUCCESS in 11m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8dd6d3b7547d4cceacb1bc4a9540c5db : SUCCESS in 1h 07m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4772bfe1cd9e4ec0a478a1d2de41699c : SUCCESS in 1h 32m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/45f5e387e7914e688fbb5e2e87e71336 : SUCCESS in 1h 01m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07f95db422ff56a465f1383f0ec79213d77862d8","revisions":{"07f95db422ff56a465f1383f0ec79213d77862d8":{"kind":"REWORK","_number":1,"created":"2019-12-12 10:23:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/86/698686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/698686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/698686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/698686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/698686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/698686/1"}}},"commit":{"parents":[{"commit":"4895a30aea623c343525ef540a7f0125e3fb7c15","subject":"update constraint for python-openstackclient to new release 3.18.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4895a30aea623c343525ef540a7f0125e3fb7c15"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:23:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:23:26.000000000","tz":0},"subject":"update constraint for os-net-config to new release 10.4.2","message":"update constraint for os-net-config to new release 10.4.2\n\nChange-Id: Ieebd532b78806e38ce092c3cc00a4094b73a624a\nmeta:version: 10.4.2\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Commit: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Change-Id: I3a9b2ccb88c92a2da92bc1bbdae8380a5a835d71\nmeta:release:Code-Review+1: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta:release:Code-Review+1: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/07f95db422ff56a465f1383f0ec79213d77862d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/07f95db422ff56a465f1383f0ec79213d77862d8"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
