)]}'
{"id":"openstack%2Frequirements~698688","triplet_id":"openstack%2Frequirements~stable%2Fstein~Ic21670813bbb54990ee3703b20f045d93677f830","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"Ic21670813bbb54990ee3703b20f045d93677f830","subject":"update constraint for paunch to new release 4.5.2","status":"MERGED","created":"2019-12-12 10:26:00.000000000","updated":"2019-12-12 23:42:31.000000000","submitted":"2019-12-12 23:42:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698688-1576194151115-600d80bd","meta_rev_id":"e55b368d836ba644b673cf1ad7647ef6715e2f62","_number":698688,"virtual_id_number":698688,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 23:42:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-12 15:45:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-12-12 18:11:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-12-12 18:11:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 15:45:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-12 18:11:47.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-12 23:42:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d4241f63073c88d5ec1493377a8c4629e4bba90","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-12 10:26:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c6f4807e71e18e0a3f808d7fdb472db505ab4f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 12:02:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4c2f9e13582542129d9f580337050663 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e5f6505d8994f3eb190f29f91b50f0a : SUCCESS in 8m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3be6fedb01184bd696d25fdbe03206c7 : SUCCESS in 6m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ef36f4adcdd24aeeb3c20908023944cc : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f57c3bde3ee4393b98590943fa2f829 : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/737cffa825ed46ab8aa95d80f6c7d463 : SUCCESS in 5m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d9c1b23a305e44c68efa082e81195c18 : SUCCESS in 5m 24s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ea18793b79b046b68666f033627a52d3 : SUCCESS in 12m 48s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/3d2052b25c6d46619079920a79b31672 : SUCCESS in 12m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/73c17814c9064cc4aeb49ba71c3aedec : SUCCESS in 11m 36s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e4a5dd6f829649fbba545e67bd61df0c : SUCCESS in 10m 12s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/b79f2470a8924969add28b054990fe50 : SUCCESS in 11m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4b6782e2c7ae483aa429c149aeb6a703 : SUCCESS in 6m 56s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/9d41299e540d4cd6a2c2cb68baf0c67a : SUCCESS in 7m 11s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/6b83246cbfb4402aa55eaf406c75d896 : SUCCESS in 12m 58s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/6281b8813a504b18bd30d1c1626dcb75 : SUCCESS in 11m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1a941b7ce8b44f4c8accfa9b4faffb88 : SUCCESS in 11m 18s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2e8ee3fbbeac41c58f5459c91c9a93d2 : SUCCESS in 21m 48s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/f2f5569eeec248768ade472ecd5be0a1 : TIMED_OUT in 42m 17s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6615e71886934a3fbe2fa058ff170704 : SUCCESS in 27m 37s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/f855fc04ce634f4cb5d5ab78e30c6aac : SUCCESS in 26m 23s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d9255aab8d864bf385e062efe5c26740 : SUCCESS in 13m 43s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/05cef4d03934459580a9f174e52e3a68 : SUCCESS in 13m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5270811750624188b21eeb376ca20247 : SUCCESS in 27m 38s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/aea53c38aa5040548f5c4904b26fec3a : SUCCESS in 7m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/14a4c62dd97e409380bec47caebfc28c : SUCCESS in 10m 57s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/9b7472b24d98459b9d238f7bfe5469a4 : SUCCESS in 12m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c35c4950edd4810a39f267cb7349005 : SUCCESS in 1h 12m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e706dfd332574df7b879483f06d0fca1 : SUCCESS in 1h 30m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2da8939ef22748feba49262c9541401c : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":1},{"id":"ef06c62ff41253b10689288ea11be2e48766b452","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-12 13:54:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"44a3b254058348373a57d1641677a3f6a6220cca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 15:35:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6b6e3696f59d4d4cbc4f41cf8559147a : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f72cbd5f3ef4b09b285b8578b30eef7 : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6032912d6dc74f18b49dbaceb8e2217a : SUCCESS in 6m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6d2021f054934a2093b0ddae5297d033 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b893ba13a5d5419e8b6c757ad5febf5b : SUCCESS in 6m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/18b39dcd43d04bcaadbd63ef3d2ee9f5 : SUCCESS in 7m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9bb2b242ae354875947186e011d571d4 : SUCCESS in 5m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/28769973566545d9b42bc22f369a2600 : SUCCESS in 10m 11s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/1f88beda36734c0292b418e2b5e4f6c0 : SUCCESS in 8m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d5085e3fbb674daaad8b36d80e29e9c7 : SUCCESS in 7m 59s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/395e5e9bd3434ac09a1272a5006803f6 : SUCCESS in 11m 21s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/f7c7b116a7bf459ca742284444f4e584 : SUCCESS in 11m 03s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f367907471324cecb174855408db24eb : SUCCESS in 7m 57s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/c07a3432b2ec467986f365fed7eb2c58 : SUCCESS in 6m 01s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/44d099186a3143c285ca4e31d2deca15 : SUCCESS in 12m 44s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/dc60fb602ed64a20ab984581ea37a283 : SUCCESS in 11m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c9aed87ca84a4348a41878c78b13a56d : SUCCESS in 10m 51s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/68f9b94a288d4852980f2056840e8355 : SUCCESS in 22m 25s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/733bdc359c594274a9ab50f5ed5c0e32 : SUCCESS in 32m 08s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e54b6ffe4ea841078d13a7e07d5f4eba : SUCCESS in 23m 30s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/faa2d2430e534b81b5cdaeb8b9208e02 : SUCCESS in 26m 20s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a128ad2080d04c3ba171aee9b4815007 : SUCCESS in 14m 39s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/3388d7e7bf454290948b6fbe01b46f1d : SUCCESS in 14m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4a3ccbdac74743d8884a07502dfffe4c : SUCCESS in 18m 41s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/944959df064845d3b35c8f2247f93416 : SUCCESS in 8m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3973d41a743042ee9aae91c59ded3573 : SUCCESS in 10m 21s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/ad98cdb7628a44bb845cf2442474c469 : SUCCESS in 9m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd05c051094e49dc8a56274aa167ef62 : SUCCESS in 1h 09m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/600bc0eeb36749ecb08d50ef05bc1fbc : SUCCESS in 1h 36m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/87f2d72e0fdf4e53a826071399514311 : SUCCESS in 50m 58s","accounts_in_message":[],"_revision_number":1},{"id":"fac9405ab7e9d8f5c1ae2ba2c48280ad2b11a812","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-12 15:45:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ab8b909c4a60a40fed20a2c94f9248056d657d8f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-12 18:11:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks good from a stable perspective.","accounts_in_message":[],"_revision_number":1},{"id":"38d95d813baf85655f58c8d14e0dbb7f997c3882","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 18:40:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"115a05a95065ab6f6dc6514a9290ea8563a9c4ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 23:42:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acdbf8f8ca374289ad775856ad68896d : SUCCESS in 6m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/045e9904809f4251a26a31d33038c1dd : SUCCESS in 6m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2418acfc3a9044d78352501180d7a4e4 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82b7afed2572432da34c470606c4d277 : SUCCESS in 6m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/108d6fa5090742e08605f6c0ded04480 : SUCCESS in 6m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/49ddff2d2044415b94f42276148d4375 : SUCCESS in 5m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/65fa3731da9943d19da8f3343f03a8c0 : SUCCESS in 9m 13s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/9f9246a4daec4629806729b25b732931 : SUCCESS in 10m 29s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f85abc6f40b94b408eed3880985f6258 : SUCCESS in 12m 38s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/b3796d1a07de41b29bd517e4138bb1bb : SUCCESS in 12m 11s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2ff785de1eb54ca1ba74d6fd4b4e56ec : SUCCESS in 8m 22s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/ea0a745d8c7d478ca563651035d01d90 : SUCCESS in 7m 47s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d558697e6de04f90b35dea682f2fb239 : SUCCESS in 15m 04s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/4d009cbbe5b746df9ee4e874c0db4e9b : SUCCESS in 12m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b11b79e0ee4e47099021a8e33f0c04c1 : SUCCESS in 11m 02s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7a82cbfec99348f484cd561dd9eec455 : SUCCESS in 21m 34s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/da7f39a136134b88988787efa101d402 : SUCCESS in 36m 54s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0769b7f63d1e4af793c959427c10e8b2 : SUCCESS in 26m 12s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/d3db996df84e42d6975a5904efa60c70 : SUCCESS in 30m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6755931d379f46bcb1fbbdcbd8c2bd2b : SUCCESS in 13m 09s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/7d1c50302be74951bb66d5b19d0f1e5f : SUCCESS in 18m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4ca2fca9865649dcb790b1edfb35bddb : SUCCESS in 21m 32s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/15adab253dfd4ad9a4482fcc14c08b81 : SUCCESS in 8m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f18c66ec7da64c9bb9216bbb70927a98 : SUCCESS in 11m 04s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/937ca331eb914457b4433328ab2765dc : SUCCESS in 11m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/705c7ffa90004acb81f3c3f1d3dd9133 : SUCCESS in 1h 17m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5e309696708d47a78dadc1dad8f9957d : SUCCESS in 1h 53m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e8b055e8f8404131b8b3a6f90a8fdb45 : SUCCESS in 1h 03m 41s","accounts_in_message":[],"_revision_number":1},{"id":"11a7641110f2987453d96ebb939dab1f9510a7c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 23:42:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ecf74a72ccbbdb5c3639e6cf1a9a8bf46c451bf","revisions":{"0ecf74a72ccbbdb5c3639e6cf1a9a8bf46c451bf":{"kind":"REWORK","_number":1,"created":"2019-12-12 10:26:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/88/698688/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/698688/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/698688/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/698688/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/698688/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/698688/1"}}},"commit":{"parents":[{"commit":"4895a30aea623c343525ef540a7f0125e3fb7c15","subject":"update constraint for python-openstackclient to new release 3.18.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4895a30aea623c343525ef540a7f0125e3fb7c15"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:25:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:25:59.000000000","tz":0},"subject":"update constraint for paunch to new release 4.5.2","message":"update constraint for paunch to new release 4.5.2\n\nChange-Id: Ic21670813bbb54990ee3703b20f045d93677f830\nmeta:version: 4.5.2\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Commit: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Change-Id: I3a9b2ccb88c92a2da92bc1bbdae8380a5a835d71\nmeta:release:Code-Review+1: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta:release:Code-Review+1: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ecf74a72ccbbdb5c3639e6cf1a9a8bf46c451bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ecf74a72ccbbdb5c3639e6cf1a9a8bf46c451bf"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
