)]}'
{"id":"openstack%2Frequirements~698689","triplet_id":"openstack%2Frequirements~stable%2Fstein~Iaf656f9c4b8cd9925a0c555cd23cf46fad59e0fe","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"Iaf656f9c4b8cd9925a0c555cd23cf46fad59e0fe","subject":"update constraint for os-refresh-config to new release 10.2.2","status":"MERGED","created":"2019-12-12 10:27:46.000000000","updated":"2019-12-12 23:42:27.000000000","submitted":"2019-12-12 23:42:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698689-1576194147652-9875c5dd","meta_rev_id":"75d0201b2a9999547a5cc8537195bdae70878061","_number":698689,"virtual_id_number":698689,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 23:42:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-12 13:53:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-12-12 18:11:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-12-12 18:11:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 13:53:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-12 18:11:28.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-12 23:42:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb06257e36967237a4da58a447982f25b1fee6d2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-12 10:27:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccb84aef33fdc700ff2fdd2a72b147905d281118","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 12:22:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6d2066a0def142c58396947474f891be : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87a6448801054e309c6aaa03655d4d56 : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3d427973ebfd4001ac6608ae4f5bb599 : SUCCESS in 7m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/235e7aa8cb98452b927dc73baab2547b : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c847252ef11540ad9a05827861141a41 : SUCCESS in 6m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd3132ea1a4e4e7b8cd933c379d7b975 : SUCCESS in 6m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dea7ac99218a45c0acd7620d0afad309 : SUCCESS in 6m 33s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d74ea99b12ff4405ba4b7a66903cef77 : SUCCESS in 12m 37s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/4845a5cdf51548578792154985eddaf3 : SUCCESS in 10m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7af759e3acb24dbda82ada69cbb38c3b : SUCCESS in 8m 55s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1a5c86370e1e431aab446cd019b110f4 : SUCCESS in 10m 43s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/d9b78c51d53f442c93e9b6eada3f1213 : SUCCESS in 14m 18s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2c9ed606bf9f4906bdeb67d6e3491bb1 : SUCCESS in 7m 56s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/b02aab7b4e4b41f1a27a9603a3928868 : SUCCESS in 7m 41s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c69031e4c8f443f9ac02a41e21989e1a : SUCCESS in 12m 52s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/d17e22a67a0c403e9cb6df2d42987731 : SUCCESS in 11m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8f70c33d86734991835692fd091ea46e : SUCCESS in 9m 37s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2cf9787bc089406a898f7e622dd3ab9e : SUCCESS in 23m 48s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/d645d3a3df884c5f8e5af14720e7bc82 : SUCCESS in 29m 24s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/60a5e90a53314fe98a3e10e23851d92e : SUCCESS in 28m 10s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/883d66c4c08b4ba682a616ac9f32aef0 : SUCCESS in 24m 19s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5abdfd9cbeea4f0ab250318b5b516fec : SUCCESS in 23m 57s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/1ca4d9f41c314ad4a9fa656abc9c1fb4 : SUCCESS in 18m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/16b4fa23f4744ac18b1b97efb609f286 : SUCCESS in 23m 45s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/d6dbe5a4866f4ac093a42f5edda71674 : SUCCESS in 5m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/785e1fb36a9e4d298be2b29c8bc3316f : SUCCESS in 10m 20s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/2be3a683f512460b98fcbbf172ac3df3 : SUCCESS in 11m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd29576ed40945e7908d534ce58f3274 : SUCCESS in 1h 21m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e90d37de5afe45da950d3e96d482025a : SUCCESS in 1h 46m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bbb1ac5b72f543cf8cfccca4e564eed6 : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":1},{"id":"11b50c3d5ec9daca493b2577976d0714b5c0415c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-12 13:53:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5dc049812e3661d46c35015467700c2a3773fca0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-12 18:11:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks good from a stable perspective.","accounts_in_message":[],"_revision_number":1},{"id":"ce77fa3d2532bc9869417053c21ea9b9f8810117","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 18:15:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c3c835a64cf079075f4bf12a26aff025dc26e1fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 23:42:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5a269c2b288593cccf3e8b9cb55533bde576e306","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 23:42:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adfe0500b3a547c4b77d6fabae1664a5 : SUCCESS in 7m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d57eb4368304476a8527ffe0a9b65c4d : SUCCESS in 7m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1d963555adab4251935ddc46ae620558 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d271f1aa8634aab9cf9ecb3eb273df6 : SUCCESS in 5m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c48074050a2945a6bace07336055d22e : SUCCESS in 7m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/81a4b7a2c4fd4f5aaf821d7b9f721c6a : SUCCESS in 6m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/272ee73c1bd545f9aef3fcc8eae17aeb : SUCCESS in 9m 44s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/dbac327c17704ac1ae32f095b06a648c : SUCCESS in 8m 53s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/301e6af4509b451994b07c390dfc9f0d : SUCCESS in 11m 08s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/52a7faa934c24163af7c2891e2256282 : SUCCESS in 11m 27s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/61765ada06da4e52aeb92501f49e179a : SUCCESS in 7m 30s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/c4fc4e2b91a944aab15292c60823f3f8 : SUCCESS in 6m 41s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9cce6644345a4625aec95975de0efe51 : SUCCESS in 13m 45s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/1a59de9c4fa441fe8c8c99cda2f6d498 : SUCCESS in 12m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3207518135db48fe8953a1678a036962 : SUCCESS in 9m 46s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/230c75f5f0f84bdca59735c170d58486 : SUCCESS in 26m 27s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/500005f4cf354af0a28f8cf67df34e22 : SUCCESS in 29m 05s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f5283fbe558a4ead8af5702ad1640ff8 : SUCCESS in 22m 18s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/e8d4a4540aac4467a7a33238f7690bea : SUCCESS in 26m 04s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ea8bc86434b3459db76a0d3eb6cdd1da : SUCCESS in 16m 15s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/7bcd04a2b9bd4dcc901bc811fa358f03 : SUCCESS in 17m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cb97821a1fd64512b57b0981f17fb6b1 : SUCCESS in 18m 20s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/8becd021b5d64659822f65af3565b446 : SUCCESS in 7m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ea73bfcb56a14050a52ef545e366b6c7 : SUCCESS in 12m 34s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/8cf53d2c61e04b52ab691b60292636b0 : SUCCESS in 13m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/40923e2c5add4f5c98b5eebe36a0fd49 : SUCCESS in 1h 05m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/843ef901b34f4232b937baac867ad907 : SUCCESS in 1h 28m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d010272284dc49baa3c32df286430796 : SUCCESS in 52m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b83b541bd37d6c428ce8e1d538fd25bf2569fcc8","revisions":{"b83b541bd37d6c428ce8e1d538fd25bf2569fcc8":{"kind":"REWORK","_number":1,"created":"2019-12-12 10:27:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/698689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/698689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/698689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/698689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/698689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/698689/1"}}},"commit":{"parents":[{"commit":"4895a30aea623c343525ef540a7f0125e3fb7c15","subject":"update constraint for python-openstackclient to new release 3.18.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4895a30aea623c343525ef540a7f0125e3fb7c15"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:27:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 10:27:44.000000000","tz":0},"subject":"update constraint for os-refresh-config to new release 10.2.2","message":"update constraint for os-refresh-config to new release 10.2.2\n\nChange-Id: Iaf656f9c4b8cd9925a0c555cd23cf46fad59e0fe\nmeta:version: 10.2.2\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Commit: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Change-Id: I3a9b2ccb88c92a2da92bc1bbdae8380a5a835d71\nmeta:release:Code-Review+1: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta:release:Code-Review+1: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b83b541bd37d6c428ce8e1d538fd25bf2569fcc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b83b541bd37d6c428ce8e1d538fd25bf2569fcc8"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
