)]}'
{"id":"openstack%2Frequirements~698750","triplet_id":"openstack%2Frequirements~master~Ib3cc44469db106e81d63390d0cd4051fc4fdde7c","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"Ib3cc44469db106e81d63390d0cd4051fc4fdde7c","subject":"cap sphinxcontrib plugins that are incompatible with python3","status":"MERGED","created":"2019-12-12 16:17:52.000000000","updated":"2019-12-19 07:33:39.000000000","submitted":"2019-12-19 07:31:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698750-1576740688758-fead5b33","meta_rev_id":"e5c3d16a00ca4f744fc3daf69af19a0d11df662b","_number":698750,"virtual_id_number":698750,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-19 07:31:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-12-12 18:10:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-12-18 17:20:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-12-12 22:05:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-12-18 17:20:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-12-12 22:05:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 18:10:33.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-12 22:05:37.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-12-19 07:31:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"746511d68401a2825e72d0b70d04523d0a539172","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-12 16:17:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e475f0db14ce38998901b54e4d7b10cdb993efa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 17:49:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df1243352f274c9dadd03bde5e05bba3 : SUCCESS in 6m 48s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ebd96510cadd4a51b200d531f1fddd4f : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0623189246d3499589def69ea13cc4ff : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9cc92c6d14b7449281f84bd761c3353c : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ce78cba4c9646b5a60d0ee3f9b79b5e : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b5cf7dcfecb4032aa94bd2dc9f3556d : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a8c25b6ed3f34a76b1eaac02ed6bff10 : SUCCESS in 4m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/02e99b2f9f034ad187fa582bf440137d : SUCCESS in 1h 19m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4fa2a1864968412cb3a5010c8e994bdd : SUCCESS in 1h 30m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ca9549342abb4564bb38642400940d3b : SUCCESS in 1h 01m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6b2be025613e853ccee51b47be6ca3e110415adb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-12 18:10:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"110c0c20010cb37004355a976f9ae0a734903744","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-12-12 22:05:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5337976e0dc9108d5b83fc3834bb12ef4cc88f23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 22:34:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1b10a08dd045e20e0bc55b65be4a42af2ce4939","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 02:52:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f2da5288b294dec9a91453093264acc : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4278dac23b4544eb8947045493faa497 : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a50d4539491740a2b11e5b7e8a6bb9ac : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1926d40b6c96455e890b13e9800711de : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7ae9a7495fe45f4af2db86afbc3e05d : SUCCESS in 5m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/afef2ab30f644be1997f497e70409eed : SUCCESS in 4m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f755d1117d0541b3a115040b7b53d278 : SUCCESS in 1h 39m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b967555567a345a0aa8ef2c087d1aede : SUCCESS in 2h 04m 07s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e18bf81b5f0e4b129c8f086654591fa7 : FAILURE in 1h 10m 45s","accounts_in_message":[],"_revision_number":1},{"id":"36c09a9708aa65d4b1087acc6175a2fe94c45423","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-13 04:09:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8a22691ca303ce145288ebd7c6ac281dc7c78873","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 05:38:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eadae98d3fb549ca89f8d1dcc7a92068 : SUCCESS in 4m 12s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/84873bd479e646dd9c1b1dd79f7dc968 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55de4d71d421442faaf493d1bf8731be : SUCCESS in 3m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b4585f1a680d44169a664e59f5e7f8cd : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4dcec6c91da5450f847cc723b5d6c0d3 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71a4697899bb4116b6fd9b057082f320 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca71f35ccb6746f1a8aa0dc15ba8d4ea : SUCCESS in 3m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac6605df95ef4ef2b872b626246d6bb8 : SUCCESS in 1h 07m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dad5ed8f7c744fd0aa15cfa9e2fdf2c3 : SUCCESS in 1h 27m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/64ec541ebb3f43ad9510712e292da012 : FAILURE in 45m 27s","accounts_in_message":[],"_revision_number":1},{"id":"02f99de02254189ab15d9d3b7575270ed8866e9a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-18 17:20:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e2014aff7f6db8ee30ed9169b461812584e0d872","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 19:25:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2f08a30b41b42568bc3bb39c7fd9e1c : SUCCESS in 6m 05s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/edb043be0f3e4541b5de6a0e65c24217 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05c4aab265a04f93ae3dffc07b4a969b : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f09be752a92247399405e832ba6515ba : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aaf5d5efb0194c758f79e0fcebd8705f : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/23e9e6d40aa24554ba75c8b226d17e97 : SUCCESS in 6m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/091db1c38dcd4dce85a5c0b894347299 : SUCCESS in 4m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7dcc97df4c3459489c1676dcbe7422b : SUCCESS in 1h 22m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4e7a1d914042468ba9af9a6c430d1564 : SUCCESS in 1h 57m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/79c713a6d4fd4bd091ed757a94f32522 : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":1},{"id":"aa64c88d205edfd5a7fc5cc78a39ad4fd4894444","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 19:26:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e2def1e57bcb22df8b982e80dff7de89096287bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 00:07:08.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"83d17b53c1023f2322357aee0c33b99045076842","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 04:15:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96697028c50c4c1e9013a79243840e95 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c28e583ae1b849e1a6f5194bd407ad94 : SUCCESS in 3m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/084963e5e8e94f5f8226fbace8412a0b : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76253ebc7888417cb641acdeece7c4c6 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/975c0862716c4c5fafe703de461c802b : SUCCESS in 6m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94aa420536ae499abef0a5109afd7ff4 : SUCCESS in 9m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4f20b39cc0a4e26bd741cccfc2297e5 : SUCCESS in 1h 10m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5c8d11fd85f04eeb906dfecf82e355db : SUCCESS in 1h 34m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9443834dc0c14566b21403133b36505b : FAILURE in 43m 55s","accounts_in_message":[],"_revision_number":1},{"id":"673f1a53f3083526b1320ef29cb2e04fc17d347d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-19 04:22:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bde1760113cb08fb8899f0b3a02c5a3ff04f0d23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 05:46:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86c590da684543549f2763e1df5b8f06 : SUCCESS in 4m 05s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/017e6f8a4e6e466daaa8de12b04dfc2d : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17eb837c2ecf46dda6d6f070d82b2630 : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c2ba62f1a8d647b681370c990228530b : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fe261c5eb444baba199cf82c1cb8bcf : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d34d61c8a6d4777b69e5e97024bda84 : SUCCESS in 5m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/211e4c8b39204e2880fac1697776618a : SUCCESS in 3m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7fa2175ed45145d8919563df87cf9f4b : SUCCESS in 1h 06m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8280c77642fe4acb93a39f3b29bc503f : SUCCESS in 1h 21m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d00d3090c63147aeb9b6c2f7a12e3d9a : SUCCESS in 56m 53s","accounts_in_message":[],"_revision_number":1},{"id":"6d810ca286b0796daa34d282fa78c76b845e4f8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 05:46:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6a423b4c24a83e2f43bfb277278efcc276d3dc57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 07:31:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0eeb44d43924047af4e102c2ad4bdba : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2e6eab0e273417c9d64f1b9541e0c31 : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f87d5ee6a9a94749ae87370658a93608 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e8e067595aa43dda105902d8efdc25d : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/27f1271540cb470bb36c051c77df8360 : SUCCESS in 5m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6cccf321ece4483d8d2ebe0691b3a510 : SUCCESS in 4m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3c69b9152054547b5ac33e99d5e8aec : SUCCESS in 1h 15m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ca17f43bbe1b4988b3656d9ca1d0d92a : SUCCESS in 1h 41m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/53e65e6e45064ad4b8aa956b84434300 : SUCCESS in 1h 09m 19s","accounts_in_message":[],"_revision_number":1},{"id":"1847bdf894141e65eb8a3020ad92c0e603c6c050","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 07:31:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e5c3d16a00ca4f744fc3daf69af19a0d11df662b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 07:33:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12066ded351c4f3786f344512d53770b : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd48d7eecb53b6231d06dbedd483943f3f0febdf","revisions":{"bd48d7eecb53b6231d06dbedd483943f3f0febdf":{"kind":"REWORK","_number":1,"created":"2019-12-12 16:17:52.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/50/698750/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/698750/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/698750/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/698750/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/698750/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/698750/1"}}},"commit":{"parents":[{"commit":"0dd1c8f446912eebb668ed8946e3dc071865e8a1","subject":"Merge \"update constraint for openstacksdk to new release 0.39.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0dd1c8f446912eebb668ed8946e3dc071865e8a1"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-12-12 16:15:13.000000000","tz":-360},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-12-12 16:15:13.000000000","tz":-360},"subject":"cap sphinxcontrib plugins that are incompatible with python3","message":"cap sphinxcontrib plugins that are incompatible with python3\n\nChange-Id: Ib3cc44469db106e81d63390d0cd4051fc4fdde7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd48d7eecb53b6231d06dbedd483943f3f0febdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd48d7eecb53b6231d06dbedd483943f3f0febdf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
