)]}'
{"id":"openstack%2Frequirements~698959","triplet_id":"openstack%2Frequirements~master~I640290359c0c15035187a62da1bc57dba5912b05","project":"openstack/requirements","branch":"master","topic":"reduce-py3-noise","hashtags":[],"change_id":"I640290359c0c15035187a62da1bc57dba5912b05","subject":"trivial: Variable rename","status":"MERGED","created":"2019-12-13 16:48:26.000000000","updated":"2019-12-20 02:26:45.000000000","submitted":"2019-12-20 02:25:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698959-1576808731467-a981c286","meta_rev_id":"156e7caf69f1351ebf95ac0038e9f05d87b4cc2e","_number":698959,"virtual_id_number":698959,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-20 02:25:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-12-18 19:34:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-12-20 00:59:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-12-20 00:59:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-13 16:55:59.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-20 00:59:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-20 02:25:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b6fbc4e4e735743581734f00eb5caef95eb71ab5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-13 16:48:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67a08d982f6254c0c63187dbeaa9eee29a2eaaa3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-13 16:55:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dd5b0b1de3558e0084a97d1b29c9f7ac8d08a6de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 19:06:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef02378ac05543bcb719c67663881791 : SUCCESS in 6m 22s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f789ef2c479048039185a1c7cccfc819 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65da2c8b30b04730bfcc473a8b057eb3 : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5cdf5875445d4e4b99e26bccb1605bf9 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30afd55e50134b67b590817fade65410 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f7dbe9efb6b40cda40c5411a16761a2 : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3c5f1f80398540bfa08e706750ac8303 : SUCCESS in 4m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b03f912124844cd193311261b3a6a6b3 : FAILURE in 1h 16m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4934a7acbe344a68855bda7894a83f79 : SUCCESS in 2h 01m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/83a980bdcfab46d0a556ade929f63c8c : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":1},{"id":"d881869403cbd6ba35b03299c05b1c24c69e5181","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-13 19:22:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"187bd7b91f870823d0a4e73ee15c25f05b6ea646","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 20:53:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e9ebc0efbfe49999d9b47cbb2cb7af4 : SUCCESS in 3m 34s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/31ea9d18825d482e92dbb7275df76ffc : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ac0aa67b0fd49749fde84d2a2d06592 : FAILURE in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/93dbbb986b8049d4b81ec353047d60fc : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/345a1b0a4cc941e483b9ca8ec89d652f : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35d89f9394ab4f6db03a8968b231d79f : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad9e72d91f37499ebdb63a320d22aeb5 : SUCCESS in 3m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b74056eecb6b4a07a6133ae8c3170a6c : SUCCESS in 1h 19m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/14f4eda4010a4c0ab1c46dc47d6ca108 : FAILURE in 1h 29m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1bfb576bce3b49fc89b41cb8e2c69841 : SUCCESS in 57m 09s","accounts_in_message":[],"_revision_number":1},{"id":"47d11f7a7a5d9b365d8a475f947e848ea55277bd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-18 19:34:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"040e946721b826c4014d3d3c4abec5a4ff3ea210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 21:11:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6422993b9aca4da8a9645f7891fe4f6a : SUCCESS in 3m 30s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6fbc24acc3d341d89919fe70a582a710 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7abd321a1e604fcf89b33ba1feb7a224 : SUCCESS in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/56672f66beca489493782cd5cc3afcca : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aaacce8dee4541e4bd9234b74449574c : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b0ca79a10a384dd099e93e1378e7a7c0 : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f1ea548c0a6d417bba6ce50a728c14c8 : SUCCESS in 3m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e927e3e20ef04893ac79cda9c14161ec : SUCCESS in 1h 12m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/09f79e6de4bf4f08b86fb02490c5ef44 : SUCCESS in 1h 31m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a4243019c43b407ba715f8081b6a29fa : SUCCESS in 1h 03m 49s","accounts_in_message":[],"_revision_number":2},{"id":"37c7785c8649991719664020d6545e30ec5fc703","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-20 00:59:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2ce50177fd67cbeaa9e3efcb635b2339a0b2eff2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 01:00:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bf1b155923bfc9b502636d0e51a6f17e130be171","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 02:25:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7e1e96ea77192890566c059973c48ce99b9c8a29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 02:25:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/047c1563d38e4275927d98b6cc9913e4 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7abd572e7a2f4d2c92d87ecb9cc28f25 : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7c441c4093b7417990f492c7e1ec3b31 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb11e0401dc34178881f44d0585cce1b : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c89177fdd6d4c648245ba61495f0a48 : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e7c9af3fdcc4ad7b1b7659fb008f492 : SUCCESS in 3m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7a80f48355c41cbbb733b8984825d5e : SUCCESS in 1h 03m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3a686cb850f54994a82f4e36b48176a8 : SUCCESS in 1h 23m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c5d6ad2ee71f4c77901221dc7113e062 : SUCCESS in 57m 00s","accounts_in_message":[],"_revision_number":2},{"id":"d5d6a26dba03a7fed6752267a502cb6a2410289f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 02:26:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35b07191ece44a20b2f7258c0f870887 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"40de09edad40f02ec73604c875201fd72f1883d7","revisions":{"129c378b066183b9644d5bce992b0138ccc84566":{"kind":"REWORK","_number":1,"created":"2019-12-13 16:48:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/59/698959/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/698959/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/698959/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/698959/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/698959/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/698959/1"}}},"commit":{"parents":[{"commit":"5620d2c1f9abc6c278c13f3ef867e4136923efbc","subject":"Merge \"Requirements-check: use local repo like any other library\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5620d2c1f9abc6c278c13f3ef867e4136923efbc"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-13 16:05:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-13 16:05:06.000000000","tz":0},"subject":"trivial: Variable rename","message":"trivial: Variable rename\n\nBefore doing any work on this, let\u0027s use some more meaningful names.\n\nChange-Id: I640290359c0c15035187a62da1bc57dba5912b05\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/129c378b066183b9644d5bce992b0138ccc84566"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/129c378b066183b9644d5bce992b0138ccc84566"}]},"branch":"refs/heads/master"},"40de09edad40f02ec73604c875201fd72f1883d7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-18 19:34:03.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/59/698959/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/698959/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/698959/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/698959/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/698959/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/698959/2"}}},"commit":{"parents":[{"commit":"d55acd3ea6fdf736b9c3ffc0b8bed2b8bdb71931","subject":"Merge \"update constraint for oslo.db to new release 5.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d55acd3ea6fdf736b9c3ffc0b8bed2b8bdb71931"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-13 16:05:06.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-18 19:34:03.000000000","tz":0},"subject":"trivial: Variable rename","message":"trivial: Variable rename\n\nBefore doing any work on this, let\u0027s use some more meaningful names.\n\nChange-Id: I640290359c0c15035187a62da1bc57dba5912b05\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/40de09edad40f02ec73604c875201fd72f1883d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/40de09edad40f02ec73604c875201fd72f1883d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
