)]}'
{"id":"openstack%2Frequirements~698960","triplet_id":"openstack%2Frequirements~master~Ie6dba8bc78dcd5fa95c7ac15b0a3770cdef8ad95","project":"openstack/requirements","branch":"master","topic":"reduce-py3-noise","hashtags":[],"change_id":"Ie6dba8bc78dcd5fa95c7ac15b0a3770cdef8ad95","subject":"Allow projects to drop \u0027python_version\u003e\u003d3.x\u0027 version specifier","status":"MERGED","created":"2019-12-13 16:48:26.000000000","updated":"2019-12-20 20:21:24.000000000","submitted":"2019-12-20 20:19:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698960-1576873168255-716aec10","meta_rev_id":"81c424b057b74b521f48ba845fe057f66bd9d1e2","_number":698960,"virtual_id_number":698960,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-20 20:19:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-20 05:23:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-12-20 13:09:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-12-20 18:32:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-12-20 13:09:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 06:22:04.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-12-20 05:23:27.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-20 13:09:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-20 18:32:11.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-12-20 20:19:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2378826de5d4c4a60690bfba70ff89fd124965b1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-13 16:48:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"794768c1525e22b4f04d63b107d1ec707e255267","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-13 17:01:33.000000000","message":"Patch Set 1:\n\nLooks like this is out of date and doesn\u0027t have the updates from https://review.opendev.org/#/c/693631","accounts_in_message":[],"_revision_number":1},{"id":"1097607a9a582583cecf7611f807c9ae853b1ae9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-13 17:03:59.000000000","message":"Patch Set 1:\n\nWould be good to add a unit test like test_new_item_matches_py3_allowed_no_py2 to cover this case.","accounts_in_message":[],"_revision_number":1},{"id":"a2a8451b3d8e98080463583805c2ebdccc8f517e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 18:51:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18a55f516113471cbe1bcac5c1bd4e1f : SUCCESS in 4m 29s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ddbdd4ddd43343bd90c7344b2c671d2e : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0de4cd8602a4a549428226086a0a417 : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/88f961d40b764d068640eeb9d8e80a00 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/948ae16812db4952af56b7b461a01de3 : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4871a8fc2e1a4aba9244a56b0316fe54 : SUCCESS in 6m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fc3c11bf298f466da75fa4fb5664b05d : SUCCESS in 5m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/305450220dcd4327af3f5da1bb50bd2f : SUCCESS in 1h 20m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/000163e388d143d794faa6937c3441e7 : SUCCESS in 1h 56m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/be213274cfd94e018c9e54e35a2df6f8 : SUCCESS in 57m 37s","accounts_in_message":[],"_revision_number":1},{"id":"5aaf222c5b1f1e70c9018240a642ebfa7fb5508a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-15 17:45:56.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nthis looks ok (with one note on the regex).  I think this should be expanded to allow for python2 being defined in global-reqs/upper-constraints but not in the project.  I don\u0027t think projects would be allowed to remove the python2 lines with just this patch as is.","accounts_in_message":[],"_revision_number":1},{"id":"b91228d9f595080524d93145a8a248b25f5f7ba1","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-12-17 06:22:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de65e6db6f2a664d9a668d99cc058063d1f8acdb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-17 12:44:54.000000000","message":"Patch Set 1: Code-Review-1\n\nStill needs to be rebased.","accounts_in_message":[],"_revision_number":1},{"id":"cd05606d6ffc75bd74f5a2aa338092dd427659d3","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-12-17 16:27:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d5f0b8c96bd04bea9a57b55861cf2214280cd3e5","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-12-17 16:28:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"55b3f758c3030969307eaa080ee9a64e04b313cc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-17 16:46:53.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks like this is working as expected when I try it locally.","accounts_in_message":[],"_revision_number":2},{"id":"58c8c22eb759dc4ad1cfa5a02a6dce90f828afd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 19:13:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8927b984d76942f0a434e74aea037b24 : SUCCESS in 4m 34s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5e7c81bcb0214283b64f05e216edad36 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c95c8cf3d6048e8b3c9b5e2cb0a8b3e : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/67c8ab2c34b9408786bf7a789c2f8a2e : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4513491d78934e2697f96a49696d953a : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a162e76874f4387a66c9318cf2b0680 : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed896fc45691413395777391e7d3b187 : SUCCESS in 4m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a20d18870cf24abf910928edc3a9dafd : FAILURE in 1h 25m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e7e2ede6ee1f476a815606e57e43567e : SUCCESS in 2h 06m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8405889ee10b40898b24da21807f7945 : SUCCESS in 1h 00m 18s","accounts_in_message":[],"_revision_number":2},{"id":"ae931a16bdd879a8e224acefff89a8cf9cff57f9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-18 15:31:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6f686820d122ccd5b325e810117a34a763a95188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 17:17:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74fc2932b00f4c16b2b15bbde5264f64 : SUCCESS in 5m 52s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a6b1ec33fbc649c2a8895b383a04da90 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19e4fa73aee342e6bc174b5566f15cfa : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7af59497d66f4ce2934df4cbb67270f4 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c5666e5bca24761a14d0d2e6bb3628c : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9603b0078ad64a03b048d5dc3344accb : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad442afb57654795917acabf1fef2dc7 : SUCCESS in 3m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/422c2f7109a84ffaa5b92231d86a2f1e : SUCCESS in 1h 07m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7264279d4e264ed8b24154e32b3c3f5d : SUCCESS in 1h 24m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9cd1034d7aae4ff5a4af52158f202ebf : SUCCESS in 58m 09s","accounts_in_message":[],"_revision_number":2},{"id":"6fa699a9ae4e73d076a88a4b5332c39921fd432d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-20 01:00:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b866caf7fa243a576bcf362eb4597262ba348bea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 02:25:32.000000000","message":"Patch Set 2: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"84f40a9d48fd5477da14d5580f40732d92112e0f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-20 03:16:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ba15f77363d942f99d96858ab79721b262826038","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-20 03:16:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"afb1a6bd55c03357d6a25e9f9cdc274a8aa474c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 05:07:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87fb08a061fa40439ea43b0d1e66608d : SUCCESS in 3m 53s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/52b81cc8c3d84c8d9770b15a46595f68 : FAILURE in 4m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a58d75934b64e6580842b10166a83ef : FAILURE in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/68601799bd4f4fc1af1ed47ccc93641f : FAILURE in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89f15c03c69b47888f6d22f523e3e04a : FAILURE in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29ed584d94c64885b5a0a00f91e41698 : FAILURE in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/026203c13e4e4357bb8020a00eb4ac76 : SUCCESS in 4m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f62c16ac3fa54a8a9f5ed42125dabe13 : SUCCESS in 1h 17m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cb4aa9d1be2445849900dd58f3841cc6 : SUCCESS in 1h 48m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ca8337ed2d3546c6b53017f0327295db : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":3},{"id":"3a836f7ea63d60277256544bd7169d859444b3ba","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-20 05:22:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"543746a8cd573c273dbe69c6878c1cab06ad881d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-20 05:23:27.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b160c2b7b15f157cfd79e89beebe807b7eb88b3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 07:53:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e049d88c378d4dedafba3e75e927d86d : SUCCESS in 3m 31s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0c6eca8eab37403eb8b72846f5697b26 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ab4ead2605c481faa36806efd09c05b : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa9aee8f4a2b4e0d8b979aadff140108 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1905482a1a74483592e6744f9d4c538c : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba213624b96b4de1ae15690552670aa7 : SUCCESS in 5m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a9a9e71d23f45ad85301fc83b8e38a1 : SUCCESS in 4m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09972213f7db405d9cf6889109cc9bbf : SUCCESS in 1h 12m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e07f7c033cd546108941496c0209de5b : FAILURE in 1h 16m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fd15aa98ddb84ba3aa7627167552aa5d : SUCCESS in 1h 39m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/00b3d888f7974ff18d849ce964d3b798 : SUCCESS in 1h 07m 19s","accounts_in_message":[],"_revision_number":4},{"id":"e7d24728a8c624740cb9af29c8b29c1dc04d4dab","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-20 13:09:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"33f6f9137609cf7bce329e75d4618b39f02a4c74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 14:48:45.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da833b516ba04030937b05b5a2ceb527 : SUCCESS in 5m 19s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a8406e226e3847aa9f1f97138ffbf5d3 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a92e5b2ab8640399a45c0da27793478 : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/96532d2260024566a2ceace28ee9aa24 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19a680e02ed5404a9ec9a146c60b62d9 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a05f18a99eb4a7aa11ad6814179be4f : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e2cea4d6d4e6433195cae34f0ac01397 : SUCCESS in 3m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae0a631434cd4939954c0aeaf9afd15e : SUCCESS in 1h 01m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5da870fd9eaa41ba82afe352b0556e18 : FAILURE in 1h 14m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/49d32575fc554600befba71285f6463c : SUCCESS in 1h 31m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7adc5178515543c3b55fc79d433e8dc0 : FAILURE in 1h 06m 18s","accounts_in_message":[],"_revision_number":4},{"id":"581185c98cd12309c90a09c80c98f0b912238769","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-20 15:56:35.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"b20ae1f17b10bacad077561bdb1559ef78a6dff6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 17:24:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a291ed07419490386134437a58f9baa : SUCCESS in 4m 08s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e7411750f9d64116b930cd7f9898575c : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dd7b8cfdfba4cc5ba91a2d1ebedc2cb : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a571e071177d424fbebd636662c01cf5 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4313b01d2302464ab0d71bff3554ff84 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e9eb13358eaa491c8b86b63a21efb362 : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7ad6d499667d41ac9b4acd5198ad0f7a : SUCCESS in 4m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f63ed1437b9499f8d3f863269835130 : SUCCESS in 54m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e9587ccbc005456b8b188f9ad7d1a7a4 : SUCCESS in 1h 05m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e1ff8cb9de9940e6a72032fc6f040a80 : SUCCESS in 1h 26m 09s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e255165596d94cbf9c411df093c0e39d : SUCCESS in 1h 03m 30s","accounts_in_message":[],"_revision_number":4},{"id":"a677e41da4f4ccfe2a1a776c039159e0805d8f21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 17:24:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7a397aaf47ebb6b953ef8961082932b378a6c7fe","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-12-20 18:32:11.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e948b5c43985c5aaadd6b9b6de29052f83260318","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 20:19:28.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e4d514fe1ea47638ee43ff4705b43b2 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4910cd34a9a44c97b4bdf1c9c86eba1d : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5c9ae248d0fa400eada65fbbccee92f5 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f076e1ff6ee4e14b56986e8b43306b0 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f51eb38a25fa4ae69094f422c836ec28 : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/23d85a78e9584a4ba17cb23ccf8bcb22 : SUCCESS in 3m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2071a93dafb424784ecb85c5aa7224c : SUCCESS in 54m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b810f9ddae344e04bedbe83a554e0a08 : SUCCESS in 1h 12m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/683b40074dc34ce7bb6d6a3d1e8bc4a6 : SUCCESS in 1h 23m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d2dd74b18306412f8b0d162570a8bbc3 : SUCCESS in 53m 21s","accounts_in_message":[],"_revision_number":4},{"id":"d76e2a6ea932130fb8c59b4c11fa5e4deb8bd979","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 20:19:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"5397bb702608ad4b8cedc7d87af8836dfd0a31da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 20:21:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ade3ed91f6145c5ab15b87353cfbab5 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a6a75d1f75ade37ea0c0d449d5f36c085ef6cb41","revisions":{"2c7b3af360f33494bed1975caef001134bb45d26":{"kind":"REWORK","_number":1,"created":"2019-12-13 16:48:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/60/698960/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/698960/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/698960/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/698960/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/698960/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/698960/1"}}},"commit":{"parents":[{"commit":"129c378b066183b9644d5bce992b0138ccc84566","subject":"trivial: Variable rename","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/129c378b066183b9644d5bce992b0138ccc84566"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-13 16:12:47.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-13 16:47:56.000000000","tz":0},"subject":"Allow projects to drop \u0027python_version\u003e\u003d3.x\u0027 version specifier","message":"Allow projects to drop \u0027python_version\u003e\u003d3.x\u0027 version specifier\n\nFor projects that have gone Python 3, having to keep the\n\u0027python_version\u0027 specifiers can add otherwise unnecessary noise. Special\ncase these until such a time as Python 2.7 is history and we can drop\nall these Python version specifiers.\n\nChange-Id: Ie6dba8bc78dcd5fa95c7ac15b0a3770cdef8ad95\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c7b3af360f33494bed1975caef001134bb45d26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c7b3af360f33494bed1975caef001134bb45d26"}]},"branch":"refs/heads/master"},"06a931344e512d961ee1430f359b6dafb7beb4c4":{"kind":"REWORK","_number":2,"created":"2019-12-17 16:27:48.000000000","uploader":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"ref":"refs/changes/60/698960/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/698960/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/698960/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/698960/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/698960/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/698960/2"}}},"commit":{"parents":[{"commit":"698b571749982f61e65814ca3de489177c14d30e","subject":"mark regex as regex for pycodestyle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/698b571749982f61e65814ca3de489177c14d30e"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-13 16:12:47.000000000","tz":0},"committer":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-12-17 16:26:28.000000000","tz":-300},"subject":"Allow projects to drop \u0027python_version\u003e\u003d3.x\u0027 version specifier","message":"Allow projects to drop \u0027python_version\u003e\u003d3.x\u0027 version specifier\n\nFor projects that have gone Python 3, having to keep the\n\u0027python_version\u0027 specifiers can add otherwise unnecessary noise. Special\ncase these until such a time as Python 2.7 is history and we can drop\nall these Python version specifiers.\n\nChange-Id: Ie6dba8bc78dcd5fa95c7ac15b0a3770cdef8ad95\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nCo-Authored-By: Zane Bitter \u003czbitter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/06a931344e512d961ee1430f359b6dafb7beb4c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/06a931344e512d961ee1430f359b6dafb7beb4c4"}]},"branch":"refs/heads/master"},"9036dc2656d4d9a2b68a0a3eaa47fd5913763882":{"kind":"REWORK","_number":3,"created":"2019-12-20 03:16:29.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/60/698960/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/698960/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/698960/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/698960/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/698960/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/698960/3"}}},"commit":{"parents":[{"commit":"12aab4eec9bbf3d1548c369772c12cef69845a95","subject":"Merge \"trivial: Variable rename\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12aab4eec9bbf3d1548c369772c12cef69845a95"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-13 16:12:47.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-12-20 03:16:17.000000000","tz":-360},"subject":"Allow projects to drop \u0027python_version\u003e\u003d3.x\u0027 version specifier","message":"Allow projects to drop \u0027python_version\u003e\u003d3.x\u0027 version specifier\n\nFor projects that have gone Python 3, having to keep the\n\u0027python_version\u0027 specifiers can add otherwise unnecessary noise. Special\ncase these until such a time as Python 2.7 is history and we can drop\nall these Python version specifiers.\n\nChange-Id: Ie6dba8bc78dcd5fa95c7ac15b0a3770cdef8ad95\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nCo-Authored-By: Zane Bitter \u003czbitter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9036dc2656d4d9a2b68a0a3eaa47fd5913763882"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9036dc2656d4d9a2b68a0a3eaa47fd5913763882"}]},"branch":"refs/heads/master"},"a6a75d1f75ade37ea0c0d449d5f36c085ef6cb41":{"kind":"REWORK","_number":4,"created":"2019-12-20 05:22:48.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/60/698960/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/698960/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/698960/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/698960/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/698960/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/698960/4"}}},"commit":{"parents":[{"commit":"12aab4eec9bbf3d1548c369772c12cef69845a95","subject":"Merge \"trivial: Variable rename\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12aab4eec9bbf3d1548c369772c12cef69845a95"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-13 16:12:47.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-12-20 05:22:40.000000000","tz":-360},"subject":"Allow projects to drop \u0027python_version\u003e\u003d3.x\u0027 version specifier","message":"Allow projects to drop \u0027python_version\u003e\u003d3.x\u0027 version specifier\n\nFor projects that have gone Python 3, having to keep the\n\u0027python_version\u0027 specifiers can add otherwise unnecessary noise. Special\ncase these until such a time as Python 2.7 is history and we can drop\nall these Python version specifiers.\n\nChange-Id: Ie6dba8bc78dcd5fa95c7ac15b0a3770cdef8ad95\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nCo-Authored-By: Zane Bitter \u003czbitter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a6a75d1f75ade37ea0c0d449d5f36c085ef6cb41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a6a75d1f75ade37ea0c0d449d5f36c085ef6cb41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
