)]}'
{"id":"openstack%2Frequirements~698981","triplet_id":"openstack%2Frequirements~master~I10837f52a206df835e91e648cfa7a1b23efae28d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I10837f52a206df835e91e648cfa7a1b23efae28d","subject":"update constraint for ironic-lib to new release 3.0.0","status":"MERGED","created":"2019-12-13 17:53:49.000000000","updated":"2019-12-18 05:05:10.000000000","submitted":"2019-12-18 05:03:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698981-1576645404811-0d4ca9dc","meta_rev_id":"30cfb4f0b89216d96968756c5eb5ea9d05f7ab34","_number":698981,"virtual_id_number":698981,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-18 05:03:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-14 00:59:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-14 00:59:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-14 00:59:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-18 05:03:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f7f16a045f34e08bb4cd4159df5a690cd05fcd7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-13 17:53:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73933fad410e5d74a4ed286379c1f59629f6eeab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 19:38:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f89a50bce9c643e88aab1f06e17c7e00 : SUCCESS in 6m 08s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6ac89e58adab4ab89c3d82181e0ce1af : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae76edf3c8a84c11b4d6745279a69cf9 : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/881b577f948843f48e2c12295d9b8a16 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1582063542cd4a639e9bce152ec219bc : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f8a84727b4f47f5bddaa0f148720f1e : SUCCESS in 7m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0857a57d85034837a5a2a112fc541200 : SUCCESS in 6m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b24f964b4af947f6af9fde53cddce67d : SUCCESS in 5m 33s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8ce09ccf0a18400cb8c24a5fb721fbde : SUCCESS in 12m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/674b57137a01488783656c9bf16cec59 : SUCCESS in 13m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/00f3dd744e7647c28a7449cafd6bce36 : SUCCESS in 11m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2d7d8447b42941edbeb7700868fd0a9d : SUCCESS in 8m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dc165f377bbf45fb94963006e356452d : SUCCESS in 11m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a95002c471e8477aa22605fc643f0374 : SUCCESS in 10m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f8792a456ccc490280e0b7681002d7e8 : SUCCESS in 18m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eacdd5e1998d48d88feba6d7b436a2f6 : SUCCESS in 24m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4458d45660274c9f8775c8b8fc031876 : SUCCESS in 15m 19s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b35e3c9f973b4374a054fd3856c20316 : SUCCESS in 18m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9471221e7fb24b69bba83d59cddd176e : SUCCESS in 19m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c38489f0555d4343a80fe9729d8eb9c1 : SUCCESS in 7m 08s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/fa59fc6aae514d06bf354c1c109eddda : SUCCESS in 8m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a1ad521ef340406f971e8f5d83fcb5af : SUCCESS in 7m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/05296f3a1cf4458fb9ad5f43e0d8d63a : SUCCESS in 10m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/304fec89346140ed92663336c2092973 : SUCCESS in 12m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4126ad137fab44aca2ca3b6ec4e70014 : SUCCESS in 1h 08m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d3227e6dd6844e6283a03c611e0a1c8a : SUCCESS in 1h 36m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/95d5c4e2c1004193be39a4e640df896f : SUCCESS in 1h 03m 15s","accounts_in_message":[],"_revision_number":1},{"id":"ac9fe7757b5f7e25971eb7d66cf440475051847f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-14 00:59:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b4ce51926b0a194583dae8f4e49010d0e42ab5eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 00:59:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fd5a5df832a4c13e7887b757ff34b3fb8e2a3bfa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 02:31:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29281f3129394059ae4568e8286a8188 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/861a6ba0a7d7412fbcb50f7535e3dd8b : FAILURE in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c89cf79b721145b5ae4a51f6d4f5a587 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26496d06225344f9a4949d381b473373 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30674182bf8448c6a906a78892a5fe79 : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c7fa16b1d15d477d82f71c532930d9d9 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3194e36f7cd344a1be65fbc467fcf854 : SUCCESS in 5m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6cb835a2b246456e9b2632f9053d81f5 : SUCCESS in 7m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1435664920e44ec981436f3b248e045c : SUCCESS in 7m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b02c906f524f4075bed851b9c54bc7ba : SUCCESS in 10m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2d1ca31f4e9d4fca888fcac41f1a9d98 : SUCCESS in 5m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f8ee5285c2614cc0b288c209bb88f149 : SUCCESS in 9m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a129f597a4ca4578860dc842b162cb81 : SUCCESS in 7m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e7211cf77128467490bec1c3923ab885 : SUCCESS in 15m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/928f27c6b9f84fdb9ba211c1c5fc062c : FAILURE in 5m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/474eddaa6d26429bab44da759137caa5 : SUCCESS in 15m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a04036b104224019b2f205e4bf8a9611 : SUCCESS in 14m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3b3e7a6a7e404f28b31246ac917da4a9 : SUCCESS in 16m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a6c2ae5c0b6b4c92b09f06cf0766f941 : SUCCESS in 4m 48s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/116abf31076f4155a69cc557a8073832 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e30ab6d126144530b44fcd2ecf8b6345 : SUCCESS in 7m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/779a9577eedc484dbd4726e447368773 : SUCCESS in 10m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f9fcb2b4b7304c43956abff9d089093a : SUCCESS in 10m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9237a156df3047f79006a8a59a1a6d77 : SUCCESS in 1h 13m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/08a75d3190574ec0bd3cc9a6e64f6cba : SUCCESS in 1h 18m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d6190ff06a194a9bbf412c1d0689efee : SUCCESS in 49m 45s","accounts_in_message":[],"_revision_number":1},{"id":"7d3167fd596adc95fca50b3b0b4ffaad3227bbb3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-14 02:32:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0b5e27b2e1a581a07b5d5cf460a4483d1ef240e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 04:13:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be7c112ef10245718cf64e57e610be18 : SUCCESS in 4m 05s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/89a14c558bf64df1a18021b75481dd81 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e66cce06251a451d90b9c7df987e7335 : FAILURE in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cba294be938f4ecc8df6e7428dc9bba5 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0874fcb7f81f45eca84de30c6a4c1b66 : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba97de124ac44acd87671c808017a870 : SUCCESS in 3m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ee46b3cdcd1849ca8b058d55a71cde33 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1929c0cd90334755b4f23c30ea8d691e : SUCCESS in 2m 55s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/57116a022ef54c5b88b01270f4181de5 : SUCCESS in 6m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e400254280c74ebb8665f801d0bde7a8 : SUCCESS in 12m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a09bda24f63241ce843c437464598581 : FAILURE in 8m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1dbeece1a66e4f03b5f0364a734979db : FAILURE in 9m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d8b4a2d36ee7487bae940b770a50a13f : SUCCESS in 7m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a51849e79b9d498a9561ac92ec41c226 : SUCCESS in 6m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2bb0a31ac28a47ec891df45950d990a0 : SUCCESS in 12m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5c1e1f1a1123475b8ca8b045f3bdc6e9 : FAILURE in 9m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/60d6cfb391a9433aa2de07f00d31a05e : SUCCESS in 14m 21s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9d948b1e5b324b01bfdb0979afe1971c : SUCCESS in 17m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5b3a525d3beb49679fb4add47d13d9a0 : SUCCESS in 14m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ab422bd6b06847fea31439d99d1cf5d2 : SUCCESS in 4m 36s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/78a741bbdfc2476cbcd3a2c153619670 : SUCCESS in 9m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b604406bfa94815a229c484ba36f616 : SUCCESS in 6m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ba9c548beb654e649c58dca2742b5b83 : SUCCESS in 14m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1b7667e10fa647e48c9b91bf749e4127 : SUCCESS in 11m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/179e32b071054c3b9d17d596e5c8fe96 : SUCCESS in 1h 00m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a949395ca31547c98fe58386dff0a768 : SUCCESS in 1h 32m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c98a2770cfb7470c88c3b716d1d9e74a : SUCCESS in 48m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ce70f19c17324975b29acf19fa4ee3ed48d5df43","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-14 05:07:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6eb07367ffb305c8836a2f9042ff32f4495887dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 06:33:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9d2174c150d462dbb181cd2c52adb0b : SUCCESS in 4m 02s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d65cbfeb0c3848d0bd11a93559623ebe : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f7a60ad13944747894368652b280311 : FAILURE in 2m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f340a7f6467453c90b832e65976afff : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e03a63062ea406b8f24457aab378cee : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/39231a041f2b496cbe451891af9ffa97 : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0dcdbb78d64442deb809237a64c991f0 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1d74ddd7112c43fabcf9e563a32f0890 : SUCCESS in 2m 45s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0d45382a92354a9bbe0f7f9a58d9616a : SUCCESS in 13m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/91ab0e436f54435b9c3c0369911ddced : SUCCESS in 6m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1e019d72ea794bc89a2148194fd497ce : SUCCESS in 7m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a88c9e39e6f0421488dc1f3e956cc6a9 : SUCCESS in 4m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b63bb7fb367d47219a4725bf97df3a10 : SUCCESS in 9m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c94f6c99caca425a861e0ff4da03a448 : SUCCESS in 6m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/70ae31acb26341968bff756a5625c75a : SUCCESS in 20m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/63d4a120935f48209817c6d8117521ff : FAILURE in 4m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c0960099dcaf476781f881de7cc84b53 : SUCCESS in 11m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d567a924c74842889547b2b30856b6f5 : SUCCESS in 12m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b3c75661279248958440b79be8b5c3a6 : SUCCESS in 13m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d157e1559aa143648405eecfb4a46337 : SUCCESS in 4m 21s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/83f250c39f974305b6803d58dedcaa25 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b5ef13a7fa54efaad7647ff6255f2c0 : SUCCESS in 5m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/147684b7fbdf4bcdbc34ab1e23f24ab8 : SUCCESS in 8m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f9f58d6e6b8745f2ad2a28037c03dc67 : SUCCESS in 9m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e476df302afc48c1915e8eb0fb8f4414 : SUCCESS in 1h 11m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0613b3e87a15432183fb6168d7f983d5 : SUCCESS in 1h 24m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7edf353cf08144629d708b693884e3a4 : SUCCESS in 54m 13s","accounts_in_message":[],"_revision_number":1},{"id":"fe6b07293eacd72656e68d0007785dc9b53177e9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-14 18:02:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cd723d08ec570207899de4d1c549812453e109ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 19:05:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4792865be3d4460780c5a207429d4001 : SUCCESS in 4m 29s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0483196ed6be49469b1ca64b02a9a566 : FAILURE in 8m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/942b3f9636f7413b8ece5c98accbcb37 : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1439f2d90bd44de18d979ce03bca9ffc : SUCCESS in 11m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e33e7c34519948739c3e9d32a482d626 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c28d89c30ef24d1cbafeba28a4dc6be8 : FAILURE in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/49aff1e221084c3eb84cb587e519a927 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6757e95e23664d899e272d4d07d36602 : SUCCESS in 3m 24s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3560a5dd21d040308142a5c070f3ac4b : SUCCESS in 7m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ec73220ca0ad48dd8fb7f2d92f447e13 : SUCCESS in 5m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/065a78649f6944b48ed6d2132bfcc458 : SUCCESS in 9m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a6c42935c8af4e279a18374872ce8f47 : SUCCESS in 4m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2a4ab80862d54ef8bd877b96789cc4c8 : SUCCESS in 7m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/95fed4c58277459bbc7be10afbea6afe : SUCCESS in 15m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3cb3e435da4d45408edf99a2441f242b : SUCCESS in 23m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9397547a1353413292c53ecfa0beae5d : FAILURE in 4m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7ce502c606fe41f6be9b32970f8943a5 : SUCCESS in 12m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/323be5b0722d4c2da874f12ad918aa02 : FAILURE in 14m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ef60403cdfc74d2282d2c3bca9d1feef : SUCCESS in 15m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3fd11d2b40a14eb6a2f494d7355ff10f : SUCCESS in 5m 06s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c7b931611b6e4bedb795ab1348ea9b5f : FAILURE in 3m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a781c467ec804e9d97730a4a7f88a225 : SUCCESS in 6m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7851ba12c9f2478bbd14170036e36474 : SUCCESS in 12m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f01421c6185442f8b9e4ebb3c57bb7c0 : SUCCESS in 9m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e55ef3543b540e281aaf18c1efa28fe : FAILURE in 46m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ca4db44da7654e66a568f821c0fd8528 : FAILURE in 11m 09s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/61bf175fb2814be2ad857106319ddbf3 : FAILURE in 49m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ef2fa2ad07b09c75a5034cb4a0bddcf4f872811e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-15 17:33:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f6251ca370ec755898fa105ec2ad30d648326b18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 19:13:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d84c60dd24f8423282a7b20da55c78ce : SUCCESS in 4m 58s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/55d0f546281344bb882266a259e2832a : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afa8d8c1185240e39c073f736aa61e8e : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d8973318efb437da2257caa935e6763 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24b8b95e0f084cb0887a057f76c241e9 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/31a88b4d656e4a9d88c99ee676dbf3df : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6672135de74240efb57516a5733b1ab8 : SUCCESS in 6m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8ba16b0b1a304c2bbb929d9570bc7930 : SUCCESS in 4m 32s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b5c440de5ac54f6bb72fd56ca18d75e3 : SUCCESS in 6m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7d545470809640f0b9dc478b66fdeff9 : SUCCESS in 6m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/70a767d2d5794b36892fe65c0378e9ea : SUCCESS in 9m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6579effb500e4515b7784137e0d65927 : SUCCESS in 5m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/213195bc07f7404d84656d52fb68863a : SUCCESS in 8m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/745d3223e9ea4d22a03e0b39cad08881 : SUCCESS in 6m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3551760cf1984e45be401718dbc50fed : SUCCESS in 11m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/40f4467db88640a89456d6e92c0d6dbf : FAILURE in 6m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/950c80f84c75495db3d60a3daadfa182 : SUCCESS in 13m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/dfb4c9cac1564b838b2ecaac32c60c16 : SUCCESS in 13m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ae87a10ba3e944a4b4670495134a60fe : SUCCESS in 21m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ca49f4bb1bb8465ab174e7860a657670 : SUCCESS in 5m 16s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/41e3d65b89a64dc3a4d8abfeef72beb8 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f1cea7be3ff41b08c1aea5edf1b2f0e : SUCCESS in 7m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/35f1a8ea7ee648dfa3bf7ec87897435b : SUCCESS in 11m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0f610d8f69c84cf79765c16c3e0f2c0b : SUCCESS in 10m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/98519816665848dcb7a1fb9b949b5d16 : SUCCESS in 1h 03m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bb0e3417c15a4402a853cfd90c6518ed : SUCCESS in 1h 22m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c564499d6b6c47f89df4f29f3ff9df1d : SUCCESS in 56m 36s","accounts_in_message":[],"_revision_number":1},{"id":"dcf7f3f5df92302c2c226c7cb6dfd3c33ab39cf0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-17 22:40:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4a9c3d820ab9a246b4c2d6cb98da253929257ed5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 00:08:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0f60f3d2cc34afd802c08ee4c8ba5c7 : SUCCESS in 7m 07s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4f14f55b5b1f449eb337f55cd87b6439 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d63488d800c44cc486ac8885bdebc550 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bdcdba44e8a949dca9d048f75295abef : SUCCESS in 10m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6059b4155f584a4385498e0740aa662e : SUCCESS in 7m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ce4a7f182e243c28b887c9e15f883ff : SUCCESS in 7m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2671cfc0f6d1407398d59550eec07e62 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a2354dcab99941878cade8fc9f788f80 : SUCCESS in 4m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/eb2293d46105401a88c086f9de7ff6d7 : SUCCESS in 20m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5401cc0a3cec453396b0db296d84b711 : SUCCESS in 9m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/226b58638bdc4c6b8027106330727bfc : SUCCESS in 10m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d3f41474e9a34e9ab76b016d540c2493 : SUCCESS in 7m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6cacbf3a75544b8e8f00dc1f0e16882f : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1920cf7740124a538a097fa5e7433f2b : SUCCESS in 8m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0b7b1ae64191495a9f5a816228e5ca57 : SUCCESS in 22m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/44eaeddd437c4ec1b3454fe0706fb044 : SUCCESS in 19m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d159dae430514676bd6245555b34a509 : SUCCESS in 16m 55s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/65e48d3ab0a24ae3a2acd1fdf54e665f : SUCCESS in 20m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/38c9bc34e2f44071b39d379d1071d909 : SUCCESS in 21m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/dbfbed13003545439e6614a574488412 : SUCCESS in 5m 02s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7cf8de9e892b4e8fb419597eb395fd9b : SUCCESS in 7m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/38ab1b1e508543c3b9c852c4a2781942 : SUCCESS in 8m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a6e8e2bbe02e4f98bf940b6f6a3010f7 : SUCCESS in 11m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1bec1ad3739649a4ba8698b3368427e7 : SUCCESS in 14m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/62082d2fca324cf28ef6952d736aab07 : SUCCESS in 1h 06m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6d936e7c609c451590f77dc5e0804d9a : SUCCESS in 1h 26m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3a31f67562244195a0389cbbf7f337f7 : SUCCESS in 1h 08m 44s","accounts_in_message":[],"_revision_number":1},{"id":"00aa007d2bee08c3cf73c1dedeb619aa3c5826ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 00:08:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee291d1672383f4e7fce2d7a9c0390d05654a2a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 01:46:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d93cc3d15c04de2ad66bdfe482a0e37 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f25b8e0aeb0841b1b1423b6978bd9311 : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/04d4df8d8f2b4d74a58cf8cdb7bdb757 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71bdc61aa9fe4dafae491aad18b2542b : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41ad4664c4be4481bc89a48d48b5746b : SUCCESS in 7m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d48e5ee715d94f588ba3479b87a32a1d : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a17cef98317040959c4620cf32460c44 : SUCCESS in 4m 42s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b6d2392a25af4e618873cb220ecd240c : SUCCESS in 10m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/39d0113a6a4941f0b176e4e748aa4648 : SUCCESS in 12m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/09ad3b44be1943e194762f8fafde206d : SUCCESS in 11m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5ec9dc0c98244ab6a54f0d94ebb9bfb8 : FAILURE in 6m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/56e3b942efa64acdb43f974938b86d4d : SUCCESS in 11m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/de8ce20576f8423bae525ee528d4db8f : SUCCESS in 7m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fa1f442b62254ae199ebf8a7680140b8 : SUCCESS in 21m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/370267f2ca704ebeb79df0330feae1c8 : SUCCESS in 22m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/24c7fe56fa4f4b5993469d7ff7d9e3bc : SUCCESS in 13m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0854a03cf5ff47b5a3e5455dd1b326b2 : SUCCESS in 19m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8f32f88b0b7e48adad56b5a12f660bf2 : SUCCESS in 18m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f64771397a914166bd5bfe103d7c4348 : SUCCESS in 6m 55s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0b5ce163de94477cacab791367709ea6 : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5e12dceec1464ee8abd0653090ebab70 : SUCCESS in 6m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/044b0de30c58474dbac0c8fb920d1fe0 : SUCCESS in 12m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3abd59ed5b72492598ff075a2b7f2c5b : SUCCESS in 10m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e527b9c4af745ddbbb98975735991ad : SUCCESS in 1h 31m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/76c8a93c2c9f4b8a8ec81f74dada3b49 : SUCCESS in 1h 31m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/be9552c8eb1c4691ab809353aa429fee : SUCCESS in 57m 26s","accounts_in_message":[],"_revision_number":1},{"id":"fefaadc79aae021dd3066e5c6d8782b5fad7f4ad","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-18 01:52:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4bff538e46d9060849c0a0a6744dfe57c073a956","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 03:23:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c08107e99f31439e82ee4231ce35f494 : SUCCESS in 4m 08s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c526ab1b947d483480e6a58591cec5fe : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4e98a457b3f4fd186991359193370ec : SUCCESS in 3m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8b840cf5d2548fb836a48cd1524b0eb : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2c419cbd7a6463886ff346807707b22 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b953b6da88a4f6b99d15153a2186a1a : SUCCESS in 5m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ba28cc3e3b947f488e7ba9c84f40289 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a4b7ea21027c42c6911d3ff12ecf0af1 : SUCCESS in 4m 02s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/91292fb123fc486484d5534235a82467 : SUCCESS in 7m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0c8920be9d324f1080747d512e780a1e : SUCCESS in 8m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b6b64c74b0fa47149c28c09a973d71d5 : SUCCESS in 8m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9021582d23a14bef822aba90d209bfa6 : SUCCESS in 5m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bc555cd6a6694f309860b31fff5cd90e : SUCCESS in 14m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a549d245a89b499aa60699b6c02f67d9 : SUCCESS in 7m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dc51c8c9d55740e7bc6e3ebeca6c24b4 : SUCCESS in 14m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a85e806a54af4bdfb16a562da78d370f : SUCCESS in 22m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0082aed471fc40cab4893ab2cbf554d0 : SUCCESS in 12m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9f0f8468552f4ceeac574506b5c457df : SUCCESS in 12m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/529dc34de00c4601914e1da0f3a06de2 : SUCCESS in 20m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/af1255fee00847868cde4fb513f5f437 : SUCCESS in 4m 21s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/20d7e51f34234d33a4709329737cbd8d : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4ab299bcc52e460ca4f8097da99fda43 : SUCCESS in 6m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/92a2976a6a9240bdbcd09568f895c2ef : SUCCESS in 9m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1a55384b9cf343909ac066228920d876 : SUCCESS in 9m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b00400f9e52c40d4a9d0a653dabe5ce1 : SUCCESS in 1h 09m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/68018433d82d49d6a4431ef80d94d5cc : SUCCESS in 1h 26m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/02734392833f415e85eb96ded05ef6d6 : SUCCESS in 59m 41s","accounts_in_message":[],"_revision_number":1},{"id":"9701afb2e149fbe2ca1e6fb2ce89159095fdb1cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 03:23:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"568eab60aa76a4f65fc6462144ff31ecdac95902","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 05:03:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"71793dc6fb413a04e9b5e55a1f9202538f2de4e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 05:03:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78771ccc6c7c4c2bb61dca89731a2fad : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae51cdf3fc81429eaf1f22c3c34c87ca : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c800445ac10b45d3ac199874f8d154e2 : SUCCESS in 9m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8aa8f076eb664bc092116eec2ada48ea : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a271f5bf6c0545ce920c74c8a1c0f739 : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/52bb0fd8f63b4e2a86e32c27ffaff181 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c248c770b1b0479b9ed893f9128b454a : SUCCESS in 2m 52s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/12fd97b22e394f79800da7eaa913d1f3 : SUCCESS in 6m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b050fdd1f5ea4bdb97bed42a27cdcdf2 : SUCCESS in 7m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/33cad9a16e854d7486dcef58528074b6 : SUCCESS in 11m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ebd959d4b808439bab8503484894053b : SUCCESS in 6m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7a689549fce841c5b40be359aad1eb6b : SUCCESS in 8m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/de4a31801bc6419395246c831ef9aade : SUCCESS in 9m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/97b26c9f594b492c9017c0a0d3a6afc9 : SUCCESS in 12m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5fb608b2c8ee4be8a143a1b93f5c3aff : SUCCESS in 24m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/be09486449ed475d9abaaad9ebf9522f : SUCCESS in 14m 10s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/444ff5b79c81413dbff8d4791d3a9dea : SUCCESS in 13m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/565fb68b6d6344d6839604e30466495a : SUCCESS in 17m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/135fb337adeb475d938010381833ca1a : SUCCESS in 4m 13s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/be546da81ffd45bbb324917f63f233fe : SUCCESS in 8m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/edfab11de08a4028be087459637dacba : SUCCESS in 5m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/791aa7809f8b42d6a2ad4f6808dacc63 : SUCCESS in 9m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/785c045797d141d5a51f5d0a2e52f4aa : SUCCESS in 8m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7aa775180856462f80bab51151a89b85 : SUCCESS in 1h 06m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c8e5cb3d94854616b91798f26eb88b1a : SUCCESS in 1h 33m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2e0d6c6746cb49538e3ab5c400bbbe85 : SUCCESS in 1h 03m 58s","accounts_in_message":[],"_revision_number":1},{"id":"91cf262e5d462dc262674cc6d4464f4d696181da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 05:05:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f50f368e543640b8bf4996b3625967d5 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"17ca0eb855fd771f73a74f218441321e8c184b03","revisions":{"17ca0eb855fd771f73a74f218441321e8c184b03":{"kind":"REWORK","_number":1,"created":"2019-12-13 17:53:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/81/698981/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/698981/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/698981/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/698981/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/698981/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/698981/1"}}},"commit":{"parents":[{"commit":"b26e50aa194594c03af1413b09d93ec0dc364494","subject":"Merge \"update constraint for oslo.db to new release 5.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b26e50aa194594c03af1413b09d93ec0dc364494"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-13 17:53:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-13 17:53:46.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 3.0.0","message":"update constraint for ironic-lib to new release 3.0.0\n\nChange-Id: I10837f52a206df835e91e648cfa7a1b23efae28d\nmeta:version: 3.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Change-Id: Ia31c1d992e76fbf21b6360232efcd27b99c3eb66\nmeta:release:Code-Review+1: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17ca0eb855fd771f73a74f218441321e8c184b03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/17ca0eb855fd771f73a74f218441321e8c184b03"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
