)]}'
{"id":"openstack%2Frequirements~698987","triplet_id":"openstack%2Frequirements~master~Ia61773d8595c546483c3adaa62efac8f6f98d815","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia61773d8595c546483c3adaa62efac8f6f98d815","subject":"update constraint for os-brick to new release 2.11.0","status":"MERGED","created":"2019-12-13 18:32:25.000000000","updated":"2019-12-18 13:37:35.000000000","submitted":"2019-12-18 13:37:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698987-1576676255257-70b46864","meta_rev_id":"2692a215efbb11ef9986a014351da91a8f5dafa0","_number":698987,"virtual_id_number":698987,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-18 13:37:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-14 05:08:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-14 05:08:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-14 05:08:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-18 13:37:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7f9535b50b963dfabbd43d9c61e544dd3d71df55","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-13 18:32:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f403e8f392871a20b9f732ae2f2387ab6d332c56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 20:08:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48eb4dff04bc4f9089ee6ad1726205fd : SUCCESS in 5m 18s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/92e3675630c94632aa448b36bdb20e32 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0193152374934a18874c943f671fc74d : FAILURE in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e4f6d098f26d469c89a9f6a9a60beb5a : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f97bbd9b1df74f0fab18ef0d375c01d1 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/48f1af81fcc249f2bd6ae9eca39885a1 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ae057a941ad476a874fc3070538b38a : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a0962e481e054e0cb09c60a9849f712a : SUCCESS in 3m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9d3d8bee1b0d43c98bd7d6ac5614e6b5 : FAILURE in 4m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/51efbc6a37f149068b0998bafbef605c : SUCCESS in 7m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/188bcb2cca0a47a0a14eaca3bbfe6874 : SUCCESS in 9m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7bd67a85cdaa484797c8ee2f1892aeec : SUCCESS in 4m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5846181a28194d25bbfd90b4754f6a77 : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1a24270e89814a4e8f3d51226be2fe82 : SUCCESS in 8m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/05e82fdffee94309a494c23095d2a86f : SUCCESS in 14m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9b72b71900e747d5b1b31c2ba9f55978 : FAILURE in 5m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a38841b2533844f6a11145e900f5431f : FAILURE in 3m 53s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b3733dc8ffd8466e9cd7632438b3c558 : SUCCESS in 14m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/47b9730e05b744438bbaf65689dc2531 : FAILURE in 5m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e6f5b0170b4d4ad190ad52e2429e25bd : SUCCESS in 4m 12s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ab45dbd4192241c6aa23a7761f40bbd9 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6b2d9eda6d4e4fc7803fca2ba50eeca4 : SUCCESS in 6m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6c3dd9c8fad0457b81ea2a9808406941 : SUCCESS in 8m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9ed2d43eb7e94435852b3f683b05a68d : SUCCESS in 10m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6269710d6b7a400583733160fd11c88b : FAILURE in 1h 18m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8082bba0a8084272855220a90c686792 : SUCCESS in 1h 34m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d4b93d0a380a48a79d45273019af83fb : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9817a989e179d91eaea23419c540e2d00b8061bd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-14 00:59:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"941adffaf09607db5e906c68c1f3d3760d45b3ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 02:30:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4505bcf55d134a3081f9849b7e643cdc : SUCCESS in 3m 40s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e9e0d2e13b304758a35cdb20228b5bdd : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdc021e844ff45a7ad672a1c9f31a012 : FAILURE in 3m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c0ad830bd184c76a51fe99eeea87591 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71bff113f14448f9ab8f87aa8209011d : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8f92f45465e46ceb4f7a3c9bde9417d : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd82a1dde5ef4b9191f2e1b29870867b : POST_FAILURE in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c847e711a3124acd946f89cdfbde494f : SUCCESS in 3m 12s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4b88b81806794a5faefde06d8a494139 : SUCCESS in 6m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a13054e976144d11acc0a8824618112d : SUCCESS in 6m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2363c1698c0d472aba03af7c0ab32c52 : SUCCESS in 8m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b63e212af0d34bbe9d698cf316d228b6 : SUCCESS in 4m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bd5ce89d5e07421db857f08f03e735a0 : SUCCESS in 9m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e32864a8da2241588402d255eb55e3f7 : SUCCESS in 7m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b7c83d891c5148b694af37cbf3aaf70b : SUCCESS in 18m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a878357821a14533a09c1a31054e4bdd : FAILURE in 4m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/319d065dc339495bb37f0cc56fcf0873 : SUCCESS in 12m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/69a22308c60048fd9d00b9fe1b1259bb : SUCCESS in 13m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b87b031012e447f5af52777fd34d4ae8 : SUCCESS in 15m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a02da04d05ef4a61ac6a2af43846d46c : SUCCESS in 4m 09s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6fc5325b98a0453081fa2772b86bf897 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/76aceffb1b66401687f6b3303a5f43e8 : SUCCESS in 6m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/18b9dcb35e0e4164b9adbf1788e0963e : SUCCESS in 8m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0b30c49766144c2fb5e3672d94d7c107 : SUCCESS in 8m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/adac0a7abf7b4310b1dd01b320670e2d : SUCCESS in 1h 01m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d3760e839e5e4606b01d6d0eda1884f4 : SUCCESS in 1h 24m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f4b3b611bfd94484b144893d0d62a877 : SUCCESS in 1h 03m 13s","accounts_in_message":[],"_revision_number":1},{"id":"863e51d6c6746ecd37cced4d6d5ace51a0ed3d9b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-14 02:32:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"69a25639f716418bdc7b6d511b0665e58198c32f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 03:58:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df096f26009d4c34874b6bfd97310197 : SUCCESS in 4m 49s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/77d34353143e41219ed58c229611be29 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce9e466b97984fbf8e20b2fc1902a004 : FAILURE in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cbf6249d6bb74103b83c185004f59fa3 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/143ca09a9d7646da8a3924b14da3cd95 : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/92c3a1d88dc940ca9e6efccfea16d8af : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/56a26317a0d04e828acf7f6edeb00ce5 : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1f47694581d842c1af1b6ea4a920e0cf : SUCCESS in 3m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a7c54b1767024fc59b1ab0916790964b : SUCCESS in 10m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/decd06a19ada4b70a6815531b2ff44d5 : SUCCESS in 9m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ece3cccda3da4c9f949776aeb64fe7aa : SUCCESS in 10m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/67c5b194822e461ca4d8e758be402443 : FAILURE in 6m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2bc73ea98eb54984ac4f0d505318e52c : SUCCESS in 7m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/909a21354de54d04b3f8ac9553d0ac5d : SUCCESS in 6m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a15ca3cc4f454815bd91a17ae0bff40c : SUCCESS in 16m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0b60b585c2ee43bfbacd74266ef3b1f8 : FAILURE in 4m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/04d9301651344c0787de9562253948cb : SUCCESS in 13m 00s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/34bddb37b40e417dac6a7ccfd87834f2 : SUCCESS in 17m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e2c1024c4dc54bda9b503ea2d9a46f8e : SUCCESS in 17m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a9a0c4fb003e4bd899805e77510da4e0 : SUCCESS in 4m 27s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2dd215eec710475fbcbd408eefb4cdf6 : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b04e1fc9bd32498b81f1614ebb816398 : SUCCESS in 6m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a5892d676c424b8ca330ddc070c3c3de : SUCCESS in 8m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ab8f513cf968429da967521f43a986e9 : SUCCESS in 10m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/18ec585221fa41c68091f707c94272d0 : SUCCESS in 1h 03m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e824e21d29324faab83f899a5dd2b97a : SUCCESS in 1h 25m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/95e6beb09c3240ad9e2fc5fea9e8938c : SUCCESS in 55m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b698d8c84b9d3cff34d28a7a64d009c48c15129d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-14 05:08:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c86e036e71aed079fff5b4fd06bc117e75e5e6e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 06:35:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cb81f17cdae49afa6dd17b0fc2789de : SUCCESS in 3m 51s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c5404f4ce1fa4543b0fb553796271ec5 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6583636291d42ceb8df60a476311f0a : FAILURE in 2m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f39bc63f1df34fda81e10f116d64b7c1 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a960d6a897046a3bd2b7d07c0a5af52 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bfd99ca1beba41eb870c5718e489eb7a : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3ebf2dcc89614164a4580a8610becad6 : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fe252f25bfae40d5b8590e9ec48f602c : SUCCESS in 2m 59s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f79259049aa94eba887a2cb3295a3ff6 : SUCCESS in 7m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1ef609d178c6453aa0f89235de82f141 : SUCCESS in 7m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/66c3d2a8edbf4355a1279796135500d8 : SUCCESS in 9m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0c56383abfb24ccbb45f272f7713c4af : SUCCESS in 4m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/588513a8b3ff45f6a77e29948b0c20f5 : SUCCESS in 8m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e7675330bed840ec9056efa6230cb285 : SUCCESS in 7m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cf571b04343a4817b0dadd1f9bebebbf : SUCCESS in 13m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/62c464ebdbe04e50a5a95853d5028217 : FAILURE in 3m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bb7746ed76754509bc2c8c77e6b9648e : SUCCESS in 11m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c2b3b41ad07a4f89818ea886baf063e4 : SUCCESS in 12m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc189e23ab8548ac8e93d4349a8df277 : SUCCESS in 17m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6916e7bee78b47cfad53556afbabb453 : SUCCESS in 4m 08s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c879a36ac5ee4fb086f45121064bdf9b : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c8fd234f6acd43f1b07cd589d34e17cf : SUCCESS in 5m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f8687c5438224c44bf2138552060a6a9 : SUCCESS in 9m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0e3ea87727094552ac9c3a3064903b39 : SUCCESS in 9m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b4cfba23cfc40bb92fe0aef6bf5e150 : SUCCESS in 1h 03m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/34a54fe067334dd8a45c34aac7baf577 : SUCCESS in 1h 24m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3b1714941ca44285a97b747855f3fca2 : SUCCESS in 50m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e1dcc76a4c2b45fcdaa71c6e7b741c7e7317c907","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-14 18:02:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"223658ab2748141968455b01620658e801ba7684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 18:57:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9b7558fc0724833a9871756765657cb : SUCCESS in 3m 44s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/36bb7794855943749ee2b33a8f3d7689 : FAILURE in 3m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eb8ad1c3bd2492294ed9b6359cc79bb : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e49310a7340470ca48a68b95e45d151 : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bb727e8e5ce4b9c85e946c1e537c0a3 : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a503c0cd2714404fb097f335df9be264 : FAILURE in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/464df2ea1bbd4cbab0d40cd86c30984c : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/39f13a334371459c86b92ed244906fe9 : SUCCESS in 3m 03s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1de222543f7944399bf0deaa82d3c0d7 : SUCCESS in 7m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fa919f245a2647f684dbb12f80c5f954 : SUCCESS in 7m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3485cae0bbc04e889d58a628d869083b : SUCCESS in 12m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cbc838a3ec7f4adf9c594273dd019b72 : SUCCESS in 4m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/634339e78a22415c9b01a9eed0f80824 : SUCCESS in 10m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/617453aa51d0433e92e82363860c56a7 : SUCCESS in 7m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d7c4a18b4bcf4b4da9507027d49e467f : SUCCESS in 11m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6f372f6204064430b26441f18a19e829 : FAILURE in 4m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8c46d2beddf642839d898d7d34d51773 : SUCCESS in 12m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/931d2c8294364b8b80565c589c4cc7ef : FAILURE in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e98fccd786ff4bf588adc4c3ce8a7daf : SUCCESS in 18m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4cda69ad5a6341f1af47f83f6098988f : SUCCESS in 4m 14s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/66f341954d584ce29f2e06bc8fc75ae2 : FAILURE in 2m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1079928e616240139dc956d4dc9230c4 : SUCCESS in 6m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a5fabdb849164b2f9ef50eed0be205f7 : SUCCESS in 10m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9ffe094b02b94a4db4f099b2c67dcac8 : SUCCESS in 8m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a122ca2c9604d4eaf6db3fb728787f6 : FAILURE in 52m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2f3529a698d549b1a24c28b86dfb15fa : FAILURE in 13m 17s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/911046a8e512461ba2f386683526b209 : FAILURE in 47m 27s","accounts_in_message":[],"_revision_number":1},{"id":"84add5defbf66a62e9a28dbec911564744bf5f16","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-14 18:58:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"61897d843c32d0eb01698898fc47658def3c19df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 20:21:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d1c38776b6543a0a8227ec5714087bd : SUCCESS in 5m 01s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/2857830e7fc0488194fbe928d1ef4496 : FAILURE in 3m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/339264a8d6ac42cab2411ad5a6c29b23 : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cba200aefc7e4bb5adb1e2c5d3b11a9b : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d454ade833134087a3496ad4134854c5 : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac93505e456345328b308317533e78aa : FAILURE in 3m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae49e2333d2a431b887727a2af2b0dbc : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8ca9c2edc3594e74a374ed421da46f33 : SUCCESS in 3m 25s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/877eaabe8a23482282133029ed85b7e5 : SUCCESS in 7m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a5f4127c49544a3abef05afcfe859548 : SUCCESS in 6m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/da20df0d86b14ecabc22e989ed4b33bf : SUCCESS in 9m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3709c4e5671647c9907afc236a570d81 : SUCCESS in 4m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5167f5ab88484709990ad950d09740cc : SUCCESS in 8m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/141851d91a0e48ad80d8c95bfaa19a00 : SUCCESS in 6m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8664fe68d27247a1809d13e9843f1248 : SUCCESS in 14m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a61985a2fe0844a0b7ae16c1f3ed9dca : FAILURE in 4m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/57ee14e7c990412d819b0234813e9205 : SUCCESS in 15m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6852a36bf0cc473181a21563e0911bc2 : FAILURE in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6a413152c9fa4258b36248f49ab2d5e5 : SUCCESS in 18m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0e7ce0b0a55845f6a8907fa238078758 : SUCCESS in 6m 16s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/25e2808d81184792bb94b2ac3bd6f622 : FAILURE in 7m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d21b847c577c48d49b2ac71449c609bd : SUCCESS in 6m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e39f0f72c80147a58cb2f637416c82cb : SUCCESS in 12m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8a00ab74eb244e828f34734ebb6a538d : SUCCESS in 11m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/042fd9b167d64fa0923d2c3da8155e5b : FAILURE in 1h 17m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d06fbbba9bec4fe9a711334da05ae45b : FAILURE in 10m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4fd2c56a12f842838543adeaac6ebcbc : FAILURE in 46m 53s","accounts_in_message":[],"_revision_number":1},{"id":"20c51b338e54b6241ed76e4dae575386c3f015d3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-15 17:33:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2194b7c4a3ed05936d15c6609871d7548cde1701","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 19:14:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3b9c6bb08a544e0b74b81b2a208f62c : SUCCESS in 3m 32s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ce1e2bdee405445d9b61350cf7ffd159 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/556bd6bce54c4bc7a65a798bbfe45b08 : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a5c335eb4266462c8ba91317bbb18d4a : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c530ffe94cdf4df8bf3764ec4f0ed4d7 : SUCCESS in 3m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b82fa8caa7ac4ad4a10798a4981c5e21 : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/690c8916d76b4d66a550a7550b5eea5f : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/895ec7c03eda414aad72bcccfcbdd704 : SUCCESS in 3m 14s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2de0a36bd9f7463ab214427a3f99a0d4 : SUCCESS in 6m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2cc662dae4d24bdbacd00e2a61b569b1 : SUCCESS in 7m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/76e9c823e3254215b5f900263fd1fd50 : SUCCESS in 6m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6d94617c891c417f919b4feddcc79452 : SUCCESS in 5m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dfddf5b3de294a4eb294c9cc9ca34ef1 : SUCCESS in 8m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/98afc41970254aaabcef1e85da8f8b5f : SUCCESS in 8m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/aa1140d89a544e71acb2130b2df80b0d : SUCCESS in 19m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/17f302c409d7471dbf1e290bb145c00e : FAILURE in 4m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/29a668976d944c5d98f6784702afc962 : SUCCESS in 11m 59s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c66a0a273c744534a2ff0706e80c6174 : SUCCESS in 13m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aea37828c06e4078a8d098cfaf087aa5 : SUCCESS in 20m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/21265c3c6783462896ba26b608fb8921 : SUCCESS in 8m 02s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f50ac98313d447c78d17852681a6ae01 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0fefe8a6233c43329df579085ffb0c5c : SUCCESS in 5m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8508120870f04c7a835c3c8678869a1b : SUCCESS in 9m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a0b4ee64f79b4b8f9a7cfd7a564becda : SUCCESS in 8m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/779cd04051cb4fc89abd7368eeb14c57 : SUCCESS in 1h 03m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/aa7d2e1fcd91480493f8bedef6a8db59 : SUCCESS in 1h 24m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9020a99f791648a49fddbc6d9fbcc415 : SUCCESS in 52m 37s","accounts_in_message":[],"_revision_number":1},{"id":"5686077d5ab868bcb148138f7639a5dc6042dc6d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-17 22:40:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"67ab19b4def9c652794e3bdb9dc5ee1ffdd0b47c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 00:31:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07d64f81038d4a2fa9265f4cf5b45ca9 : SUCCESS in 5m 01s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8a44d5820cd34f9fb1d22ff1686f4707 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d4372bfbad943ba9cf5f97ac2907001 : SUCCESS in 6m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/28d780e13e064754887b07d6576faf2b : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcd74121ef70424ab440f85bca2d18be : SUCCESS in 7m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/08d844f7cd2145148345dd9498a1fee8 : SUCCESS in 7m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/adf29626d4574ae2b4c7c3a07a2cbfc6 : SUCCESS in 6m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/729cec1bc8b346c0b7fcb63864d213ca : SUCCESS in 6m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5d848a63f75844cd9f18e50c1c60201e : SUCCESS in 9m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e09ab79fac3c4c25a8142bfd9623e3cf : SUCCESS in 12m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ae99f72148d84e04bf8301f1fd8a3600 : TIMED_OUT in 41m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1966416b6dc44236905fe1621e18ee96 : SUCCESS in 7m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ad93c80b6a704bedadf0dd1482d67c7b : SUCCESS in 12m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e1723978c38e4987bb44b572d66414ee : SUCCESS in 13m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/03101c31875648f2a9ad5e0d4e9f5c04 : SUCCESS in 15m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f065140c207a4feb8ddec6dc1e69ccbe : SUCCESS in 27m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c093d09744c341c6893d03d55cef2e2f : SUCCESS in 14m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f1d5da3f5a314a398edabc9c8c854c44 : SUCCESS in 15m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1658e912052c4f879cf5250445025fdb : SUCCESS in 23m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/87c523e1e2654c7db69e7d4e025a8302 : SUCCESS in 6m 29s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/684a3ef9fa404b0bb35f3d7c2c6c5477 : SUCCESS in 8m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8e3929d4529e4aff9b8d24842592e1b4 : SUCCESS in 9m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fb643f7bea0d4454a2bd72b83b1a8a04 : SUCCESS in 11m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/39a6d86d5de547178c7d1387ab3c0f3a : SUCCESS in 11m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d67ea3bfd6c94958adda18a7d3ebc0bf : SUCCESS in 1h 16m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f8acd196db60455388355c1c2d08ae1d : SUCCESS in 1h 40m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fcd929f0524f4125be8b94bb7867987d : SUCCESS in 1h 04m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3faa1036087a1563f6a72f3e055a551c9e1e65bd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-18 00:36:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a033d014c2a58f1780394ef211d8dd2ad00884a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 02:16:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/151ec962362e429495cb312d399aadaa : SUCCESS in 5m 39s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/2504dfeec82b46399d0dbcde2d514316 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/704a75126a56457992b182db8bae8338 : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8c24d6148b64ad385720913deb4413f : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26efec61abed44e295886cbc3d915e1b : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b637ffcc669a43b4b4293af6b3b5a6d7 : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b636e7ecb844762b606ae15ad87b236 : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1d243e2049c645bf8ef9a21835c924b5 : SUCCESS in 6m 06s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/975556b0c7dc477dbc3f9b0379ec45b5 : SUCCESS in 7m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d8bcd30e0d8e4e008a17c660c41927dd : SUCCESS in 7m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5cfc887b23a246028b3056259e32b4fa : SUCCESS in 10m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1e3a9bee57214de8a535516f526e907d : SUCCESS in 6m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cf78c38690e9452b98ca27f84bbd18c1 : SUCCESS in 9m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/94a33b592cdd43d7b3f8e245886b5717 : SUCCESS in 9m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/910893f526274124a0d0415c7e19e94e : SUCCESS in 14m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ec7a982ab2e94fae9b9abb2105834efa : TIMED_OUT in 41m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c093dace20894bb493314ef59f6c0602 : SUCCESS in 12m 55s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e1ed944f70ae41a190d007421df9d1c5 : SUCCESS in 15m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5c97dd52ed0e434b906e82c5c882bbc7 : SUCCESS in 27m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/296b44b0d1c340a1806c8cb08129e937 : SUCCESS in 5m 39s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e1a7705a1b8f4ce180ab09ac2594285c : SUCCESS in 6m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/86c74de419ee4cee97485653474ba477 : SUCCESS in 6m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4d3c952576834122ad6c6044eb7b01c8 : SUCCESS in 9m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b4c4d918730249b7ae66c25b96e00cf5 : SUCCESS in 12m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/51f9b10804954213afbed375683cd616 : SUCCESS in 1h 11m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9053ea3e02de4a81b8fa691ab43b38c8 : SUCCESS in 1h 32m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4f0b0de760694050acf46e03150f7685 : SUCCESS in 1h 05m 20s","accounts_in_message":[],"_revision_number":1},{"id":"51bb00c173148a5e2d94ce7fb314d4e9a3723b0a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-18 04:48:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8522f108334d9e0f556885c9fc29f15efe280bc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 06:38:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1ae5948533b4473b686ce253b2d295d : SUCCESS in 3m 38s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0a3e85029d34404e87267bd160f13508 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83408cad5cce4fceb3823ae2b7762bc0 : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab9d8b7727c245dcb48fc4b2317b4212 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31921c26cd9d44588a2cfd3a2232ed86 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e213ed6dc81a4517a3934dd641af91c8 : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aef87937a3754fad9dd81ef621d59181 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ce1aa4be4f77422e83d676c335ecaaa2 : SUCCESS in 3m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/463a36b87d12488bb169384eeb38b5be : SUCCESS in 8m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/be1ba7ba06f746429ea5f6fb53e2e999 : SUCCESS in 6m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8c9c3dccc10f423fbcaa00fb344e5372 : SUCCESS in 8m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0b9956f69e044ce88ab1266483b6baa6 : SUCCESS in 4m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/877dbe7eccce41efb07f49d01ac91d09 : SUCCESS in 10m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/77b626d1262e43a49e78bf07d02ee841 : SUCCESS in 6m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e520957a8fc4422098661e687dba4ac3 : SUCCESS in 12m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c9660d8facd541e9ba68baa6c5c5859e : SUCCESS in 20m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f3e17093eb234700a4b300bc66049718 : SUCCESS in 13m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6d1dbb20b41e407685ca9d4aef090f36 : SUCCESS in 12m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8edacbc071254ca8872b036ec48dac78 : SUCCESS in 19m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3930226529444ee48db87d260df824e0 : SUCCESS in 4m 24s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/483d0c410ef9400682e89ae373e249c0 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/42b06989578d4cde81aa8a31acc883b6 : SUCCESS in 6m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/37632cc650f149188f372eb0e347058c : SUCCESS in 9m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6c20dccf27bf48db8125c4a2fb56787c : SUCCESS in 8m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a8f4106754654c4ea7c5143bffb4f424 : SUCCESS in 1h 10m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/14c599d003af4e34acafbb1317b5b173 : SUCCESS in 1h 43m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/dcf4544b15ab47949d5e1639af8997f8 : SUCCESS in 1h 18m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b59021bac449909ee5143162074d5c0815dc4d3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 06:38:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a0d106b2abbd9886dca5e3f2e53c78dfab4a37c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 13:37:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0364c55fcf046ee82a3eed2c638b8d4 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec805a8c4e2d44a0b94ebda925e17550 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c1654a5402d3440c81f701cd8c921e56 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec9bc21796f741a78df2419149ed6fd0 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f173a5f4f654463841cd53926925e02 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/da04ac485c3045fe9771af930ec96488 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8edd12fe300d4b959e5c2000c158a37c : SUCCESS in 4m 14s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/88be1ff0070042f7962e9b619d74f32a : SUCCESS in 8m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/89ccf6845b4c4e97b29f6ba7f2422817 : SUCCESS in 7m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7d3ea7b601c14f84be0ac27236650473 : SUCCESS in 9m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ff2160bce0414087a3312abe753e5e83 : SUCCESS in 4m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/492b5684e7264b3db35426de9602957c : SUCCESS in 9m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/660e5a0a2bc845a5a08e83166b593556 : SUCCESS in 8m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/550777cc356049849cf87d9d3f3d9b04 : SUCCESS in 15m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/49191c3cd6d7451d8c11e2d989e502e8 : SUCCESS in 24m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/28012697721947d0827790fdbc20712d : SUCCESS in 13m 10s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5b1346740a5c4c8d9ad7294ec884fa5d : SUCCESS in 13m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e76a930ec93346a9a6169095883a05a9 : SUCCESS in 20m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/57cecd69828a46da8f60ae63a9133070 : SUCCESS in 5m 17s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/dc4efe93f56a414e8d6c7a7e3dbd23b1 : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f60740e54a40469498e85c6179aefe55 : SUCCESS in 6m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c9ee1e65c0d1479bb1712875b5e9d30b : SUCCESS in 9m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1b866a1914fa44a6bf1dec53b057b277 : SUCCESS in 10m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/71caf6506fc045cd9474f754193eac48 : SUCCESS in 1h 13m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4fac76cc418c45709727840c6e3f4411 : SUCCESS in 1h 33m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ac9068bde7f141b2a594729224e7e7df : SUCCESS in 57m 31s","accounts_in_message":[],"_revision_number":1},{"id":"105cb1fe2fe49b1404d9569980c83759a16bc2ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 13:37:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b3f6dec2b42f54c60dd75ef1e6fb20893ee528f","revisions":{"8b3f6dec2b42f54c60dd75ef1e6fb20893ee528f":{"kind":"REWORK","_number":1,"created":"2019-12-13 18:32:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/698987/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/698987/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/698987/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/698987/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/698987/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/698987/1"}}},"commit":{"parents":[{"commit":"b26e50aa194594c03af1413b09d93ec0dc364494","subject":"Merge \"update constraint for oslo.db to new release 5.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b26e50aa194594c03af1413b09d93ec0dc364494"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-13 18:32:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-13 18:32:23.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.11.0","message":"update constraint for os-brick to new release 2.11.0\n\nChange-Id: Ia61773d8595c546483c3adaa62efac8f6f98d815\nmeta:version: 2.11.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: Ic2bba3c7783a66844dc4ec626d72d694b2740dd3\nmeta:release:Code-Review+1: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b3f6dec2b42f54c60dd75ef1e6fb20893ee528f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b3f6dec2b42f54c60dd75ef1e6fb20893ee528f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
