)]}'
{"id":"openstack%2Frequirements~699183","triplet_id":"openstack%2Frequirements~master~I069d4c2d0910702d16c3c0ead58b3b376a443019","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I069d4c2d0910702d16c3c0ead58b3b376a443019","subject":"update constraint for tosca-parser to new release 1.7.0","status":"MERGED","created":"2019-12-16 10:59:25.000000000","updated":"2019-12-18 02:18:04.000000000","submitted":"2019-12-18 02:14:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699183-1576635299787-70b46864","meta_rev_id":"a7b3d78f2c956e14a9bd54c8b527fdbfb901225c","_number":699183,"virtual_id_number":699183,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-18 02:14:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-17 22:40:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-17 22:40:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 22:40:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-18 02:14:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ae5fa11cc103b3af754ecee16cb43b667193fcb","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-16 10:59:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d20ce6d090343b08dace926361ec45065c6591d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 13:35:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f04118be9f14c40804e520ae49e64b2 : SUCCESS in 5m 24s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b4cbdd0bd202443aa8018769d1aad671 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a45afec54ef44d6b2a871c802364fc8 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb1f3388b1174319b2f53ab9acc45f20 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56fe29074034443da619a56c9f0feacf : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/525764e3ccf94481bb0702e545382b85 : SUCCESS in 3m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f502ffdbd3a641c8a9db4d7fb5cab2dc : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2238f379a9cf4d2cb134061a4aaac931 : SUCCESS in 3m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7f65c43c61a349a9ae039a89d966296e : SUCCESS in 7m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ee7468a1653347d8bd6c348a52210313 : SUCCESS in 7m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4d648440d50144f1887c3bfa720c1137 : SUCCESS in 10m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0f68fa8cab38496bbbc82639c333f802 : SUCCESS in 5m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/aa12929a7fa34cca990ebcb29e9c5ed1 : SUCCESS in 8m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fb7a1547150d42f4aebc51f05e3023ac : SUCCESS in 11m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/879a54e6a7e747e0bfb8a3667ed204d0 : SUCCESS in 17m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d6037bcdb1184798aa3c40461e976b28 : FAILURE in 5m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/594c673a0c164a30af62690e5095b38e : SUCCESS in 13m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a39472b3f7844e95a3de211d8ae99de7 : SUCCESS in 14m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5c71632ddb13444aad310bce70e2ec6d : SUCCESS in 16m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/583e08b0a78b46999eb66704ab698cbc : SUCCESS in 5m 42s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1e1bb474f58342368c6e33a8a5c86ae3 : SUCCESS in 6m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aff88ee629b54fa183fdf512e8f8bb3c : SUCCESS in 7m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0f36a6c2440e4e75897dd4d5c8e65de4 : SUCCESS in 12m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f77ad43166e34a068dd8606f2afa4795 : SUCCESS in 11m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/023f37568d1946d0948cffb956b47514 : FAILURE in 14m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/23814dbe1ead47d9ae4d884c293f0f4d : SUCCESS in 1h 30m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/46c18e555e9b41219ef2997f19b86b10 : SUCCESS in 55m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a236419a0d7370c7eba0e972e5e92c41487e2840","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-17 22:40:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c36e2df649e22ac824557fe76479806c6701c2b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 00:08:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f64a0172d3434356bccfeb352532b2ab : SUCCESS in 5m 57s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f1daf710de464fe5bd3c7f9626ea7116 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77ebdc6631014f14b15b7219df74d616 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e61a397936074dc9a688d1fa187b9fe3 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75f4e903cf9d4805b0c959f8c52152a7 : SUCCESS in 7m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f328de6e16b454598cdcd19e2d4c500 : SUCCESS in 6m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a7adf1b0ba11442283017a41767fc241 : SUCCESS in 6m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3fdbf383720c4f48962e18f90cbfe41a : SUCCESS in 5m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b00c39d711564d778a37ffe326761a4c : SUCCESS in 9m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/31f30c662e664f619601091e7a52f6e1 : SUCCESS in 14m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/153a6eb61a8f4aa5ae1922ba6f2284dd : SUCCESS in 11m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/eb998a86f56f4d9485bdd3b1833b830e : SUCCESS in 7m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fe0f5b909b584c1b8b9ebe01e51bef22 : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/afaca5c09c8b41d59fac69a6bb6533cb : SUCCESS in 12m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/23604b055a184f328d279c133dd863cb : SUCCESS in 21m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e5a64b7dc914405ba0ed47c51fced75e : SUCCESS in 25m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/259ec1ea520548b9a2c461edff52a946 : SUCCESS in 16m 04s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/54f13e1dff8b455fbd3d86ce7e9f57eb : SUCCESS in 17m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f0bee32d34cb402f9bf8c1b502bbbef4 : SUCCESS in 19m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/be982898699f4da18c4316488677d821 : SUCCESS in 7m 53s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3966f4353a9040999602d214ba37c377 : SUCCESS in 8m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ae6b849abf2e47c2904d6aa465e1afae : SUCCESS in 8m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/09ed10e793674509b4aefa2887dd756b : SUCCESS in 12m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f2e52bb09443492c950c548b8110deac : SUCCESS in 12m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/10fcd9b8c4654e398469b304164330d7 : SUCCESS in 1h 06m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7e77a984a9034ebbb88fcf499c92882a : SUCCESS in 1h 24m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f4c11a687a5844109ab983f7e0f0603e : SUCCESS in 58m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a821e00df2277a183316bcd14283fe322e14151d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 00:08:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cd1c7ab89c38263d8da69c1c377d73d8def70a7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 02:14:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"41ec64f08f69961570507d85bb54e72323e6e486","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 02:14:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d9de807a0354b8e911d6c8ee49ef9ee : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40ad3a70cdc443b4b04dd3483c83d627 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a221b3fa9b3142d290d4434f825fde26 : SUCCESS in 10m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbc397934c91458b9ce07535f95ac17f : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6761cd1ecf8f4abcac5b13654eb3d13a : SUCCESS in 5m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eaa4986c849142599a6173aff9e36b38 : SUCCESS in 6m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc3ece42252f4fbdb2b70fcec5e7091a : SUCCESS in 4m 18s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e4c77912e9a649bca7996f35da595bcb : SUCCESS in 26m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3319c95472d34485ab300ad891ca65c6 : SUCCESS in 25m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d80e871c86eb458691d1bcf6d8b2b5db : SUCCESS in 10m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/276d3f4fed19482e94b165c87e40b37e : SUCCESS in 5m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/112e1685f52d452c802f03fe60a52462 : SUCCESS in 13m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f357c505de374c98954a1fb0a792d311 : SUCCESS in 6m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4928a8d95eb74e32ad589c7ad0604d3e : SUCCESS in 47m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fba26fa174fa42bc90d9cbeb3a8ee8ed : SUCCESS in 28m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8eb3d1a4f9404790a83ea15293f7362e : SUCCESS in 23m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6b0ee123fbf949d19812838704ffee9f : SUCCESS in 15m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/468775080d55469ab84b235647cdf69e : SUCCESS in 20m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2e790c998efa4cefa96da9c825da51f9 : SUCCESS in 4m 58s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1f1c0a3c8de243b39e6cc038ed803475 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b73cec49535047429b67dec775288a55 : SUCCESS in 5m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1850eaed5d6946b6bbf809ea120b9602 : SUCCESS in 12m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1dcbd0903f494dd19da4f9ddc5a53867 : SUCCESS in 10m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e6161e922104f22ab9292593776ccf4 : SUCCESS in 1h 07m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bbdd2f46fbb44c0cbfa48b7683d21bb7 : SUCCESS in 1h 53m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/19769cf9e0714b06948f93ea2dcd3f60 : SUCCESS in 1h 02m 49s","accounts_in_message":[],"_revision_number":1},{"id":"60b719275e3ddff2c24661cc1ce77ed277fd14b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 02:18:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ad789072aaa4b51afc26ede6011330a : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4eb5a2ebc18fc12df19621ed7f0f8220fda3d0ac","revisions":{"4eb5a2ebc18fc12df19621ed7f0f8220fda3d0ac":{"kind":"REWORK","_number":1,"created":"2019-12-16 10:59:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/699183/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/699183/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/699183/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/699183/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/699183/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/699183/1"}}},"commit":{"parents":[{"commit":"698b571749982f61e65814ca3de489177c14d30e","subject":"mark regex as regex for pycodestyle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/698b571749982f61e65814ca3de489177c14d30e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-16 10:59:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-16 10:59:22.000000000","tz":0},"subject":"update constraint for tosca-parser to new release 1.7.0","message":"update constraint for tosca-parser to new release 1.7.0\n\nChange-Id: I069d4c2d0910702d16c3c0ead58b3b376a443019\nmeta:version: 1.7.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Iec878d5978840b981fe95fae826ee87e9f2c7a06\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4eb5a2ebc18fc12df19621ed7f0f8220fda3d0ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4eb5a2ebc18fc12df19621ed7f0f8220fda3d0ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
