)]}'
{"id":"openstack%2Frequirements~699185","triplet_id":"openstack%2Frequirements~master~Ifeae952972902cc48833f9507aab63d6f7f30a97","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ifeae952972902cc48833f9507aab63d6f7f30a97","subject":"update constraint for cliff to new release 2.17.0","status":"ABANDONED","created":"2019-12-16 11:00:43.000000000","updated":"2020-01-07 17:58:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1566f1bfe731ed561d2432199ed2e29564e5918f","_number":699185,"virtual_id_number":699185,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-01-06 20:14:49.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-18 19:05:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-12-18 19:11:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-12-18 19:11:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"date":"2020-01-06 20:14:18.000000000","_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-18 19:05:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-18 19:11:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-06 20:14:18.000000000","updated_by":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"reviewer":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"state":"REVIEWER"},{"updated":"2020-01-06 20:14:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"f62d82d2ad3515cf85462cc88aa9f86448bd7f10","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-16 11:00:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43c62fcf25457784abd5d18c18cd9db13ab70b0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 13:31:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cc31e2220b34e1580e10107bdc26fc5 : SUCCESS in 4m 05s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/46d7d9fb3c6e408a98e15649a85fb632 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74e733f5ce5d415e9f6060ae303546ae : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d6c184a69f3445069cb8c2130e6b997f : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fe4bb7120bf4bfa916b9e508b73f9f9 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09ab5280785f46ce8ee4e6fe7e3a742a : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ab830e2afa1e4bdcb8f10031950a6d2b : SUCCESS in 4m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3a03ae2c6e804fa9a31e3a8c04a002c5 : SUCCESS in 3m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6848e2d7c6f14833ad7ef8c0ebff6170 : SUCCESS in 7m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f634d403d0664c61bc8ca5cbdc120d9b : SUCCESS in 11m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3ba6b08387854eff8dafa38ed873bae8 : SUCCESS in 12m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/518f18d081ba4f63b313069fd72abf24 : FAILURE in 7m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/72d46f9c654d47f7a19d2c2f7a4548c7 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/65c0f93cbcbf44eba98bf26c87f2a29e : SUCCESS in 7m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/34d9d08f194041b0aaa70d7ab32af4dd : SUCCESS in 13m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1b5ecb5c40624b81822815b95ffd4188 : FAILURE in 4m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a7bcaa9ad84f49d4947367552dde18a7 : SUCCESS in 15m 42s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b97e08dd23ca404fa02ab4019445122d : SUCCESS in 18m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/744be7c5002f4f1a80c940db9a2494fd : SUCCESS in 19m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/61958b07b19042f0883701afb9feef39 : SUCCESS in 4m 41s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a012e95cbf2d481998482c304b3d32b4 : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1657c757f42f4e29b0fc7f9188bfa678 : FAILURE in 3m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dc5a9efd5b8e4ddb9ed4c9e3c27e7f27 : SUCCESS in 11m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1ef5b59064e04784b536bc439c76b1b9 : SUCCESS in 10m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9112b65a6ed4c3db910fd24908aae46 : SUCCESS in 1h 12m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d2a51540c9f34e40bf8a5969b61b1d76 : FAILURE in 1h 19m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/aeac2db45f9c48e3adeaf60b1ce1e444 : SUCCESS in 55m 09s","accounts_in_message":[],"_revision_number":1},{"id":"7219b06b9125208366cd5277bb158f9654a406a6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-17 22:40:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9676599399a120c08ddbbdf47c3b6720933fce4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 00:07:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2beaa329909d4611878bf5133e2cb3d2 : SUCCESS in 4m 41s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/07826f31ef5f474e83b0f9250337679a : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6f0cb22c1a24153bda87aa003019ade : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/55ee6097aef54035ad6e67eca0d7cf77 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20fe7c027ec54f13b1cb384c76c57584 : SUCCESS in 6m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c8d4fe6a0a8742d8997a23129890fe7e : SUCCESS in 6m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8cb573a36b204340a3be4eb34db0398d : SUCCESS in 5m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/af053431de904c5894e360d48f9fb865 : SUCCESS in 7m 43s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ad746c71863f45dd995d195a0a125f5c : SUCCESS in 10m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bcc88e4f27f24901ae261f8b7008e46f : SUCCESS in 8m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5f3f0b0da11b4348899cbf82582052de : SUCCESS in 11m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/16faa6d7b4494292b60ab4c2ed104a15 : SUCCESS in 7m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1d3581ef7a0742bebab08d5f0365b110 : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/473baf214dda4887a7cd8c8debd61f92 : SUCCESS in 9m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ab83504725164f44b36790df62aebf41 : SUCCESS in 20m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/463d2042db2d463895d9a73a40fe2975 : SUCCESS in 21m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7a9216ef08204adf9872fb95b507957d : SUCCESS in 12m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5bcc001b0cd94947b3ab831b54cd17a2 : SUCCESS in 13m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1d63b8c566f3445f8d32f7b9a10a41e5 : SUCCESS in 16m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/26f6fe55937a407c902ad606e1b0991b : SUCCESS in 6m 24s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/23844ba31ba743b6ac3a4977de69e353 : SUCCESS in 6m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5d47b77fa2c14667a8b9887d9b236fe7 : FAILURE in 5m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2834b01e8457449d9f4cf4e4882b2831 : SUCCESS in 11m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/05e68cbed0964c8b90a4403b089145a2 : SUCCESS in 11m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/835359f836354389884776345a0a5375 : SUCCESS in 1h 11m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dec26121ab8243a4a33a6ba2f74239cf : SUCCESS in 1h 24m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/35e867013fc94a65b101a8809d7e7678 : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":1},{"id":"29cc2b58b7f75babb7e440c5f7ba3e00f2deaead","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-18 00:36:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"462e55fecbbef3b30966865e84d2cf6dfe6ec473","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 02:09:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4130d9916e07481f98edf86191f3887f : SUCCESS in 4m 12s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5c3dd371721c47ae9fb889a7dba14cb3 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e325c1dd14c6483aab4bc8fef3efca50 : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dceee1f1b00149edb84a529a15701185 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e85f7351f98a441c99546c1dd6c18203 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7df41a6270414a2aae43948d3b86dd95 : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/92719b05bde446cdaa2123e8f69e6a3e : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/311dcf22f5e34963909b7c9d38fa272e : SUCCESS in 3m 33s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/72adb591eddb491cae43a0bc8d3b8e8e : SUCCESS in 7m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b923b7f53e224224892bf48a38f33600 : SUCCESS in 7m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7c5e5035104d4207988f2361676df82e : SUCCESS in 10m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0b6dcef0d820456ea03ce460fedce9fb : SUCCESS in 5m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/830286490c874f5aa42c1047a03144ef : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2f73e9bf42ce4b33ba2913182e31abc0 : SUCCESS in 7m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8daaa516b0d34faf9fe60df434710355 : SUCCESS in 14m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4f01f166e3064eb8a91f6f0ad297c869 : SUCCESS in 22m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ab998f5cabd24e88b9eb0d6a9d186d98 : SUCCESS in 12m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9d332e581b5546058cf93401cda113ee : SUCCESS in 15m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/152e2b673bcb493380cf1e31579982d3 : SUCCESS in 18m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c7b48f967736417ebc91d2a5c48d5bf6 : SUCCESS in 6m 07s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1d545fdea2814d5e8581fcd0d9f3d9de : SUCCESS in 6m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dbdfdd4e33694103aa2c612d36cdf1e2 : FAILURE in 4m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/624be230ac564f13b778ba1f4cfc3cf0 : SUCCESS in 9m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3abec13a3d4a4229b5fbb60d6af6ecb8 : SUCCESS in 11m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/17d27df8b378481d9a7bd0104739f98b : SUCCESS in 1h 09m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/53c9c8ba623742db90bb4af719b70baa : SUCCESS in 1h 30m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/56ce60fdaf2c45e793fadee593944c58 : SUCCESS in 1h 01m 10s","accounts_in_message":[],"_revision_number":1},{"id":"bee48c0ced79756f5c3fc3b4e5698c8485df4153","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-18 04:48:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e40a3ff7049b06cdbe1e3e383029bbf01cd57338","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 06:28:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eddd125a563a4bd5958319b873d41084 : SUCCESS in 4m 12s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/581927b9445b43bbaa264ae748dacd2e : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/336e416e79ef4164a50d01c206375f24 : SUCCESS in 3m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b1e88801ad0e4e94bb8d5e9d3097d7c3 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47afad10019d417399441670656840ed : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7adc2f2314064c31888838b7586d948a : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f3e49da0fb244e3b7e5bea80c0f8dbb : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/996a1c391b3e416394dcf5a49b1afaa9 : SUCCESS in 3m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4a3182fa7771484692fcfc4bec4a7813 : SUCCESS in 6m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a840fdf056254d9b88f8e18db39620ea : SUCCESS in 24m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5761d54fb5b8413da7c15cd599b07098 : SUCCESS in 8m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/186a472621a146deb413fd6c00e4278f : SUCCESS in 5m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3553f36c9a154087a0ab8385c4ce0c11 : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5cbda402b9fb49ad86c109dfec7113ae : SUCCESS in 7m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cffb4ba40c124a1598dcaaf32bbee362 : SUCCESS in 21m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/02a28cacae89440cb52ed998e0320565 : SUCCESS in 21m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d02c960946a0484caacd4378c6172529 : SUCCESS in 11m 35s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d781c0e33cff447aae891e77a199bea2 : SUCCESS in 20m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/697e1e522cfd4469a7a4cbc0e3eca45d : SUCCESS in 17m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/895b1657e7f840338544b2282e69d72f : SUCCESS in 4m 31s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5d073bbd334843938c6a4b3a73552dda : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bb905972474642b7b1771b13d622e431 : FAILURE in 3m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d2a1756efc07468e9d45f2915b959d73 : SUCCESS in 8m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3a6d56cb2fb34ab194491ee6210b5b45 : SUCCESS in 9m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7078a8176f6d47d384e894c8828171ac : SUCCESS in 1h 10m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9dfb89b758f941128bfc211766a16494 : SUCCESS in 1h 33m 29s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/df92b68120cd4898a74e67ea29a45742 : SUCCESS in 1h 20m 56s","accounts_in_message":[],"_revision_number":1},{"id":"355b1b2d0b19e935c70cb6c5c10dc4ea4b2651b1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-18 06:42:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fe094b989eae12891f854b54e749a4aaf388de5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 09:16:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e235eb5498a4ce2b9d02221b51a57fd : SUCCESS in 5m 58s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/1a98a516ad0143248522cc3d64dcaaaa : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f95c171d6f3b48efb62531dc8d6bd24e : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0bc1c14ec3794a47a92d677dbdf0bc23 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6d21af7d7dd4a83925f930f8df50056 : SUCCESS in 6m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bec78d268bbe4982869705b2be4991fa : SUCCESS in 8m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a72c781f5b64f949057c141289315f4 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dbe832910f6d44aa8cd51298e4a0bf0c : SUCCESS in 4m 14s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/30212f7371c241a296c6c279ca337fca : SUCCESS in 8m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f77784277bcf4ddd8600bf49149b18da : SUCCESS in 26m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c8de2e2128b049d79d03264e5e3b46e5 : SUCCESS in 10m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4234b68901934e429d23d04cc19ff026 : SUCCESS in 6m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2196187a91a148429f445570a8c3887d : SUCCESS in 15m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/596a53a43e7644719ee20da007a504fa : SUCCESS in 8m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f4fab2f6feea44a094f1786dfa9bf430 : SUCCESS in 16m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/293d80ce164b412d925138a5571152b6 : SUCCESS in 25m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/be9d6a5f1bd941e7abdb1bbdadd96232 : SUCCESS in 14m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/73dc402ff19b4a9da3d0b89df888b393 : SUCCESS in 14m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8ff9c318f0ff47418a33b61eccb86ed7 : SUCCESS in 18m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/35ad6349c6d5405e896c8929eda8009b : SUCCESS in 6m 27s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/46280fa50f7d4a99ac4227a6a945c84e : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d16391afc96e4c5597737e382cc27bb2 : FAILURE in 6m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5000ecd474414fffb4b5fd6d8ef2c122 : SUCCESS in 11m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/834be373c1864f9694865efb6b3b119c : SUCCESS in 11m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bc91802c2ead4e72b5eaf6d8692b655a : SUCCESS in 1h 04m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/539089d1abde49f1aa96fdecedcfbcfb : SUCCESS in 1h 41m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/18ae5b2f501546c58f2eac09f6aff2d1 : SUCCESS in 1h 07m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a0fcd9d683d117c32e2891be051235e12c6a2924","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-18 17:21:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2297f47d8eeb566f1265718dd66793419d462804","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 17:21:21.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"2c7e8233d820957c25d86f6b610118d6c7b6e947","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-18 17:30:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"11e6248bdb70418fcd0cf24e5ec62de8c2040d20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 19:05:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/804230aba9e44aa291d58e0ed46106ac : SUCCESS in 4m 50s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/452047eb9db14b639c953fb1e0649e35 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afdc7821685e40ce91dc7785ae847508 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e2e18da3f9994c6c8989d4699af8ba60 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f71a94389df4be48326c7be4d38f5e8 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c8ff7d8ec3f54f4f8de6c24f540dce2a : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/852208e8dee2484d97dcdf79e87d53f8 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d76cf4ba98a342b793b7c6c0bb85a3db : SUCCESS in 4m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a543a97e035843d299135790db11ebd7 : SUCCESS in 9m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/19f7642df1144d4c9db0e84e3aeb881f : SUCCESS in 8m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/070b7351c1484b9bb8be9d3a282feea1 : SUCCESS in 10m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6a5e6ea2704b45108a53175d2a90e2cc : SUCCESS in 5m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/db0e2b75894d40b880bb070d6597a644 : SUCCESS in 8m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/826ada0b729c4689aa900c8190ace64e : SUCCESS in 9m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2bbf9e69589f4fd08fd3615e6e7b12ca : SUCCESS in 14m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/328e5d49346544349b4df2622ede7ea1 : SUCCESS in 22m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a9ce9be4cfec46d886fed8bf5869251a : SUCCESS in 12m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b83afb9ac4a9449592f2f0480baba03c : SUCCESS in 16m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c3687414cc604d398c97f3c001fd938b : SUCCESS in 18m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/eb5ec7e16600431cae0f011f2ddbed32 : SUCCESS in 5m 21s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c55cc46204974c34a5465640289abe31 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2b73ce68746740a6a29c443b307a6f90 : FAILURE in 6m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f6bcee8ea75143d78a23994fd7483685 : SUCCESS in 10m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1ae0eae7f8a548b29f44d434bd6ff832 : SUCCESS in 12m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a139888064fd4701a30c67de40425b5b : SUCCESS in 1h 04m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7bb4b9469ea34980a7badcc666b775ff : SUCCESS in 1h 28m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3259a514dd774b5aa206d72e88e7f553 : FAILURE in 1h 09m 59s","accounts_in_message":[],"_revision_number":2},{"id":"de83c9ce60ae930cede8a0628f465eb6031e93c1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-18 19:11:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"04736e8cf58fb786b0a2cb04c15f3bd532555dc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 20:56:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/495cee859c384dde9160c1bc9ea19db0 : SUCCESS in 4m 31s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/849a0e3c99934291a8a5d4c2dd08842b : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b17c7d2f323d4499b921f4561a5d03bd : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2cd8902206c416784200439a052f121 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d03228fe2e9485da69d2cb1ebd8013a : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19f7b71d80e744a5adaf609316f92568 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e53e1e45955546b0b2c07eaac1a0bc05 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/350041e54e1940d68d7d856a2e0b554c : SUCCESS in 3m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a107a35f6d03469db4d5d4eb476b9cb9 : SUCCESS in 6m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d790835914a0495bb1e99f495582f367 : SUCCESS in 7m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9f7f752075604266bb927f7b188deea8 : SUCCESS in 8m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1c7700c63c7147989f828dece9c58fd5 : SUCCESS in 4m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dcf36f1698ae441d86ccd1a6abd3326c : SUCCESS in 9m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/55c80b36c57346b88f911ebc6def819b : SUCCESS in 7m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/90eb6287c64f46d48894517f8127dde3 : SUCCESS in 17m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d793402419ca480693ed81c1e911f893 : SUCCESS in 19m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e0bd9a11fc1b46f09082e2eae52cbb83 : SUCCESS in 12m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/09c4658355884966a692d61aba55924b : SUCCESS in 13m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1d15bf9e75cf40f58866294e6e5b6634 : SUCCESS in 19m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/374a6ff193f64870a2dabc4c4f72671c : SUCCESS in 4m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/408feb6aa8e44653a6c5f74b27e0901d : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dd8a3650feed4f3f8faaffd3ac8e321e : FAILURE in 4m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/15c80b32ad5449b8a48f812046d2197d : SUCCESS in 9m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bfb02d09a27d4f24937cd881361f1ec8 : SUCCESS in 9m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/40260ae46b26445a8f717cd19174e892 : SUCCESS in 1h 18m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2a3bf3f25ae64c11b1fa2093ff04b50c : SUCCESS in 1h 39m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/11a1f6563e4047b18363651b654ea321 : SUCCESS in 59m 17s","accounts_in_message":[],"_revision_number":2},{"id":"0b9830a19233e81c4414c229280d55667b342852","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-18 21:00:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5c1883726479ff32994304289d04b177ef352783","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 00:43:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7abfdad432dc467facd30e0a0edb8893 : SUCCESS in 3m 57s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/1105e6f8a5d44dc2ad5bdb5a60a3436d : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26139ed659a84df385f730dd4634484b : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a204ca370434401a9f6741fb4a7e371 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15d62a1dcdb8496991bf6bd5ef0e66c0 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46d2e2344ade4ced9f160d4a2772bc2c : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3b82aff2d7824dbea5aa23dcebc6f2e0 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eb7f99358b964ba98032b39146425bda : SUCCESS in 3m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b92dda20c5d54294944952c58fb3aa57 : SUCCESS in 14m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a82acd3093ce48dba80417501919a0c5 : SUCCESS in 9m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9d247da247354c83901256f404d62b0a : SUCCESS in 11m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2fb426773c144167bf6fef0409b5c135 : SUCCESS in 4m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/de36db133deb453583fc9f23fbfac99f : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/38b7d1f8bece469086f64adee0a80bbb : SUCCESS in 8m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/880a0917b1a941d09fff446ea5f70126 : SUCCESS in 15m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a7fa489fbae84f6da9c1d0cd37990cb8 : SUCCESS in 22m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3b96b1afa5574f63a112892a06f02321 : SUCCESS in 13m 35s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/409201714b7e48b98bee95e654d0130e : SUCCESS in 14m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4a4c7cd9852b4228893245a35c0ad128 : FAILURE in 19m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c9183c0ce4e645268adbc9fddd094b3a : SUCCESS in 4m 49s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8ac5b98d71524f6e8d1e167b85555df6 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/986426f7652c434685aef5ed12259009 : FAILURE in 4m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/955daac83f86424fb0019d93d8764574 : SUCCESS in 8m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2e8fd77349ac47b0b1cf94e57ba178aa : SUCCESS in 9m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/53b52a078dd946ada641981817dc7e71 : SUCCESS in 1h 21m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d7a8bf43e7774682bd665ace857db92e : SUCCESS in 1h 55m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5bd8e16fae1e4ada9d43ae44ebff0dca : SUCCESS in 1h 11m 34s","accounts_in_message":[],"_revision_number":2},{"id":"5d71a7b39585810c1e3cb83b3fe274c69406eb3a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-19 00:47:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"078b7d44b95d095d0c5f8adf4c4afc702f9af5c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 02:18:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0261a4065caf49c3a0643cf60d0e091b : SUCCESS in 3m 18s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5dfeca49d4054a2494fe28dd5c4ab656 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d181c0a34c9e457ab0620f87fc940048 : SUCCESS in 2m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2b6328a0cd24db0b8242c56f712becc : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51526111bb3446e3a9a31c0be837d73d : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8acd058f653409599976224d429fab0 : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ddcc27bd24b447e09eb342b5d3c5a767 : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/06932ef59b114d4680eb09bfec4460e0 : SUCCESS in 3m 31s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/453d7c6620e4430ebffd060de1f8c19a : SUCCESS in 6m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/71a2785a044a4f76a1b96e6da715ed2a : SUCCESS in 6m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5e089010f7d94dc99e01cac5b89e7838 : SUCCESS in 9m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3843546579b54a40a5d318533d3b1380 : SUCCESS in 5m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/408f2ad6580944148739c5acef998e42 : SUCCESS in 9m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a6c5ef493a844960b5150c07ba778e73 : SUCCESS in 8m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/26816204ede74c3091c53920a4f52ed3 : SUCCESS in 15m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/75f1e18f69074c64970bf251602afcaa : SUCCESS in 20m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f279b96481c1453c86dc148736aee09a : SUCCESS in 11m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e2a21634fe214d819865646046e2786f : SUCCESS in 13m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/07cc65f9e49f48e08df219f8ae92c8d3 : FAILURE in 18m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/aca0b7acba4d41d9880223022d4b66b6 : SUCCESS in 4m 14s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0f0dd764c7f94ad1b6b0ac3bc85ffc78 : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a3f5812b287947ec95837daa691f426c : FAILURE in 4m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5639c79a46f84558b8cc09d66b5bb8e8 : SUCCESS in 8m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d527e12ce5e6474aa61fc3c0cbdafe15 : SUCCESS in 9m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/89d43ecefc754455bd832bf0d14e2762 : SUCCESS in 1h 06m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f6a6794f0400476e98d8d07a5075a1b0 : SUCCESS in 1h 24m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/81c36f6f06e9458995e3dd897020ab38 : SUCCESS in 57m 30s","accounts_in_message":[],"_revision_number":2},{"id":"3f9d48a4eafc0c97722795d8d9168f6bcf02ebf9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-19 02:27:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5856d2f294f3167dcd764fcba89c158eb4e0c6ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 03:50:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32ee6bcf6f414d21848841690e7ad168 : SUCCESS in 3m 28s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/be05ed50f502401e863f1bce8b63d13e : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e48249bb597545179164e4bb5890712f : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7da9279bf98948138cec3063fb0cd3b4 : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54becfd09e894a5f9ac61734b1b0a8e5 : SUCCESS in 6m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8aa0da842a44f0eb583472b37d3625f : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a5d2840fab749168a1fac69bb4873c6 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/89a436ad14404117af4fc2ff0f78dc28 : SUCCESS in 3m 25s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/582bb3ef600f44d48614589696b2fd5f : SUCCESS in 6m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c9a03a20f98945b995180fb8549dd838 : SUCCESS in 10m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/469692d0ebd84db981b0ea270ca7ad53 : SUCCESS in 8m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/299d5659eab642d0a2e976e31ae13795 : SUCCESS in 4m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/71625999222c4f23adbbdc9e1ac0a7fa : SUCCESS in 8m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/47995e9e031b460d8f40a19a13a40a84 : SUCCESS in 7m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4492902ccd754f768d6755463e5dd5e9 : SUCCESS in 21m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/36ff90c979e24e73918c404e3de36178 : SUCCESS in 19m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2d20e306c7654de4b0c37d0e80de424e : SUCCESS in 13m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5daa02a5646f4ac6ba49eae57a3bf4f5 : SUCCESS in 12m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/08c0b0f84ed84724acff6ffbd09ca61b : FAILURE in 20m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/575166be68534541a1bfc69ab268b97a : SUCCESS in 6m 26s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3ce2c4ff34914c69858da5ba27641226 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e18c9b3019484353904ed3f995302d91 : FAILURE in 4m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1296613d9c8443caa0c3981520294ce9 : SUCCESS in 15m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/abba75031ed9481cbae9e6c47b270020 : SUCCESS in 9m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/20ed26feac944efea415d816fd7b2cff : FAILURE in 1h 02m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b400f6a6bf4e4073a7ced9add9376b59 : SUCCESS in 1h 21m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4a36538f48cc414e90f9f7dbc9cc6eeb : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":2},{"id":"2b736cdd8783b8c9338a1e6a7be6fe395519ea38","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-19 04:22:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f222863687acca558ecb68170e60b748aeb0f04e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 06:11:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e023ff75e14d41a375677ff23d8ab7 : SUCCESS in 4m 03s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a3d5834b65544c53b6ca4f6a5cf6a0d6 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/143e07849e02408aae0b6b8769a995bd : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fcc6fde2f3aa4f6f950af3b5bed2e63b : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e962e6052ffd4b059a1b431e09afe004 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/27dfc61ce3934e92bca8097e80d420f9 : SUCCESS in 9m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8dc9808173f0464fae37dcef4baf82e0 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bf26236288d5496f872fc141ca11a2c9 : SUCCESS in 7m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5e0830f6c59d4f6db182825a734592ba : SUCCESS in 7m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f8dbe5582787404cb509f21c15fa524f : SUCCESS in 17m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/420990f4022f4504a6e52f89283951b7 : SUCCESS in 9m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2c103b40adb943858904d2f9c3896932 : SUCCESS in 9m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9059461bea7e4e08919a654c24e54411 : SUCCESS in 9m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0f721f0a433040dbb1499accf71825e5 : SUCCESS in 13m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6a4b384831ff4494b412cdfab3ee92c7 : SUCCESS in 20m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b2954fd42f7e488089dcdd28d98b8264 : SUCCESS in 25m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3d45e7b32a3046d4a511e832d0f750a2 : SUCCESS in 12m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f06c9e02cbe94ef69d4a37dc75388540 : SUCCESS in 20m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2e2b8c57c1fb4352a02ef9b0d39abe87 : SUCCESS in 23m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c5919e4a1a474ec69e988a4cf07dd480 : SUCCESS in 8m 08s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/cca993a8988b440dbd04c15f6fbf9e65 : SUCCESS in 10m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/09cc4d4d803145aea059b642130eeffa : FAILURE in 4m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/44ef99bffbc24101bb7728ae40ca38ce : SUCCESS in 10m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ed921710bb7e463fb8486b08afdb64e5 : SUCCESS in 10m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/94c97da5b2184389b4d59eaff4add1e7 : SUCCESS in 1h 04m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7fd9d0a7ace6422ba772e952cbbcaa2d : SUCCESS in 1h 43m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/254538f3324242d08022567b305e6acc : SUCCESS in 1h 01m 39s","accounts_in_message":[],"_revision_number":2},{"id":"35c6f7d2ee3597781c782dfeb2fbcbe4e037af14","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-19 06:14:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"211790fe227efc74021d3b2378512a7dc061cc9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 09:01:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4442a759fa64c9eb1d76bb8941a37ad : SUCCESS in 7m 38s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/eb9ac856ec834d938b47e851e88170fc : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd6aee1db0da48ce927d9c1ee992fc89 : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6e8aee8854664250b7e508d02850df08 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6dcea202c21745baab3feb188356b06a : SUCCESS in 7m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/312c22ae64d9413fa1bcfe15f01caa52 : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e52f5c92312341b4bdd605ce02aa8baf : SUCCESS in 5m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/de5e8295499b48e4a49b8e4c35a9d941 : SUCCESS in 5m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/facccbd3d413449e89bb232c0546a6ca : SUCCESS in 13m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b21240990db74c2d959838095fab4f07 : SUCCESS in 10m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aecaa490ad31441a8d1b5673ba314b16 : SUCCESS in 11m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5f1a9da97ecd4c0692593ad67ae35285 : SUCCESS in 7m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/01f8000dc1d14fa49ee35e60cc78ad04 : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/325994b376ba49c28613d5e45589a3a0 : SUCCESS in 11m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7bf6595f0cfa403baed4dc57702768bf : SUCCESS in 18m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0a85707c86194adebd816f2a3431a938 : SUCCESS in 24m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3aa63183c82f48259802e41760f2c315 : SUCCESS in 17m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cc4300d3cfc1474fb2246a9adf7c57c0 : SUCCESS in 14m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a05b8f398a724e5a9d6138f7b17119bb : SUCCESS in 16m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d55411751996431c8c1c492761b0f8a3 : SUCCESS in 7m 45s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a052fe20710c461e8e11784481f40afe : SUCCESS in 8m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dbf3c4ef275e4036a48ee7c03b5d9fd9 : FAILURE in 6m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bc4d17467a21495aab41a92d8e015082 : SUCCESS in 12m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5bd870c0a3ac4915a4e666a0293a2a0b : SUCCESS in 11m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/37349c1a024743a5a3db00f08ca67b64 : SUCCESS in 1h 24m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5de0dbb429134604ace7a761ee80b344 : SUCCESS in 1h 39m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b8b4882373a94aaaa626ce20f9d48b37 : SUCCESS in 1h 04m 12s","accounts_in_message":[],"_revision_number":2},{"id":"da265539366e31a002ae3b2ccde6f524f73d31a4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-19 16:49:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c12f60835589ed4e894e016c8f2c866ca8d14d11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 19:20:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a0de2b24cf54691adb5ca5385fa6566 : SUCCESS in 5m 25s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/128031efcae544b796bd3c93244ad396 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dffbaac8a4a04041a7ca228796357b43 : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a53bb06ce514d8eb465e6c7e15e4a5a : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a690b55d81c4781be11becec35ceb94 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/acce2831c94d4f69b4fb5a2c03207672 : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/089718587ed54b4f8cc47de3a5d9eeb4 : SUCCESS in 4m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/57778da763a14998b22ff13a6d220b2e : SUCCESS in 6m 03s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0dec9a4669494827b994f568d0f8fa11 : SUCCESS in 11m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e504f257ae694491b1989686b9ef7484 : SUCCESS in 20m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ac765dbf30c24427a8de1e6082b1ddab : SUCCESS in 12m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5fa5d2c1bab14493a75d54ac014ad2ac : SUCCESS in 6m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/88424d6fba1c4bdc8bb8551eaaf6ae0a : SUCCESS in 9m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/19278a44965d4fd6811969524120ba92 : SUCCESS in 7m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4fb0a54aeac94dca9a030ac976008d5c : SUCCESS in 14m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f3e79dd0a0cb46ca804123ca99d703a9 : SUCCESS in 27m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9ae213ec8bb84e1091493855cad1bbe3 : SUCCESS in 12m 19s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e47fe55d8f1143a190d96736234da9d4 : SUCCESS in 18m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8baba046123e44fdb0f25ca30bdab91e : FAILURE in 20m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c053fcdb53fb404b91623cbe46704205 : SUCCESS in 6m 03s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c063beec4b9542568b120ca969a0647c : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/79c057b14410430c886b0c800aab93e3 : FAILURE in 6m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1c3d70cd890d4aada5843bcc3c54d09b : SUCCESS in 15m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2b00ebef2bbb46139ad5f784ded13d95 : SUCCESS in 14m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/76ed0974baa74e9dba1d43348a0e2170 : SUCCESS in 1h 06m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/041e82b50e20468faddf1b036d18ef05 : SUCCESS in 1h 45m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ff373c2503714ec2a31e3f417d6fdb4b : SUCCESS in 1h 06m 47s","accounts_in_message":[],"_revision_number":2},{"id":"eb8e75b2a8c4c0e0421c96c8fd45005bb2219bd0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-20 00:59:22.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"23e8ea437a514654770f0dfa8eb2d006e3d28fbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 02:23:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49801ba0aaef4352bdae0cc0404f22b3 : SUCCESS in 3m 19s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/254e1649c3bd4d749c2672407eb54abc : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/387632b0bf2f49f885f2bccdd23e80c0 : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/43dc3755d1ca4ca2999e7c70624f0f3b : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a12892f5fb2844a1b4012a449da92e1c : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0d7e71ee5c84787a1c85481dcb073ef : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/47c7f4f6279544efa7aa57311d60e6a9 : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/15267a06e87e4068babe1a02db7503c9 : SUCCESS in 3m 33s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/295f5e487e23494db69bc99dc9ce9c0a : SUCCESS in 13m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c121adb392db4297a6a961f2b3913aa8 : SUCCESS in 6m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a21c072161244ce7b3d5caa0bcef2d08 : SUCCESS in 10m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2b27a3ae3d1f4ddba78c04f1a6a4184f : SUCCESS in 4m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/274a581dc974472ca08a99696ae0694f : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/60d6f090367a4decabc8a77e9d5800f8 : SUCCESS in 6m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d86f9aa9dfdd43d79971eef2991bc389 : SUCCESS in 12m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c2e7f87e04ed4000a8b1f9e92db503a4 : SUCCESS in 22m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5eb12bc9cdcf483e88038bdeceb56aa2 : SUCCESS in 13m 59s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cd209e4344674923aa7d0d46e73efe50 : SUCCESS in 13m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e608e9ea87dd4cd9b15c731480a1b214 : SUCCESS in 19m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/95aa66039c404a91bba3454946495b3d : SUCCESS in 4m 24s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5682350865be4c8fa385b250d0cc7e4f : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1b50573b3b4344d0a76c8d81daf1f23f : FAILURE in 4m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/81c917ce125946ad8796c9814a2270fd : SUCCESS in 8m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/564fa8edc9a34d04bb95c6663a2c401b : SUCCESS in 10m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ef75916934d433fbe8810796fbee867 : SUCCESS in 1h 12m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4ae510c271dd453f8ce2336107cb5091 : SUCCESS in 1h 22m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/64a90470a5ce4cbaa1c5a2471cc08ee6 : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":2},{"id":"53db8b349acf7753cc6d8fa7a455ab5e17e95f0a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-20 22:06:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9be8fe68213aac4be83ab51a65893f5fa7680783","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 23:35:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e4d357ef9d340439faa58983ee72180 : SUCCESS in 4m 08s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/cca36344c993424dba29e3e9fa134e04 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92a0778466f84d32beb74d07bb5f3ef3 : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/84f1f96aece74eeb86a38b0c3973198a : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6ec2012fbbb409391b642d4e5f4d015 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b8928458b46427495bef60145b0e808 : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c314fa446ff940b98a2e53f66d253427 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ae0beac95360490fa86ccfb4af33f9bb : SUCCESS in 3m 57s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7f9fde57884d4eed957b5f812c1a8783 : SUCCESS in 6m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/46a97b6a5a0d465f986e4efcd024860c : SUCCESS in 7m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1daba5d0eccc4fbcb43e692623868daa : SUCCESS in 9m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/71f5903accc84393a8c11a578554c9cc : SUCCESS in 4m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7157059cd6924d498cd44f8fc4aba38e : SUCCESS in 9m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6aa44b9a2f4d47339113d69d9ab3d9f6 : SUCCESS in 7m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/632c00e520c54385969885b0405b0cab : SUCCESS in 14m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/01430e74c7334cf1abdf9306bbb63c29 : SUCCESS in 26m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0eb2e889dc0a4a038fc18d083abb7580 : SUCCESS in 12m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f562936e6a8842d0b32aa752ae189c8c : SUCCESS in 12m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b2da691f2df74c8a9c55d523c860a9b7 : SUCCESS in 16m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4fc51d8006ed4c80bde71215a559138e : SUCCESS in 4m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c617674a36774aab99edf437e2604592 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/81e3dfcfdd544b21b6d9ce035cfaa89d : FAILURE in 4m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f77ac6104367445c939bf26c09d4e6c5 : SUCCESS in 10m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fd2c88f4543e409eb3b11210c94fc074 : SUCCESS in 9m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec9838e2f24647de9aa6b7717f34bc21 : SUCCESS in 1h 00m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/099ca0596a1e4d5a8d1bf34fcadbc459 : SUCCESS in 1h 02m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0b3aacec9f4e4aeba0e8866f58677bd7 : SUCCESS in 1h 27m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/80b6b6a67ac34e7998ffe55861fa81c0 : SUCCESS in 54m 46s","accounts_in_message":[],"_revision_number":2},{"id":"1ad8420b3d5156314191c6e0dce230f781c89284","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-20 23:57:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cfd8f9ac3df9eaeebc7016a6f9349f847af0a8bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 01:38:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bb36e9c673346c7b201ab923d805c28 : SUCCESS in 4m 09s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4514f19b23af4e95b4372a8bd2cfacd7 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a5f10850684431a9ad54d8c6b602197 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/18814f6cb15d41e8996d40a047135eb8 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69de902faa6842d5a40dca70a79895df : SUCCESS in 3m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47464a48da7b4dd993858b948eb1f9ca : SUCCESS in 5m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/714ca51d7fa94510991aa993044bec9a : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7d73b35bb0b64479a639e8fa7de9977a : SUCCESS in 3m 26s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/db7b62c4c6494204a2dc68afcc24f411 : SUCCESS in 7m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c3a3d36867364fdf857c58d8b159b919 : SUCCESS in 7m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aac319e83c304ae89bf35af5ab1fd0b9 : SUCCESS in 9m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/70efb18d6b3d47ee8ed8f174d32054dc : SUCCESS in 4m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3517e5ea02b7425e9659b3f4916778db : SUCCESS in 11m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/551b464496a447f293bc595ddb109b31 : SUCCESS in 8m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c9bd8a52542843d7b6b3bf1282cfc284 : SUCCESS in 20m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d09db96f2491430da63fee06bf40853d : SUCCESS in 23m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cf521310419948219e9ae373bdf59243 : SUCCESS in 21m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/14e8a1570223458b9ed5c7a2e8ea4b9e : SUCCESS in 18m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf48b3a5d4cc4ce8851f6fd434a4f5ad : SUCCESS in 21m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/781432be84da494aa216a6ffbc33b9fd : SUCCESS in 5m 03s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/dcc1cdda8dfa495caf55005e00b011fe : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f6d01583f2da4930b8aacd748a39b00e : FAILURE in 4m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c1f6c3ad716349338c1ebcb04e03774e : SUCCESS in 10m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/72fb634fd0814d81a2dbd967a34dd2a0 : SUCCESS in 9m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e039a52073df4f168e13d207c55e2e95 : SUCCESS in 1h 05m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/06182aa3981f46188ec34dfa6413c445 : SUCCESS in 1h 23m 43s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6d46641a09d149caa58d0addaa209ad6 : SUCCESS in 1h 30m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4c9d419da2394c26b3345232fb1f05da : SUCCESS in 1h 05m 28s","accounts_in_message":[],"_revision_number":2},{"id":"97b12c44d10e866fc7e63a054986d5d6b79ecd1e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-21 01:53:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"14b61ed336974364eb70deb4c1dfc2b807c33b4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 03:34:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2762b29c7354244a0f326fbf1e27445 : SUCCESS in 4m 05s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f49440463143475382ceaf5ee1490208 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d36114fedbda41788f14f4fc3b611720 : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/22a219a9e1bb4e0f9b5383a0ec91a7fa : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc5f97aff63d4d7c9cf82a5225d3be4a : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad5f96ca3db1498eac2e823ce01e868e : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3d8669817e8b49f7ac9c6a0a5bd7909e : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/72e93fefd9324b5db5c274c869fa9d76 : SUCCESS in 3m 02s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2f77defd9b9e47ba8c7cd3d6dcb3759a : SUCCESS in 8m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f0a36c03597043c18078949d090e7eac : SUCCESS in 7m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e5ef307987f24c3bb8710e12329f7dff : SUCCESS in 9m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0de2ef4c87e24156a3744f7bdd31876f : SUCCESS in 6m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/31a43ffbf53342ce9a613ca10325c48a : SUCCESS in 12m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f94ed8df7b014d0a80719de1bdfbe2f8 : SUCCESS in 7m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/50606c662856452a817eacd726efb1a4 : SUCCESS in 17m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9441a336557c4f049ac0323a56b31269 : SUCCESS in 21m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b6a97c962dd74d1492c00b8bb80190b1 : SUCCESS in 11m 59s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/047f2397554a4de2a4b3021f4b82fc9a : SUCCESS in 15m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/84dac4db2ec34477b057edae574b78fe : SUCCESS in 17m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7bd76d57dea24aafa3b7e16c9db8029e : SUCCESS in 4m 50s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bf9eca08a2fe4bc2a08c09431a8cae21 : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dad02d88aac34f0e820679c6177351b9 : FAILURE in 4m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5c2e748e95d54318a2da395a1c28ae75 : SUCCESS in 8m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cf155bf5f3be4aa9a5d5d78846abb5da : SUCCESS in 10m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bf5c1bcbf96412794662ff636d2a2f6 : SUCCESS in 55m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4bf6d35e3c924687834889b87fe84fb0 : SUCCESS in 1h 10m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1b638ab036914d64ba2ac5b85a62f8c3 : SUCCESS in 1h 34m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6df5293b3cbf4c0eaeac1410d9e5287e : FAILURE in 59m 44s","accounts_in_message":[],"_revision_number":2},{"id":"8cbb7295ef9954e32e7e5b52aec2f31497824940","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-21 03:40:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b7bd2d8c778917aa0f7f94625e9281cfa9780cf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 05:01:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdbc068cce314db4b8d7ee165158e3de : SUCCESS in 3m 12s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/948d08c39d674059b3108590a371cb89 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a7eefa089e2409780467b00069d60be : SUCCESS in 3m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c0df02c129843199af58beedabc2a3e : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c58ebaab6f4426793c356d2caf4a533 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98269f37944d427fac4de58e7184dfe7 : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/163a9c499c294af5894f100de83209b3 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77d8c819d4ee4cf3b728d37e4b7e17ff : SUCCESS in 3m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/dc0a0a4a465b4881a523f83e77914cee : SUCCESS in 7m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/90f712f8f3cf49aba47d2b40afd936dd : SUCCESS in 7m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/19d39255ee864fc6a22cc89541314513 : SUCCESS in 8m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/40e7023831224d0fa075341ace685ed8 : SUCCESS in 4m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f73fab1ddf5041f39f6ed4af9d0ed402 : SUCCESS in 8m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9871c96b51494e8b9aa24e2360d450bb : SUCCESS in 8m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d2808bea1b364519ac61ee0cb69f4732 : SUCCESS in 35m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3b17d69985cd43bd978a2d408111fc90 : SUCCESS in 32m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4396be9614de456fbeb9ae3bda0bb16b : SUCCESS in 13m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/af492282d741444ea9e796073a41ba79 : SUCCESS in 26m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/200010f29595456d8eed6f08df380f50 : SUCCESS in 30m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f30d04244bed4049bd75b46b6a5893ae : SUCCESS in 4m 27s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/593f237488f2411389c8cca372ed7dfb : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f1023469db1241409567cd88d3920122 : FAILURE in 5m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b4264ddc6a324e85b59589f7a6d73f16 : SUCCESS in 10m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/462129942a104b1bbdf5f9485cd2192c : SUCCESS in 10m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3e2ca8791fc44848f39fb6d02ff4567 : SUCCESS in 54m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac1bb665f06347ec891290be6e31f3f2 : SUCCESS in 1h 15m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b48025c05f784a50864de3519b31a806 : FAILURE in 1h 15m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3b249a7b6ac14c0faf11943e0ad1ac2d : SUCCESS in 1h 09m 58s","accounts_in_message":[],"_revision_number":2},{"id":"6d2f6243ac91425d96939a70006404afd900a949","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-21 05:17:54.000000000","message":"Patch Set 2:\n\nrechech","accounts_in_message":[],"_revision_number":2},{"id":"0e89a5f9985ec2cd323af3f07b939e5e1ac31bff","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-21 05:18:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"52e8286e2ac0d90aa295c3063d08b56a9cd36764","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 07:01:25.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/558b2f4ee3d74cce8f513b8aea5cceda : SUCCESS in 4m 32s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/fe926f60f3df4c17876eee3861712c2a : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f4d8a1052244d489b523cc490c705af : SUCCESS in 13m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4e43a5d1874f45888034ccd7a5a6f8f0 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1473c61550404488a1db3dcbba69e89b : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a48d019e90bf49e39fbe2a1b2e64964b : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7284c857d114b6a950678d5272b3d54 : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/32bd2a92c68949b487773e569cd9ab36 : SUCCESS in 3m 57s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/65d15def5d2f464fb286ed83a32dd3f5 : SUCCESS in 18m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e6c9088e666b442a9f64ae7070a09d29 : SUCCESS in 18m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4bf036fc95cb43d597c64635647c3433 : SUCCESS in 11m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/50e94aab987b49699f2d16e68acdef76 : SUCCESS in 5m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c596ef4809c84564bd582c99e456031a : SUCCESS in 14m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a45eb43802ca4e88a9ada5f6c7999f38 : SUCCESS in 8m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e5737227bdfb45bcb7931199dfee8016 : SUCCESS in 22m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/68b326dd3e7b437a83a22d24398a890c : SUCCESS in 23m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a8d9efd88d7b4366b84d0f5f886daa48 : SUCCESS in 15m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f0f4ebb2ce2c4c369e4fdc1caa6f2338 : SUCCESS in 17m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ae1a4571870142a19ba426c660eabf77 : SUCCESS in 14m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/29f5f4b76d2a4721866efb258838840f : SUCCESS in 4m 12s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/599b38fe303d43e38c7323063affdb3c : SUCCESS in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8a4855db1456409aa435e547413d6454 : FAILURE in 5m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d35d333b61454e78add99763b4b661b6 : SUCCESS in 13m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d00ba8281e4a48df87e0d181176bd98b : SUCCESS in 10m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c19d5dce4064b6ea28152c11e0e626b : SUCCESS in 49m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0211ab651afe4b3dbad7218ee746c963 : SUCCESS in 1h 14m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c944eb955ad94be7a3e4052e1b0abf2f : SUCCESS in 1h 27m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2f7f14b3a4fc415885965371c228d210 : SUCCESS in 56m 29s","accounts_in_message":[],"_revision_number":2},{"id":"cc52a1623a5b06ef2c2be70d67bfe87e54f64694","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-21 17:27:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d5969111ef32566ea1fc07a1d0ac231f51f33437","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 19:05:49.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fa7db1fae6a4802aac94848667dc792 : SUCCESS in 3m 05s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f9b2ac06bbee416295d6660cc6849bd0 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/675d999052484459a10e94fbd1618e9e : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9c5c1af552f4567a9bc40e7b8784ad8 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/678b63984e3e4d43b007f1ae6dc8ea03 : SUCCESS in 3m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d03c2564f8ea442a89f8a9b204ad9166 : SUCCESS in 3m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d6f7306ca8b14ce597b414aa624345f4 : SUCCESS in 5m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/416c44c96a104315b4d69180bb2367d9 : SUCCESS in 2m 52s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9ce8584f009a46fc9d0ff16b3b5e2b16 : SUCCESS in 8m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/54b435fe1cf643aab10974fd100cbbe5 : SUCCESS in 6m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c3921a887f83408483edf8a07ea528b2 : SUCCESS in 8m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f5a784e4e5f040e5b954302803450844 : SUCCESS in 3m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0ad87a5c07544b7bb4364a7bce375c38 : SUCCESS in 9m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fb29f8d25f994f0cb2e1491a009ac616 : SUCCESS in 6m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d0f7246bffc5459e9a2e4607cb867c75 : SUCCESS in 13m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/766cd6250bac4f72ab1e2343afba40ce : SUCCESS in 23m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b80c23172ed34f919f465c3feeae4651 : SUCCESS in 10m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9a1331d436184990a0d1718a017a84f0 : SUCCESS in 12m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d6f48813fb847e992496f7669e6c79c : SUCCESS in 14m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5330897350de4f98acdcf9954367cec5 : SUCCESS in 5m 16s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b92f3ed40a514fb1a0ca6a3d52c40a51 : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3f0b73cf7f8c49ac97b0e1e1b8e938a7 : FAILURE in 4m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6b7f8519ad90420b93eced37c38bd6b7 : SUCCESS in 8m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bf15f78691af49918b636af225ed65b0 : SUCCESS in 8m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c70fda5c597a4e619a228eca1fbcb6fd : SUCCESS in 51m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/42c40d4cfd9a42449e2e3a6d7ae197a9 : SUCCESS in 53m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/350057d395114a06995bf4253d318957 : SUCCESS in 1h 31m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1f19f0bc43d64339854d08f315cf1bf6 : SUCCESS in 57m 58s","accounts_in_message":[],"_revision_number":2},{"id":"558d5f4c56e15286eb66fe2fdaf19f45338ec248","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-21 19:19:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dfdd635cd57aecb5b4afcd3eec594e8b029ad32e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 21:06:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ee252489ec842a190ed90304359372a : SUCCESS in 3m 39s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7f09528b3cef49a2a377f7051ba6ed49 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc9358dc7b774b228b661ce790d6806d : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a654e90f08724d44a3f4c0092f1a9965 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df6f42b822244eff97344673da0937ee : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67903c6598304cf6b9efbd37c91d51b7 : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/773e8a744bbb4203be13c43185ee2409 : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/580412399a3a402aae415eb381223693 : SUCCESS in 4m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/38c21212cbd44892b213e0b927a5948f : SUCCESS in 6m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c422b67af45d42f8a1dbf50de837cd1f : SUCCESS in 7m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b6909ac1f07547f68109ea14e041ad19 : SUCCESS in 11m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a9a9d51ef06c4b85a1f0c8af43cddd7f : SUCCESS in 4m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4b61d26c6ba3407e9fd383987f1a205e : SUCCESS in 8m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0a8ec4c174a74f2bbc5070067cc1c7e0 : SUCCESS in 6m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6d75681b26384652bbd147d4eabdd4ce : SUCCESS in 41m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fa9dbc200e2b4d67815e4c246a7f2f19 : SUCCESS in 23m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/aaf277067d2649b98ef8dda37af24d0b : SUCCESS in 12m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ab6d663c15eb46c88851a946f43edf8e : SUCCESS in 14m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a07ed81eedd468b9d5adc3a75f3484e : SUCCESS in 20m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/70cffd4cd642444c9af922ee5664f5e7 : SUCCESS in 4m 48s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4b77a954cc764559a815245d174a696f : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4aacc2f4da394a5a85d2fe77ea9f5149 : FAILURE in 5m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b9da0e700b184e7fa033fa7507dd68f4 : SUCCESS in 12m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9562f8be46b544118812066276801384 : SUCCESS in 9m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e36cfedafff48ee9c71492d4dd0f452 : SUCCESS in 1h 01m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b2e049055fe940fab3e841f0ca9481ed : SUCCESS in 1h 11m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/010212fc14d6467c8d11166aaa1f8c0c : SUCCESS in 1h 44m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fdc9f8d430b94263821e21fceb7d2fcd : SUCCESS in 1h 01m 37s","accounts_in_message":[],"_revision_number":2},{"id":"f5df0994b401c40d2e2d9153f24218bcc05a4cec","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-06 20:14:49.000000000","message":"Patch Set 2:\n\nLooking like this version will probably need to be blacklisted once the issue is identified and resolved with a new release.\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-December/011741.html","accounts_in_message":[],"_revision_number":2},{"id":"f4c535a543180ff31298a2eab49d2652547e5881","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-07 17:58:00.000000000","message":"Abandoned\n\ngonna abandon this based on the blacklisting","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c064c6b25757d5a553b12d4fb30dd22b3ef52668","revisions":{"ef7e7a7327b8d0b37934775dbf63405a076d8212":{"kind":"REWORK","_number":1,"created":"2019-12-16 11:00:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/699185/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/699185/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/699185/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/699185/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/699185/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/699185/1"}}},"commit":{"parents":[{"commit":"698b571749982f61e65814ca3de489177c14d30e","subject":"mark regex as regex for pycodestyle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/698b571749982f61e65814ca3de489177c14d30e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-16 11:00:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-16 11:00:41.000000000","tz":0},"subject":"update constraint for cliff to new release 2.17.0","message":"update constraint for cliff to new release 2.17.0\n\nChange-Id: Ifeae952972902cc48833f9507aab63d6f7f30a97\nmeta:version: 2.17.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Iff6572d96e22544d6798caaeff790f46807eac42\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef7e7a7327b8d0b37934775dbf63405a076d8212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef7e7a7327b8d0b37934775dbf63405a076d8212"}]},"branch":"refs/heads/master"},"c064c6b25757d5a553b12d4fb30dd22b3ef52668":{"kind":"REWORK","_number":2,"created":"2019-12-18 17:30:10.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/85/699185/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/699185/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/699185/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/699185/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/699185/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/699185/2"}}},"commit":{"parents":[{"commit":"d55acd3ea6fdf736b9c3ffc0b8bed2b8bdb71931","subject":"Merge \"update constraint for oslo.db to new release 5.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d55acd3ea6fdf736b9c3ffc0b8bed2b8bdb71931"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-16 11:00:41.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-12-18 17:29:57.000000000","tz":-360},"subject":"update constraint for cliff to new release 2.17.0","message":"update constraint for cliff to new release 2.17.0\n\nChange-Id: Ifeae952972902cc48833f9507aab63d6f7f30a97\nmeta:version: 2.17.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Iff6572d96e22544d6798caaeff790f46807eac42\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c064c6b25757d5a553b12d4fb30dd22b3ef52668"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c064c6b25757d5a553b12d4fb30dd22b3ef52668"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
