)]}'
{"id":"openstack%2Frequirements~699187","triplet_id":"openstack%2Frequirements~master~I24159b786b5849d8b024e34fa7ef4cd9e698eb6b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I24159b786b5849d8b024e34fa7ef4cd9e698eb6b","subject":"update constraint for osc-lib to new release 1.15.0","status":"MERGED","created":"2019-12-16 11:02:37.000000000","updated":"2019-12-18 02:35:40.000000000","submitted":"2019-12-18 02:33:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699187-1576636399642-fead5b33","meta_rev_id":"2ccd543beeb52aea486c03f7c476b995d4ac9aae","_number":699187,"virtual_id_number":699187,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-18 02:33:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-17 22:40:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-17 22:40:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 22:40:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-18 02:33:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d6dd7658fe317c962c5acb2be08e92c4e0e785b2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-16 11:02:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f203cb13f710a85670536000d0416cd833183262","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 14:08:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcd29bea8e3242918778bb1c72a17dd5 : SUCCESS in 4m 10s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/da320f5dfb7e494abd75815cb010b03a : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7dea53100eb496cba178d6cb9006f74 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/04055f3788be4213bba4cb2460ac8764 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6984d5d3a2f438aaf96a90a7ed4ff57 : SUCCESS in 6m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c9ecac6e9b8f4c49887d619e80a70689 : SUCCESS in 5m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0753581ca07b48e78837d25f6de6e4e9 : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e1f6863d2e7f4c768a30166f20ed6a81 : SUCCESS in 3m 30s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5ae6d589551746749ab02f6eeb835914 : SUCCESS in 9m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1c4c2b3bceac477f99febef32319cd4c : SUCCESS in 9m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1e8f30d1dc69401585615e020f2717b2 : SUCCESS in 11m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9ff07d32557d44a7a5a98b8ee98df9d9 : SUCCESS in 5m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/55430ef7cb1142ab8fb4ee2b0279fbae : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/76be17bb0309478b8b5a5ee8a2d45f34 : SUCCESS in 6m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a7cf377b6da443039c18b2e19bbfc03d : SUCCESS in 14m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cc7689734d9a43a68ade33b5df4202b7 : FAILURE in 4m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/44e5e3910a234a468dd2431ea8c788ef : SUCCESS in 15m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fc7932fd84ad446b8eeb062f29f85c9c : SUCCESS in 18m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/298ad877a2d94829ace1c4fa45fb9fce : SUCCESS in 22m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f28de19250754ea0a179b5f56ad7e882 : SUCCESS in 6m 15s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/365de1f97d1542a8bfbcf5e72e7ab79a : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6d7be5b496644d0a835328141f372dc3 : SUCCESS in 7m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5d6f47e45ff04a78a33fcd530226eaad : SUCCESS in 9m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/941fab6c774a43448e8ee9a53b92e46c : SUCCESS in 9m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c14da299daf54113a0ec5a0547f40fad : SUCCESS in 1h 18m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e5697ed78a184ffdb7d8e717c93548ff : SUCCESS in 1h 51m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f6ee802999f5490fb50a62ba4d5ccf86 : SUCCESS in 57m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3f57a3b8a5fb89bc7c5bdc7be359819b81cfd44a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-17 22:40:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e34c80d9b1924f6eade1e70b668e4e6ddb733f96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 00:33:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1f49c24ded94c358d5f1acad71969c1 : SUCCESS in 6m 54s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8a7533eb1cac4b1080439045f140e880 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b243aa2d106f49f38b3e9ebf5f7d6d4f : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a9cdb34dc50490cbaf9d53832772087 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b5fc9043c5a4547b47f19035a64ca73 : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a2b12bf460d54092b4d689bc6e007f9b : SUCCESS in 7m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/32e832fc7a634ae580cdaffe6f729e7d : SUCCESS in 5m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0dd61e65c07646d2b4767b31873c5a93 : SUCCESS in 5m 51s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/01e0cbf1bf07475d91a84300cafc233d : SUCCESS in 8m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/11c5db38d241477e90a73075eebfbf35 : SUCCESS in 9m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/80611364f3b147188379961faae74895 : SUCCESS in 12m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/94b93c3e9609494e8f84dd546e768323 : SUCCESS in 8m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4b74f71e8fc244d180ff810c3d956058 : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b726fadaf8c94ddb9b8de5223f87c2f5 : SUCCESS in 9m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dc272a5fe2b9436f99a20e1013afbb51 : SUCCESS in 21m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8e495c9fbbe7449584cd5ecd9d2f0336 : SUCCESS in 22m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bcf8e80fb5da42bcbe58b7a2679f48e8 : SUCCESS in 18m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5de713484db940369f3ded7c2951f36a : SUCCESS in 17m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1f4277d499fc4806a2aeef9f284116be : SUCCESS in 21m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bea9560b34ed47709cceff6a4953b190 : SUCCESS in 6m 52s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/dd6afc2278384e5f8f6fba28dfe0b64f : SUCCESS in 7m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ea883163b8c4516832a4d0a51f5f7bc : SUCCESS in 7m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6446f455f6c44ba5b27115871bf7fea4 : SUCCESS in 10m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/89f3ff3573df4f5583c5e94b057b0951 : SUCCESS in 12m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fea36591c673430aafcc51ff27eb9744 : SUCCESS in 1h 13m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3610584e65394647b25981097c1718bd : SUCCESS in 1h 45m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c8067d3d0ea24e6ab65c70f146a5725b : SUCCESS in 59m 48s","accounts_in_message":[],"_revision_number":1},{"id":"199da11356b9bbd832e016005192283a24da5b88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 00:33:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"60416b53f7d54bc3a37a7b86288a6a13dc866020","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 02:33:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"44a0971c29583c23c7bebdb5caf6da050b90e33c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 02:33:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaa75db84cb8496f96a9e2a6e7c4dc8c : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0f0d5fc0035478581ef6077878bf9f0 : SUCCESS in 8m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e355f22030c04813804e176d962098a0 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e14db640a8741a1a54b966f042924a9 : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2705fd591e14f5395a66228080e540d : SUCCESS in 6m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f38c29646bbd4a73b6d6e1c5586a5081 : SUCCESS in 7m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5dd01e7978644872bbc681ce72588fe1 : SUCCESS in 5m 23s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b5b6110d757641e6bc6cc5684d5ae1bf : SUCCESS in 10m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e15feec55ce04d5ea388a7cb0adb743d : SUCCESS in 10m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5c611e281bb04199837beb30cfdac0b0 : SUCCESS in 10m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/28b1787c8e5a4693b2b5355bdfa0e97e : SUCCESS in 7m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/faef4dbe06c44d24badcca4459195264 : SUCCESS in 14m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b64f2e5518b34c5fab96497732f4754b : SUCCESS in 10m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8986b16637b147e08b4c7e7a171f3fc1 : SUCCESS in 25m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7e80087de0604a8bb837bbb5fdb0e9c1 : SUCCESS in 22m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/395bee28c8684cf38fe09b6d73808e98 : SUCCESS in 15m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4ad07776fe5a40e8abf08fea0ee40e55 : SUCCESS in 19m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2db6416597f94be5882cba0318a57eaa : SUCCESS in 22m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2bba04e81b1e41fbbc294c9f1503b90e : SUCCESS in 6m 59s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d53dc54543824b608669e85b15615e73 : SUCCESS in 8m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/000914c941964ce08957e747fbd1fb25 : SUCCESS in 8m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d9c9cff0ed394c809c0bd0872c9e5dc4 : SUCCESS in 15m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cd4bb922bc4848db94ab7acd7a9db9a4 : SUCCESS in 13m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/927ca849283346d69e737c8bed2621ca : SUCCESS in 1h 09m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/260ca943afc54b3db0b5e900c50463c5 : SUCCESS in 1h 21m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/67d161a7316541eda819fb119508a9f7 : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":1},{"id":"454e1b848016b0f0cd2bb1f9eeef201b036535dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 02:35:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6fd030be3bb4bcfb8206925a8808db4 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08c14666de79c92100ca5238dd7706906ed987bb","revisions":{"08c14666de79c92100ca5238dd7706906ed987bb":{"kind":"REWORK","_number":1,"created":"2019-12-16 11:02:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/699187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/699187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/699187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/699187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/699187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/699187/1"}}},"commit":{"parents":[{"commit":"698b571749982f61e65814ca3de489177c14d30e","subject":"mark regex as regex for pycodestyle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/698b571749982f61e65814ca3de489177c14d30e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-16 11:02:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-16 11:02:34.000000000","tz":0},"subject":"update constraint for osc-lib to new release 1.15.0","message":"update constraint for osc-lib to new release 1.15.0\n\nChange-Id: I24159b786b5849d8b024e34fa7ef4cd9e698eb6b\nmeta:version: 1.15.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I639aa13d9de4f0fd7b1023dc860934fb3429ad13\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/08c14666de79c92100ca5238dd7706906ed987bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/08c14666de79c92100ca5238dd7706906ed987bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
