)]}'
{"id":"openstack%2Frequirements~699290","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Id66a3e4d574fa4da5fbffe6a1aa24c5ea23ab8cc","project":"openstack/requirements","branch":"stable/train","hashtags":[],"change_id":"Id66a3e4d574fa4da5fbffe6a1aa24c5ea23ab8cc","subject":"mark regex as regex for pycodestyle","status":"MERGED","created":"2019-12-16 21:05:56.000000000","updated":"2019-12-17 00:29:36.000000000","submitted":"2019-12-17 00:29:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699290-1576542576903-c83103b1","meta_rev_id":"600e694fc42c55cbfcdf264f2d0fd29eabe62f0c","_number":699290,"virtual_id_number":699290,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-17 00:29:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-16 21:29:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-12-16 22:11:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-12-16 22:11:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-16 21:29:34.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-17 00:29:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac237d87dd8073b61718592a78c5eebda71f1809","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-16 21:05:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"592196b06b7faa4707e9358325774fc14042a15c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-16 21:29:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3fb69eda6ee689f91a254860ae1f1916a5b9adc6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-16 22:11:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nJust proposed the backport, but didn\u0027t write the code, so in the interest of unsticking some things I will approve.","accounts_in_message":[],"_revision_number":1},{"id":"0010dcdcf64844ac619effa8c0037f20a6f2f605","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 23:11:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0d3e98d34bfc49b48293726fa9f5b4b5 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd2f6b964ae74c089d26299518c294eb : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c15594af42341e08b9575e7e17f9fa6 : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efc2918cdcf74c3abee709e6b49dab77 : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/008b6f32749f4a1487db790728c124fe : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/df8fafe5b7c84de8b781b6b10da18e95 : SUCCESS in 3m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6544b2377574f9a9a17e6fc499a1c61 : SUCCESS in 2h 04m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5c46102423204883a5d433cb3fe73e09 : SUCCESS in 1h 35m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/45db8077496b48da8518f21b2c867ac7 : SUCCESS in 1h 02m 45s","accounts_in_message":[],"_revision_number":1},{"id":"6fdd860f6b323bed8233d76c459fe83f57a1e45f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 23:11:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e685156a0256fa77d122855847848301399e7ebf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 00:29:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b415a695d18144cc8d9462d8a4a62f96 : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8e82843c7a34081a40471a2c81b74d9 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aca934eee55241bf9ef89d08b7d91ac1 : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1cd74f0c0834c75ae44a99ae3ca8518 : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ade42bb6f3d54f428a99d5ee55b9818d : SUCCESS in 3m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5a4a8df5e3ad4f12a25730b96cde6e8e : SUCCESS in 1h 08m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f946bd1c0e93417a8e5bf810673dfc48 : SUCCESS in 1h 16m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e3c970b1e27742afa1ae60a74dd7f729 : SUCCESS in 57m 09s","accounts_in_message":[],"_revision_number":1},{"id":"600e694fc42c55cbfcdf264f2d0fd29eabe62f0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 00:29:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bfd4804c4763ef7b2990871eb5458288d54fc07f","revisions":{"bfd4804c4763ef7b2990871eb5458288d54fc07f":{"kind":"REWORK","_number":1,"created":"2019-12-16 21:05:56.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/90/699290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/699290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/699290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/699290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/699290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/699290/1"}}},"commit":{"parents":[{"commit":"e68c31180ea1ddd5bf6a24e3a218c7d59ecfb869","subject":"update constraint for ovsdbapp to new release 0.17.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e68c31180ea1ddd5bf6a24e3a218c7d59ecfb869"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-12-14 05:15:45.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-16 21:03:24.000000000","tz":-360},"subject":"mark regex as regex for pycodestyle","message":"mark regex as regex for pycodestyle\n\nChange-Id: Id66a3e4d574fa4da5fbffe6a1aa24c5ea23ab8cc\n(cherry picked from commit 698b571749982f61e65814ca3de489177c14d30e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfd4804c4763ef7b2990871eb5458288d54fc07f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfd4804c4763ef7b2990871eb5458288d54fc07f"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
