)]}'
{"id":"openstack%2Frequirements~699295","triplet_id":"openstack%2Frequirements~master~I31b8813231831bbf82adf44e113e52aba7c1e560","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I31b8813231831bbf82adf44e113e52aba7c1e560","subject":"update constraint for oslo.cache to new release 1.38.1","status":"MERGED","created":"2019-12-16 21:25:05.000000000","updated":"2019-12-18 02:32:41.000000000","submitted":"2019-12-18 02:29:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699295-1576636196839-d94f54cc","meta_rev_id":"935b15682d283de08447d72a396554ce9eb11276","_number":699295,"virtual_id_number":699295,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-18 02:29:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-17 22:40:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-17 22:40:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 22:40:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-18 02:29:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"357a48f91296949cd2ae5f7aeba3582dc61e3ebe","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-16 21:25:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29f1216623343285c5f9142bd26a12b1c69fe85b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 23:08:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/904fc6dfc8c44f2a976dec593ab16219 : SUCCESS in 4m 29s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/74ca9789c6e44d26950f4eb585d722e6 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77b667840dd649739a9573daa165107a : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bd61a2c6240e43739fa4e1bc7be2c616 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/926922a960354a7ba6636863b6cf533e : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/43e161186e29475d8d890b1c4f03e2c4 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50f7c78bfe0148e0a75d48ac8debbd33 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1d69e950067140f393877c40c5d780f7 : SUCCESS in 3m 42s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3c09218e7c2943c1a4026d6a6eccd632 : SUCCESS in 7m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7da5dbd8133a4602815d74e03293a527 : SUCCESS in 6m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6abfcf58365c44f9bda8dade354e1473 : SUCCESS in 9m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a12252b2358348389bf7ccc3ecbf744a : SUCCESS in 5m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/78ae53193ce747948765d024262625bd : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cf59804787264f84ad02c3e599435140 : SUCCESS in 7m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2da2ac2842f04f4c9aac73e85384905d : SUCCESS in 16m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a4ce0c2d394b4cb9abb8c22051653a2f : FAILURE in 5m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/810f66690029433794fa90bc56b1a46e : SUCCESS in 12m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c87d7d7807e54d81bf1bf695262aa7f7 : SUCCESS in 14m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/83028ed3b07c4b039510574902a5c549 : SUCCESS in 17m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/df1cbc8ec83745c6a0638df5e34ebe57 : SUCCESS in 5m 06s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3013123fc1724aaf9773db27642be97f : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d85435038df045e3b7e9a329dff06a79 : SUCCESS in 6m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0f92510a46714c479d4737d319ceb919 : SUCCESS in 11m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/531308c6d718430fb4b910c22f1d5a5f : SUCCESS in 10m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d78ca692390c4421b2d0b93b2d627320 : FAILURE in 1h 04m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0a06304d63b34c1b8d5304144c306ab5 : SUCCESS in 1h 40m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/94b3bf0c9b544821adf8d90f94978000 : SUCCESS in 1h 00m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e002bf9670d7f3ceeb335b0bf275323c0676e290","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-17 22:40:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7c923e06e952f5812fc8e725d89a0e9d933dab14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 00:32:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ade5310e0c9a4935b376f21421224ca8 : SUCCESS in 5m 14s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/76487822f8a74d8290c225e290be59e0 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a96eec0ab9d46d1bd589ef36e40af37 : SUCCESS in 6m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57fbe8f71ead4e2f95595a9b12cf7470 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50e192f78a6c40689bc73d7292268596 : SUCCESS in 6m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69b43473fdf14b2abafde37b333be5ef : SUCCESS in 9m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a2b658cb42244b0a4648e872f230e57 : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b98d67a9ffa44a41afd6e47f8c66574e : SUCCESS in 5m 13s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b7aba4a08152423c84aa8340448a59b8 : SUCCESS in 11m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8bdb9228c6024c0287849a2c35758009 : SUCCESS in 19m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cbdf1555786d4b519ad51cfa13db89a0 : SUCCESS in 16m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/002cc7dcda834da3ba1da7ce972a0920 : SUCCESS in 7m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0aff276437bd4bffbc0469fc4b9cd587 : SUCCESS in 13m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a354b7e11be64f119bd3d9d6645f8279 : SUCCESS in 9m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d7a678d0d0b44f3fa109d902e50b4d2d : SUCCESS in 20m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a5e363cfc8394701b115eaa3c73335a1 : SUCCESS in 20m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b79dfc3c5fe84ecc84f39b84d7f2dc46 : SUCCESS in 22m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5e91e7111eb445c2a82f4ab10cae6e72 : SUCCESS in 17m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/09140600eadd48c89034176ef14989b6 : SUCCESS in 19m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d9d1ab8f1d7942e89134a4dec81e79b2 : SUCCESS in 6m 15s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6924d76f4b7249df9f9b3aa94b708dcf : SUCCESS in 7m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ee7063bce5f4427e8adef0943ec0b301 : SUCCESS in 7m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dae8846166954b4197dccb96681a96a9 : SUCCESS in 11m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ba1af2c83ee74242ae2b31e64360ef11 : SUCCESS in 9m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3a4a0014f7ac4acbbd0d40acafbf3226 : SUCCESS in 1h 08m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b6a10e254df04881bf697dbae2ac39a2 : SUCCESS in 1h 34m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d141e5f236984edd9233c8eae60f5e97 : SUCCESS in 1h 02m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e54fffe86f35828a606bb97eb33543cffa00a697","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 00:33:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cdee36d2e784569f77e85df1327ffa6b81196ed1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 02:29:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c28e53e40e816b8b53cff3d89f80a6be4bc037e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 02:29:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d1dc7cbe11f4999ae43344d3c0bcdb0 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdc0d51648c9422291deeaaf954b4698 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4241ed0dbe5343ae8b749fe5cb7b9cd8 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1525eb5a3c04365bf6654e99653c51e : SUCCESS in 6m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/543c89dbba734f42be63ebbb459e7b49 : SUCCESS in 7m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a97f450f41f640cf9bea2d9f3867e1f7 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5fd1a80b068b4eba9a144792a706f167 : SUCCESS in 5m 59s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e773966c2c2d44558342bdf9cf3f83f5 : SUCCESS in 11m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fba0757942254d9f94d5e592d75a414c : SUCCESS in 9m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eb69d162a3da4aedad615486e3a3191f : SUCCESS in 11m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ccecfeab563d490c82507fac7404523d : SUCCESS in 8m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/41a7346e349d4466a8e2298bc13091f6 : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e518db777164b7eabcb80ec5d3eec1a : SUCCESS in 8m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8f58c726f4d7456eac334cd99885d559 : SUCCESS in 26m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7156ccf1de7d4127b20d74ba8764a562 : SUCCESS in 21m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c9e6470be3a74d34b9a5a5f8d45937fe : SUCCESS in 14m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7a9daba21d06474ea403aa2abcc33fe7 : SUCCESS in 18m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cc5d588236f84627a2a565111bd0e204 : SUCCESS in 18m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/eed946f8800d47b3bddbe9edafcfc477 : SUCCESS in 6m 52s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4c34e157848947f1a6a8918c2af2718f : SUCCESS in 6m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/af4b852ba5e442f38c1e875bf0aba577 : SUCCESS in 6m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e6c5df01245d4ca0a5cf4daa4c7d9f24 : SUCCESS in 12m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/337c7c353d59425ea18799007724cea4 : SUCCESS in 10m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/23452ab9823446b4865bc6ee4588d7cc : SUCCESS in 1h 07m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/27be02d423c44771b685a225146ebc01 : SUCCESS in 1h 26m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/79ab05f7846d40cd96bbc840c03484dd : SUCCESS in 59m 03s","accounts_in_message":[],"_revision_number":1},{"id":"29614c363d09bcaeedab870b505c08b80b74c4b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 02:32:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/923b71ebcc2041aebf9cf9cdb3da7d0c : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa2e45a9adc16677603ea1d108c2729edfb13c87","revisions":{"fa2e45a9adc16677603ea1d108c2729edfb13c87":{"kind":"REWORK","_number":1,"created":"2019-12-16 21:25:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/699295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/699295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/699295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/699295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/699295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/699295/1"}}},"commit":{"parents":[{"commit":"698b571749982f61e65814ca3de489177c14d30e","subject":"mark regex as regex for pycodestyle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/698b571749982f61e65814ca3de489177c14d30e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-16 21:25:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-16 21:25:03.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 1.38.1","message":"update constraint for oslo.cache to new release 1.38.1\n\nChange-Id: I31b8813231831bbf82adf44e113e52aba7c1e560\nmeta:version: 1.38.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I2ee5a8a85a2b93614e8a44c73513632416848f89\nmeta:release:Code-Review+1: Colleen Murphy \u003ccolleen@gazlene.net\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Kendall Nelson \u003ckennelson11@gmail.com\u003e\nmeta:release:Workflow+1: Kendall Nelson \u003ckennelson11@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa2e45a9adc16677603ea1d108c2729edfb13c87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa2e45a9adc16677603ea1d108c2729edfb13c87"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
