)]}'
{"id":"openstack%2Frequirements~699299","triplet_id":"openstack%2Frequirements~master~I42f565bd5134380ba09af0ab867bf96e87d753fe","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I42f565bd5134380ba09af0ab867bf96e87d753fe","subject":"update constraint for oslo.policy to new release 2.4.1","status":"MERGED","created":"2019-12-16 21:29:41.000000000","updated":"2019-12-18 03:20:03.000000000","submitted":"2019-12-18 03:17:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699299-1576639070396-69f755c7","meta_rev_id":"c6bc67887c1b5e8185097797025e652d4f104058","_number":699299,"virtual_id_number":699299,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-18 03:17:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-17 22:40:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-17 22:40:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 22:40:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-18 03:17:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"71676b74d6c47ba5b183fcfeb35c751da4618fdc","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-16 21:29:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d1ae600ab9acba0f23fca8e640bbad0aa8b5704","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 23:19:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e54422e4fc64bf2931014174da23b0f : SUCCESS in 4m 44s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/2476fd6a0abc4ffbb5b7a0ff584a6cf6 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce5ed136d8b54cc2b70075786f3e6289 : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/edec60a39a5b45d3833bb8f9d89fcfe3 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bcf6609b5a84aaa8b84e610a0752a37 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dbf1efa3864d404bb7d6efe2ded57c85 : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c8adb07a5cf948408be19ffbbe0ff86b : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f8f7c50e684d44d08967702f3c259a4c : SUCCESS in 3m 47s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b66b24d713994ab6a807c4782678eb94 : SUCCESS in 7m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/57b5e705747741cca83ac0bfd247b745 : SUCCESS in 6m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eb063639acfa413c8535f0be94e95656 : SUCCESS in 9m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e9f9ed920624423baf19dc17f6cd54fa : SUCCESS in 4m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/90b7877312874636a3161c1b40d32fb4 : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8ee8f68d45c04eac8d0065b5b58c696d : SUCCESS in 7m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1ba9835862ea4c1bba667fbc289dfcd2 : SUCCESS in 19m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/05f2c70172a44037b9b459629c1decfe : FAILURE in 4m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c56885cd29aa4fb1b9e9d5c57b9e87c8 : SUCCESS in 13m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c8249657dc5d429697c79507d5670af3 : SUCCESS in 13m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/03db21af59984493b2da4bc744172525 : SUCCESS in 19m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8d6dda9810064bddbf578a99cd65a5bf : SUCCESS in 4m 46s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9797c87e410a41faae7a22271e0a5024 : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/77638c391c814a62aff128f11150c99c : SUCCESS in 6m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ef98bb39f7c747259ee6b8b7d108d09c : SUCCESS in 8m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d39ae42609bc49d7a3e81ae33d7dda9d : SUCCESS in 10m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/19d3667a60c14140b7e87dc038b7d301 : SUCCESS in 1h 06m 29s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f4e87bfa10e645899d4fd44148cbc782 : SUCCESS in 1h 43m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3dbd443e3c65446084cb108fbb9895b7 : SUCCESS in 57m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c0c579581e4c3323db72b2dd221e1a91a0349d53","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-17 22:40:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"653684f2ab96ee80c90d98c76238c0ee028e065a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-18 00:36:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b6af5092de1fb330d5c1edd96e112af857eaa6f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 00:41:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/949a387ea1374391b28c88cf0406987d : SUCCESS in 4m 42s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6b1327a144e44d0ba014de43cb9eded0 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c862f93c40c744f598ccce321383ac77 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/96eb809e6e9a4bd8bd95168c12dde791 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3da2a26e7244661a50097588afe0be1 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/006244edf9ee40199fb82a01ef221c63 : SUCCESS in 5m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7817564649b54a89b29c2abd09e900b9 : SUCCESS in 5m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f73c14e1bb194cc7bfaebcee5859017a : SUCCESS in 6m 47s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9e5c6e655614463f843c0f752c9932db : SUCCESS in 19m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1f8d5cdc8acb42f2be25d2cfe0e3ac17 : SUCCESS in 16m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/97f835a3517a4419a1093734d94cfdc2 : SUCCESS in 14m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/40178ee645a2432fb97a0c5be3866ed0 : SUCCESS in 8m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4d3664d670364bc69bc80a4d502411e8 : SUCCESS in 12m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/835fb1532a204ff1bf8dafd325bca18f : SUCCESS in 14m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b37c3d28c8f04c8aa34db66ea68423d1 : SUCCESS in 22m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e40a5316025d4805bec8dfadac7b040e : SUCCESS in 23m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/66687f3cb21743038419552a66cc3149 : SUCCESS in 14m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9149f34889ea4854b9db26bb38293f59 : SUCCESS in 14m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/62c676a5db9945f7b7628a4803f0e970 : SUCCESS in 15m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d7b4d78e79e043c894b120267e897337 : SUCCESS in 6m 11s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ee03b242ace8495d8f5cdd34baff008b : SUCCESS in 9m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/931889fbd8b643f9b6de122a2f541354 : SUCCESS in 7m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/335b23c073834e70b9a77ed1f2af67b2 : SUCCESS in 12m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8c6a5ea6bb0f455aaf4d8fa4e88c9c5f : SUCCESS in 12m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/19165fbce5f040b1a1978cd748f8b276 : SUCCESS in 1h 08m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6e4bf77a353649509051fb96e2dc0890 : SUCCESS in 1h 38m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/10e9eb6e42b44a82aef5a786d28e179d : SUCCESS in 1h 07m 37s","accounts_in_message":[],"_revision_number":1},{"id":"f56ad984273de5f81c307303ed2efebd2f27a645","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 00:41:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7a0e29cd9ca970b6fe1e0b420c8dd6c5398d152d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 03:17:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c99adeca3f5f48f7a5800a924eccee52 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb59c659c9b24c4e9a031c450592d495 : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f9f732f6ae994e34acc0a9a6f05fc1fa : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d78a737455914f259740fe54f345e896 : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/390c8e7627704cd68946b755543c021c : SUCCESS in 9m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a9022d3bbbf4681a270178c9c97f41b : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8e4374f0b95543ab929c78fd7be6fbb8 : SUCCESS in 3m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ded9bdfc52d548bda6009ac62078f19a : SUCCESS in 7m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/92c743fae0be4c2e83f07bbe17c097d1 : SUCCESS in 8m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/967bf2ef8b2842da842b921c44291fba : SUCCESS in 9m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/27695bd08b5b42bda4c8d3d5b8677287 : SUCCESS in 4m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1c44ca0d8f434b84a11808b67949ef47 : SUCCESS in 11m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5adaf1a9a1924db1a23f9c399b2911ba : SUCCESS in 6m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/66ae198f10a84405801fbb863df52166 : SUCCESS in 15m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6666e607a06d484e9755bfa7d49fc195 : SUCCESS in 26m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/46de132b9c6c46ba8fa367aa99023088 : SUCCESS in 13m 10s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bd59605230a14af9abbf95213995f671 : SUCCESS in 14m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a77df545f9d344278958c19f753e7bec : SUCCESS in 19m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/eca964bd917848cca467663fc25372c8 : SUCCESS in 5m 36s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/86e4273770684e10918136bb61ffdcbc : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7570435e809a4d559e7d3e8665d95067 : SUCCESS in 5m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f366a976367341d184801120748e9e20 : SUCCESS in 9m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dac52dafb5ea4145a1529094765f9941 : SUCCESS in 10m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a573ee8be5944f9b853f0e3917e36d4 : SUCCESS in 1h 12m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8cfcbd0f975849c785e1dfb0552f4d0e : SUCCESS in 1h 23m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3dbcda5d75bf49859397b4bccb6b22ed : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":1},{"id":"d5786d28e04377dbea49612124d0e32b6ceaf186","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 03:17:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"10478ac271f9fe32f934be748bedd41909b2654f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 03:20:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/590f20a2430b4411ba5bb4373d200b03 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e2fdf6cd5342441657a22f45b4d6959fcd4adb64","revisions":{"e2fdf6cd5342441657a22f45b4d6959fcd4adb64":{"kind":"REWORK","_number":1,"created":"2019-12-16 21:29:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/699299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/699299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/699299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/699299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/699299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/699299/1"}}},"commit":{"parents":[{"commit":"698b571749982f61e65814ca3de489177c14d30e","subject":"mark regex as regex for pycodestyle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/698b571749982f61e65814ca3de489177c14d30e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-16 21:29:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-16 21:29:36.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 2.4.1","message":"update constraint for oslo.policy to new release 2.4.1\n\nChange-Id: I42f565bd5134380ba09af0ab867bf96e87d753fe\nmeta:version: 2.4.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I2ee5a8a85a2b93614e8a44c73513632416848f89\nmeta:release:Code-Review+1: Colleen Murphy \u003ccolleen@gazlene.net\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Kendall Nelson \u003ckennelson11@gmail.com\u003e\nmeta:release:Workflow+1: Kendall Nelson \u003ckennelson11@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2fdf6cd5342441657a22f45b4d6959fcd4adb64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2fdf6cd5342441657a22f45b4d6959fcd4adb64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
