)]}'
{"id":"openstack%2Frequirements~700154","triplet_id":"openstack%2Frequirements~master~Id968e906136ed74a87090c10df62e0f8704780c5","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id968e906136ed74a87090c10df62e0f8704780c5","subject":"update constraint for glance_store to new release 1.1.0","status":"MERGED","created":"2019-12-20 11:16:31.000000000","updated":"2019-12-21 08:37:00.000000000","submitted":"2019-12-21 08:35:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700154-1576917318654-17819b5d","meta_rev_id":"96cecfa48aa906b6d17c190edd1de2f063725d3c","_number":700154,"virtual_id_number":700154,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-21 08:35:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-12-20 16:39:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-12-20 16:39:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-20 16:39:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-12-21 08:35:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0296b14383ce8c8c7460ef904933be9d8f188094","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-20 11:16:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6b4e3d8041204c10bfaf0a799510588a09e9a36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 12:51:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/130d3110c5a34b9b8c23ab90ed2b6eef : SUCCESS in 3m 10s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/72453a36977f4f918a5c55d46475fb4d : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1fe8a563d5a4290a618288c8ecb4a9d : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d1f963470d9c483ebe7b025b86cf880e : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0828ecfc150940eaa512d6ca2abd9bc7 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17aa83b49dcd4c588118ea4d6577083b : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8033ef26fdf74922bebecc44706a071a : SUCCESS in 6m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d1c7ed0e29d345f893adf48caeb275d4 : SUCCESS in 3m 00s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f6be0bb6371946f688941263649d0e76 : SUCCESS in 14m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d2445ae3d3ed4f90a59fc3449bc34944 : SUCCESS in 13m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/94d15433b2f14a24a44b567caed1df94 : SUCCESS in 8m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/77de07d149534bf19a25e8c2b70c7add : SUCCESS in 5m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/809f5f20edfb4198b6645a09aabc9f1b : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/860f9685f6dd4ee4a002f13da0704ece : SUCCESS in 6m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fd6418716a49491ebaed3938e5302de4 : SUCCESS in 21m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0a6c6e3951a64d1f90c39c2a6df8f80c : SUCCESS in 28m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a67891cb55294f6bb13f3b55b60e2d8c : SUCCESS in 17m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/970ac69d8cfe463d9b2999fbd307fec4 : SUCCESS in 15m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/764e4d5c0bed4b7a9b67168d32be021e : SUCCESS in 17m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d4bf7582235a4953beeb7cf2879650f8 : SUCCESS in 5m 05s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/992bd3ea38574ad2ad6dee0fd0373833 : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5c611a3295784b17ac1b284e30eb795a : SUCCESS in 6m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/320983af5b8e4abaac8a971d33a081af : SUCCESS in 12m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/73996b325fe5413c936da0b55351606a : SUCCESS in 10m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe3c015758b14587a6cb2ce744375489 : SUCCESS in 1h 06m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8415fdb8a04c42e4a8a5b99855204f0d : SUCCESS in 1h 28m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/16582aa8d3a24ca5b0121b081793f8a1 : SUCCESS in 1h 24m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d7d9e860ddf14644968094a5d0465ebd : FAILURE in 1h 10m 28s","accounts_in_message":[],"_revision_number":1},{"id":"8b9a0f2ef0687b149291a207a6ac1cce4d004026","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-20 16:39:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aa9bf867ee7c047ecbb130c9c191803a7a868c61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 18:06:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/847114db2d014cc883c60e19c66ef9c5 : SUCCESS in 5m 03s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/83f532a6ca1d45c1934d32b0bc65d69b : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4fc9b8930eb445aba4c5e6c1e831239 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70776c3261114f59b1fc438358d22384 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79be8c1c4f12435a8c618756ab379017 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/408fb094234f4d8aa5b1412aafa30f38 : SUCCESS in 6m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/30d64e44ed3e456c839b49d56648e442 : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5382f9132d014e73b3df2452c6a6a9a0 : SUCCESS in 5m 25s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1da8e9027c0d4cfab16fc0a0c6991dae : SUCCESS in 12m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/81a3e94630d5431fb78bbba59a288f18 : SUCCESS in 8m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ad1f33370667400ca44524b1f65d7670 : SUCCESS in 11m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/249d077854ac4241934c713b4ccbe857 : SUCCESS in 4m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b7ace155cbae4883acaae1d89a33bb86 : SUCCESS in 11m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/156763510e9a4bbd9ad1e74dd7148477 : SUCCESS in 7m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3d34f16e48e34c23833cbe313a06bc55 : SUCCESS in 15m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e10069a5737644e3a12dc8d0b594a65b : SUCCESS in 23m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4638468c4e2047988b6c3053e66b1d31 : SUCCESS in 13m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/95aeb6267df94d8892131af2e92704d9 : SUCCESS in 15m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ffdb741a526f41fda70d7d63fb731e43 : SUCCESS in 19m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ef270c91b142422b9cefcf028953c792 : SUCCESS in 5m 33s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9dc305f4e3ac4fca81f21569a242c5fb : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e8606e2924c04a74ae4cced845358587 : SUCCESS in 6m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0b002eeede284c20afc4d9aadc5f97e6 : SUCCESS in 9m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dbdb8430af1640738b0a7a2d17cb4e20 : SUCCESS in 10m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe0a2dd4fef449e79eb29dd7a319902e : SUCCESS in 1h 08m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3057fa1fae1a457e90db981c6c72b93c : SUCCESS in 1h 03m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/33f645821fca4b6499fdee1169ee25e6 : SUCCESS in 1h 26m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/237226e31c9147e5b89d84af3f70d185 : SUCCESS in 57m 37s","accounts_in_message":[],"_revision_number":1},{"id":"377d1341fb22e4b23304b0fcd9b501e107a46559","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 18:07:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a050df34988b3ba2a65adc0f23149d32bb86aeef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 20:19:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/284173df2fb94df8901b9fd03bc99896 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08a1753dabbb4c70960c7f73c1284f3f : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2840b6933a94771a47b613d65b03958 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc3362e048e24eb2bcb39bff7de74d57 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/713523ab2f3c4b4dbc016a10d3716e08 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c36a7d28d2d64c51994049be1ee5e2c5 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/794c6be5dc6e4186b39ed9fff3e4702b : SUCCESS in 3m 50s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/467507fb584b420dadfb78af39800c33 : SUCCESS in 7m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fd25da368843483a9070aad7e13389bf : SUCCESS in 8m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aafaa83b92d84d84a046511c7a885dcd : SUCCESS in 10m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e2b37645f2254236acb0f3d4b9e6fbc8 : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/26b75f1d21084ab0997d0772268ee627 : SUCCESS in 10m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/270d179b475843dfa2d58637db651abc : SUCCESS in 9m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/161bdfbb921d47a0b5f3570bb419b3ae : SUCCESS in 12m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f48cc66dc1ac411bb1202fc86dd0e5da : SUCCESS in 23m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/09c0430ad0664f238d6b727bff8e47e9 : SUCCESS in 15m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0501cf115ad248fe8e0a5a367655b4ee : SUCCESS in 13m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1093e233bf374b0ba74c2f1fa34dd0a0 : SUCCESS in 19m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/807484ad83e648239d3f80984e93ce25 : SUCCESS in 5m 35s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/cd9b7ae1f37845b382079cdb297942e2 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b10a306db0b744b48f275749774aaa60 : SUCCESS in 6m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c70cd2cbba4b459799750a5376afc61e : SUCCESS in 11m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c3d54eca5db942459ac13c38f20969f5 : SUCCESS in 11m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f925270880b4315b447104bc0c1751e : SUCCESS in 1h 12m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4452dbec3907491e8652fd8495e231b2 : FAILURE in 1h 32m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ccc829efa3d040049a8cd6bf5b2b1b64 : FAILURE in 1h 37m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c0aea895dab94a10a896f98ebb2e5aa2 : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":1},{"id":"bfbda3b4f7daa478e143df185207ed5817b97a6e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-20 20:23:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9b0f1a32a2d6fbd6426d7509ab83ef506fafff17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 21:55:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/805826713b9344d78b419f6c06a42920 : SUCCESS in 5m 02s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f36e8972632b41ee8d15848c8aa29930 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46f73db1047041f5bcd3b36788b37054 : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a74cb43c90943d69ac4d2e33286865f : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b53dad6b06854734ab8c643464979efb : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b7d22e8aaebe46e8b1105f3d5caf4364 : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e46d822f6419440eb8b41ee7d0792e98 : SUCCESS in 6m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/723d19e1ece74f7db3aa21b5988732ef : SUCCESS in 3m 57s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0d72c53c966c4a2ead00edf10fa0f248 : SUCCESS in 8m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6262c1502364480fbeb8c88a38208799 : SUCCESS in 12m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7bad1616cb1c4297a67c6a08ea29de7a : SUCCESS in 9m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/351c06afb3bb44a3a7608002bce2b0e2 : SUCCESS in 5m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e6d9e0c264524b1893b1df9361ad0d0c : SUCCESS in 8m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/79a2d6a36dc4402f9b05d9bd6899dcce : SUCCESS in 8m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e208c54fb2a545bcb845776f6bd1bc3c : SUCCESS in 12m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b24d832b1eae464e92c325ec2b7f6d3b : SUCCESS in 20m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3dd2cea7100d40839c1899aea29cf01b : SUCCESS in 13m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1131ac54fae9476dad7238c10e8150b3 : SUCCESS in 12m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7f17df1a3b4148628aa5cbebe5636400 : SUCCESS in 17m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3ac1a3a95715473b821304663ee1b578 : SUCCESS in 4m 44s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c150c4eab3b64b9da7014f1f3f862a14 : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6652864d7b564884a245a25f9947b321 : SUCCESS in 5m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c0646636d53941f184c3aa027432ffe1 : SUCCESS in 9m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0ab0d1169ee243118cf536044c65eab7 : SUCCESS in 10m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7cfcf9ca7552444daeeeb57cf3b6d6e5 : SUCCESS in 55m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/506e702ff8da48c8afb196dacdb69a50 : SUCCESS in 1h 08m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2e38eff07e4f4cd6ac8c5a91f195b0f2 : SUCCESS in 1h 31m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/24f442d7c4c3435c88208d4d3dd5ea93 : SUCCESS in 59m 03s","accounts_in_message":[],"_revision_number":1},{"id":"73b62acec2be93a76bf3fabe50fefa4f62533dfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 21:56:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"50719485ec142d7612d86bacc01e9ab9ca0e0e50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 01:34:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b945c76ec5fa474097be64139c2b1b77 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d13443a5ae7420ca30cc321c4e6627e : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d3cd8e8bd83a43dabcf3e207ba2718e0 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a6761bd74004559b3158044de708d61 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b9d26dff77f4b34ba86a8b8fcd18d7a : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c27adb50ba9747638c996c6fcaf98c8e : SUCCESS in 5m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a687325680104449829a51c6131b8c17 : SUCCESS in 3m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d3c5d68aae024326a389daa50ecda9f6 : SUCCESS in 7m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4fbf52f288ff4d6789e98b4d4d2629f5 : SUCCESS in 6m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9db4d9786174436a960b5edc72faea9c : SUCCESS in 8m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0902d06b942d44b58c92fc749835e76f : SUCCESS in 4m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/54757ba5975a4ae18127e5e70ca12491 : SUCCESS in 8m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/20a6b496ff184030ba04f3de63ae1ad5 : SUCCESS in 7m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3ec4b54148b041319f1575c3d0de90cf : SUCCESS in 17m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f07657f9c7a242b28476a665c3f180f8 : SUCCESS in 20m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e5586f9d926f4f2993b2244c1beaea96 : SUCCESS in 12m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e7c3d07f43904fc7a830d023ce8263b2 : SUCCESS in 12m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/745875be13d14cdc8f9dc0f02f4dd89a : SUCCESS in 15m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b4aa37d6d0214f4386dffda5674e8740 : SUCCESS in 4m 28s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e59d662eef7c4067858b8937a0269404 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/71b83c75bc40442ea42fa2a64c0b3668 : SUCCESS in 5m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c1ddc20b23154bba8e8423d24d1556d7 : SUCCESS in 10m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/caa1b83f0bd04db7af939d1c747a603d : SUCCESS in 9m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e95b87c85684d2fa0996957ea81ed52 : SUCCESS in 53m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/825e696bb20049efbf3eca3c74fa74cd : SUCCESS in 1h 05m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c27470120ce84865933915bac3cf8877 : FAILURE in 1h 27m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3ddd66660cb04110b4f8bb86f0aa3422 : SUCCESS in 56m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e781646f2db13b36ab985fd3cd38daf43d9895b2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-21 01:53:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e1938dc29ec1e4deb2b4a7ae0b2c7d8e7e4ce195","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 03:14:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c18ca23d07d848398aafe6bd77c5d2c3 : SUCCESS in 3m 47s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/bad5f0fc8f334c0b8d53be71e069554a : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f02e0dc796c48ddb293d862d0f73c5f : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f77ba963c15d427f9dfbc134c6432d14 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d15f80373a0a4c5695d2ad79e69defcc : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8658deb241f490b848d9caa6524251a : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/709d1cf9831a446b9b6cad1c78f7cd71 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d858c6bc03ca4001a56cb8f811c4b27f : SUCCESS in 3m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f3a167d6962048e0ab436ab9932044b0 : SUCCESS in 7m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6128a16156124ec6896418d49c2ad73e : SUCCESS in 6m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1753c43ec7c741c292f641a24576135c : SUCCESS in 8m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c6d69d0c93a8451c942644b0076d94fa : SUCCESS in 4m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e8ce291d28cb471c91d4afc74578fd25 : SUCCESS in 12m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1b2776e5ec19467391c6d910f14267b0 : SUCCESS in 6m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/942ec15a07514498a493b86952c1b120 : SUCCESS in 15m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d3c8a410124e404bad3ac3f6ec93eaa7 : SUCCESS in 20m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/29001bc7e6c44e27b3e99b96bd897a61 : SUCCESS in 14m 17s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4a138176c2c94a5fa080e88f348e0e29 : SUCCESS in 14m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7aa0c55263664ec3b650bdb200ebe75b : SUCCESS in 18m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fcfa932e59f84b65b8a28017d885cc66 : SUCCESS in 4m 34s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b83349ade2b34443bd0b263b6f36b7ea : SUCCESS in 7m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6897282fba014e3f9ef755062495fa49 : SUCCESS in 5m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b38800430d324fa4935b65c8259affce : SUCCESS in 8m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a7ae957cb0d74c419369fe044f15a076 : SUCCESS in 8m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1331c497df147fc94f943a15e7272d7 : SUCCESS in 1h 06m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2193c6b11239498fa435216a3e5b1e45 : SUCCESS in 1h 08m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/97db6e6c8f2444c6a6e63f9d8da09ccf : SUCCESS in 1h 18m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6239f7a89d744a3182358f93fa4c4249 : FAILURE in 54m 37s","accounts_in_message":[],"_revision_number":1},{"id":"7ecdb0d730d30be88a82e77b97f7d201e16c0c07","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-21 03:30:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6dbf7d121f8824f949abc8fe561887161a045147","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 05:06:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f925692e31e4f2cba49b998d2ac3e39 : SUCCESS in 3m 35s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ae6d7b58a5474679bd2ca9126463aa0a : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20d42b91be80419da0deb53283a54c5a : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/031dc0441d7c401288a7ae101f33a537 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c794c10ed9f94a07830c43ba2b2a559e : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a0794bd0055406d85000b375912b6cf : SUCCESS in 3m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c6f3079fa0b243169f486715f0a99547 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/399bec8e761644a6bb3c47a2e604658c : SUCCESS in 3m 31s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/05ea44a44056434c85d2e1bb715ee19a : SUCCESS in 6m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d69ccb36c1094e3d8e61d1c562510009 : SUCCESS in 6m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1d3920c219fb44c997296c109108e25b : SUCCESS in 8m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8bf721e0f4bc49d2a42869e053a3a7b7 : SUCCESS in 4m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a9d2b3caf79945c7b7a4516de0b0b4c5 : SUCCESS in 8m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ea372136a4f4f0891bfaa02685aaf61 : SUCCESS in 6m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/80d303a295b84bd8ba8108c9cdc015f9 : SUCCESS in 11m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/069c1790b52a495aa54b6ab0fdd45e52 : SUCCESS in 20m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a9625cc7a6c148fcb695d49d4945c9ec : SUCCESS in 12m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/eb88ac36ddc742d4905ef4b4c8107679 : SUCCESS in 13m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b85ec95e39394e95b1e0fdf85ad2392a : SUCCESS in 16m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5e699a252439457eaa1803c197698951 : SUCCESS in 4m 29s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/635cf06bc9634aab8edb2de05ef06444 : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dcc5f408ed3641ae85186ef9121ebcf8 : SUCCESS in 6m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ce254c90d297485683a5adf72949a132 : SUCCESS in 9m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6ebf3297f0ad4b49a56cc5f7b4506e6e : SUCCESS in 9m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6048d05ec4c4c2f85ce1325be4bf098 : SUCCESS in 59m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fdae1c720df34ab59ad2f97d5ba1e509 : SUCCESS in 1h 17m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d7cf93f1643b4a3ca7f287f9527e8139 : SUCCESS in 1h 30m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9b87b0f04bb24025bf7dec7ccfcbb5cb : SUCCESS in 1h 02m 45s","accounts_in_message":[],"_revision_number":1},{"id":"9f184d957cabb83cf31e2d01d507e450f2b4f9cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 05:06:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0897e0f5979f61ca477a2bee881972f168cfcf58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 08:35:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4e8c3a5028539d2cc71252daf856309b185a6762","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 08:35:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79110abc1e734092b554c062f7e9afdc : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ed0752f98d14676a7e45f71ed9e5317 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b679c126150a4cbfa6caf9a02d098861 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1232175f723e4c07801a2f74498232fc : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2c317af327554011a0028c813e3f7eba : SUCCESS in 3m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8a6db80d0fb546da8f44f3af8d75f4fb : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b1ab97f6fcce4bb2b40dd07bf2dcfd05 : SUCCESS in 3m 24s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/da2ed5abcb7d4b89a77a06cd5d1ac02d : SUCCESS in 7m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d2c0f6ffb2b24c4eaedcde382bc9a708 : SUCCESS in 7m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/754f74684919448f9ee683c782000904 : SUCCESS in 8m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8adefcbc110042beb7b8531942398e6a : SUCCESS in 4m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/74061aaf6ec548c38ee95a086b6f8cb4 : SUCCESS in 8m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e30c8efb76c34b1aa002b96127e3f909 : SUCCESS in 7m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c0a69ec2597849ec89e5d66e8047ed52 : SUCCESS in 18m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1a4b6469d33a40eba7d8adc68ecc2290 : SUCCESS in 22m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7780383e99694e1d94098187b83fe3e6 : SUCCESS in 11m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/25639f9d45a54197bc14fab89c023283 : SUCCESS in 12m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0c52b93dc03c46e29d4d20336c729ccb : SUCCESS in 19m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e71e0425c281465284450cd8a2ea14b7 : SUCCESS in 4m 13s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5e8185fd92144bbeb7677131ca97de09 : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5ccae55eb5c345afa1e26c482ba22ad0 : SUCCESS in 5m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4f0447c1ea2d4d10a65c274dd9ccba6d : SUCCESS in 8m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/989b8aa5353f42eea158cfc6f4d0b7af : SUCCESS in 8m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f0608d88b314167a6fe6bb3a47774a4 : SUCCESS in 56m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e5cf9d26fbd749b7bb8a1dbbf2d85854 : SUCCESS in 1h 07m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a105b13749044815aae27250cd9c733d : SUCCESS in 1h 27m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/41cd46d978ae496a9b46379486d2d65d : SUCCESS in 1h 01m 39s","accounts_in_message":[],"_revision_number":1},{"id":"5f3cf96eea398c98975ab7dcaf7052fa94e08992","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 08:37:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca5bf8eb4b164a069b8293e0b675681c : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e9b3a7024456d5b4464fbb5d43eb77df9bb2ed3b","revisions":{"e9b3a7024456d5b4464fbb5d43eb77df9bb2ed3b":{"kind":"REWORK","_number":1,"created":"2019-12-20 11:16:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/54/700154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/700154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/700154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/700154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/700154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/700154/1"}}},"commit":{"parents":[{"commit":"96a06e4fe95bd3cc79e4b739cd06c31fbca62f6e","subject":"Merge \"update constraint for ceilometermiddleware to new release 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96a06e4fe95bd3cc79e4b739cd06c31fbca62f6e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-20 11:16:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-20 11:16:30.000000000","tz":0},"subject":"update constraint for glance_store to new release 1.1.0","message":"update constraint for glance_store to new release 1.1.0\n\nChange-Id: Id968e906136ed74a87090c10df62e0f8704780c5\nmeta:version: 1.1.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Change-Id: If8e94f9e39961c61d17110e8bffcd4057e07561a\nmeta:release:Code-Review+1: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9b3a7024456d5b4464fbb5d43eb77df9bb2ed3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9b3a7024456d5b4464fbb5d43eb77df9bb2ed3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
