)]}'
{"id":"openstack%2Frequirements~700376","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I421ee2d498ea5100d9b52e356b46f2d594638b5b","project":"openstack/requirements","branch":"stable/train","topic":"u-c-horizon-stable/train","hashtags":[],"change_id":"I421ee2d498ea5100d9b52e356b46f2d594638b5b","subject":"Add horizon to upper-constraints","status":"MERGED","created":"2019-12-23 07:02:07.000000000","updated":"2020-01-07 10:27:15.000000000","submitted":"2020-01-03 21:26:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700376-1578086776157-bbef5f2b","meta_rev_id":"ac62fcaf179cffa16ab7863217ecfeb8c97928d1","_number":700376,"virtual_id_number":700376,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-03 21:26:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-01-03 19:42:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-03 19:47:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-03 19:47:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"date":"2020-01-07 10:27:15.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-03 19:42:26.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-03 19:47:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-03 21:26:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-07 10:27:15.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"2055f934a0ac5375f378a213479d1027625d140e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-23 07:02:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9721743f74f6bca1d94bc84f29c58b3ddcd4c07","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-23 07:04:02.000000000","message":"Patch Set 1:\n\nFor stable branches, horizon in upper-constraints.txt would be helpful especially for stable branches of horizon plugins. One example of the failure is found at https://review.opendev.org/#/c/699241/ (designate-dashboard stable/rocky)","accounts_in_message":[],"_revision_number":1},{"id":"e9f1bd6ada9e007bdc0009adeadb7ab988706545","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-23 07:27:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3e505c9bcc66482ece22de2cdb057b19e3ade97f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 09:23:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/64f1479beedb43e48d7e9396569fea0c : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75195f4b3bd84def90a05b3e6a0adc19 : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/810e10c137b3493aa5be951dfa807e15 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a3da3ed00d04760af2cce9ec0828ae3 : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff062be9612446abbda1fda135f0d520 : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e27da984cb2e42799808f5242ff6e3cd : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b94a0170eabe4c1cb91a6a7e7d463b59 : SUCCESS in 5m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/10a9277d77e24991bce417f773b39cf0 : SUCCESS in 8m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8012c599b67c4c5a8086b20dd2312773 : SUCCESS in 7m 26s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a4f57ffdf9e74eb196e7f99c0a87dc27 : SUCCESS in 14m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f58d06fa37c540d79de1260f0409b26a : SUCCESS in 9m 13s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a39ce71e7e704d1e8fa0e4bd81316d93 : SUCCESS in 9m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d6a661c75f8e4d628aab8ca5444b36bf : SUCCESS in 5m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8f2217d7135747d1b4c11bf9c1364085 : SUCCESS in 16m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3ec5e34a5a6b44648792a2a0f3a7fbb8 : SUCCESS in 8m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8c00d0b3115b4e729d1b653ea9ccc749 : SUCCESS in 12m 25s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/059c22dca5644724a1e54dc69c0f5d7b : SUCCESS in 16m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/95d6cba1e83b4f85a3e12bf3ebb769cb : SUCCESS in 30m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c49d20177abc45b3bef0660891a24451 : SUCCESS in 22m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c425885c44db45b79ea4d7987c9c5077 : SUCCESS in 22m 35s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5ee859783c5b4bcabfa35072a77cfc93 : SUCCESS in 15m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/204aa216a8ee4ce3868272612a2c5ad3 : SUCCESS in 19m 23s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/89f209f9235a4f4e8bfb578c56989f50 : SUCCESS in 13m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6a883930a78f4122ae0323c9758fd3e8 : SUCCESS in 21m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6d3ba63b48a34e0fb2c2da2a4bb7cf37 : SUCCESS in 5m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/18c9fb694d194661ab7e61603ed7a950 : SUCCESS in 14m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5fddcc06491748df8a46980d91143cb0 : SUCCESS in 8m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/39c6a994473c4fb6bcd322ec84544261 : SUCCESS in 1h 05m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b746dda7e4b04ae9b771cd9d1ea14d68 : SUCCESS in 1h 53m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/611ee2176fe3480cb999ffe9a8994887 : SUCCESS in 1h 21m 04s","accounts_in_message":[],"_revision_number":2},{"id":"61ec0dd63143d3cfd2504057871e423dbe39c07c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-23 17:46:07.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"4cb71392e510ff8bd0966c0083e7933d19d581df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 19:17:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/12cd74a511114bb7a991fa08ccd367ce : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1498736b755f4d2489f000a64df8dfd6 : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dac9d8339a2740f4a87714f15cddee57 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d75a84cf87d4c76807a3bbbae4f16b5 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2410b891f82c4153844452a48fc3b0e6 : SUCCESS in 4m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/456cb624ee924a9d970cec127b7ecb2d : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a2de2ec4d2d4e34a4f94c5dcf6bd91c : SUCCESS in 3m 52s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ca2e2d6dc49849eb817920503d72bd09 : SUCCESS in 9m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/77346306544a4a3a9e7252ddec8ac1e2 : SUCCESS in 7m 26s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6afe407d01ad4f0a823885d29dfe1f05 : SUCCESS in 9m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4e256fffa9634d0b927a83bb7dff32ac : SUCCESS in 8m 45s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2ff8eae3a781472ca541e856086beea1 : SUCCESS in 5m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ca769d72c71247acbb740c040b214bd1 : SUCCESS in 5m 28s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/012ec10b1993442daeb3b7087928b453 : SUCCESS in 9m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c710eb27241d44219e068313607cefd2 : SUCCESS in 8m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d23d98e4d4294f08a89430fffab852ad : SUCCESS in 7m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/293419cdc13a41a4a10ebd58c94ea3a7 : SUCCESS in 20m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/69902d90aa804feeaf60ce89137c79d3 : SUCCESS in 19m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3016b5155f9e475e82b78226cf4d5531 : SUCCESS in 21m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/64293c1e22ea44eaac0bf642abc18561 : SUCCESS in 19m 46s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3542e49b90c7484786284f8360d1c38b : SUCCESS in 13m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ce027ebf5aea4f289ddb679a1355dae0 : SUCCESS in 13m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a5214a815dd54417a1b6c9ac8a92ea27 : SUCCESS in 13m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5a7bbfe6a5ae44c58ab36954f8c1e1ee : SUCCESS in 18m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/888c0b2b56c94958a28d10e5b27b6531 : SUCCESS in 5m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c5808dc34c764acd92491e883f605195 : SUCCESS in 11m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c9e24f14a25a4766a7e8f3f6b9b8231a : SUCCESS in 10m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/92d35de5fabc4b9880540e3eacd10924 : SUCCESS in 1h 06m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/86a1363af58c479d9a1aaa34aa5c6f5c : SUCCESS in 1h 29m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3948c53c590a4d93b8c5b4eb8cf1bd2a : SUCCESS in 59m 24s","accounts_in_message":[],"_revision_number":3},{"id":"a44b1bd736d3940a421fa763cbefa3078fe20ac4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-03 19:42:26.000000000","message":"Patch Set 3: Code-Review+2\n\nThis will be OK for stable (since there aren\u0027t many other options).","accounts_in_message":[],"_revision_number":3},{"id":"09f278d02284f1f264196318416e736a87c5b88b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-03 19:47:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3302d84b299c77419dd9588b1371d3f005660ebc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 19:52:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8ac86f4b35c7777704eb74eebed3f104be9a5185","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 21:26:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c406e937330469fb0cf99bc0412d62e : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7212caba0da84ef2b3df4ffc02d7e331 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d51df2a573c41f4938a0a3f317dc00f : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a8f7eedc59143ceb33e5dd75046a5ae : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c64be05c4884a71be23716767817a42 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/46637e57d8da43fab367ab0e45e2f8d1 : SUCCESS in 3m 52s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/08d0ea77902545458e40dbd1ba7b2952 : SUCCESS in 7m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d5ebfb33361846d18ce9a01c5a131dbe : SUCCESS in 7m 03s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0fcadaea57ca43e7a4dd55471005088f : SUCCESS in 9m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/518a03abe85c47f1aadaa97194d95962 : SUCCESS in 9m 25s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1269eb46ff004c00b1e3b858136348d9 : SUCCESS in 5m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6b5a92905f2842268b8312855e13f886 : SUCCESS in 5m 12s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/96a5712c38164f0583996ddcfebb396e : SUCCESS in 10m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/089858f9105b4574a03abec2b67d55d9 : SUCCESS in 8m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f0425370e8864f90a1ecca953a4d8910 : SUCCESS in 7m 11s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/93e13c0fe6474d7f815d47581f575331 : SUCCESS in 18m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cb85195463b74df482d005b891a0c993 : SUCCESS in 17m 53s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/657484c26ee74a6f90db00f9fc012e24 : SUCCESS in 20m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/08e04a747ead46698f076bc1f3d523c0 : SUCCESS in 19m 34s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/39aa52ccfb934350bf43ea82989b4deb : SUCCESS in 12m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f7fab784ed52416c94b519c22b47bd01 : SUCCESS in 12m 23s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a446486805634b048f27a62547a76e50 : SUCCESS in 13m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/14b836c0c806446ebefe72392f2410d7 : SUCCESS in 19m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1c4b7db199bb4dff8d330bce3b8bb683 : SUCCESS in 5m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0a8b6ff750434e31aea446e6f60a9590 : SUCCESS in 8m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5e1d1f3290414478bffe57c11381f14c : SUCCESS in 9m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c00d9e90fbbb4f2b996efe1dbb3dc33a : SUCCESS in 1h 08m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e6ad7e20b4034aaf9dd8239649ea3e2c : SUCCESS in 1h 27m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b27bf8dc853648699567afcf626c1e08 : SUCCESS in 56m 05s","accounts_in_message":[],"_revision_number":3},{"id":"7d5a85446e3baa0d0f68908ef9ce596bc00a40d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 21:26:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6c7b9ced5a6472ecc9d9ae425145bfbcb8781368","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-07 10:20:32.000000000","message":"Patch Set 3:\n\nHad to write https://review.opendev.org/701348 to workaround that requirements change to be able to install horizon...","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3e54878f9c87266e1b9775f72af06de27a3dad65","revisions":{"78997e6a3aafb6eebb48a7b30af6ba6644898264":{"kind":"REWORK","_number":1,"created":"2019-12-23 07:02:07.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/76/700376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/700376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/700376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/700376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/700376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/700376/1"}}},"commit":{"parents":[{"commit":"bfd4804c4763ef7b2990871eb5458288d54fc07f","subject":"mark regex as regex for pycodestyle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfd4804c4763ef7b2990871eb5458288d54fc07f"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-23 06:47:46.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-23 07:01:29.000000000","tz":540},"subject":"Add horizon to upper-constraints","message":"Add horizon to upper-constraints\n\nhorizon plugins use horizon as a library.\nHaving horizon in upper-constraints would be helpful\nto avoid unexpected breakage due to a newer horizon releases,\nparticularly in stable branches.\n\nThe reason we don\u0027t have horizon in upper-constraints so far is\nbecause most horizon plugins would like to work and test with horizon\nin a same branch, but Zuul v3 allows such tests.\nThis reasoning is no longer true.\n\nUpdated to stable/train maximum of 16.0.0.\n\nChange-Id: I421ee2d498ea5100d9b52e356b46f2d594638b5b\n(cherry picked from commit a7a4d42b0e20bf000c6aa33aeda97c17089bf875)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78997e6a3aafb6eebb48a7b30af6ba6644898264"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78997e6a3aafb6eebb48a7b30af6ba6644898264"}]},"branch":"refs/heads/stable/train"},"65ceb99bd6f75a88c892f18c8c7ddf2f3463e94f":{"kind":"REWORK","_number":2,"created":"2019-12-23 07:27:07.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/76/700376/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/700376/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/700376/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/700376/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/700376/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/700376/2"}}},"commit":{"parents":[{"commit":"bfd4804c4763ef7b2990871eb5458288d54fc07f","subject":"mark regex as regex for pycodestyle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfd4804c4763ef7b2990871eb5458288d54fc07f"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-23 06:47:46.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-23 07:26:52.000000000","tz":540},"subject":"Add horizon to upper-constraints","message":"Add horizon to upper-constraints\n\nhorizon plugins use horizon as a library.\nHaving horizon in upper-constraints would be helpful\nto avoid unexpected breakage due to a newer horizon releases,\nparticularly in stable branches.\n\nThe reason we don\u0027t have horizon in upper-constraints so far is\nbecause most horizon plugins would like to work and test with horizon\nin a same branch, but Zuul v3 allows such tests.\nThis reasoning is no longer true.\n\nUpdated to stable/train maximum of 16.0.0.\n\nChange-Id: I421ee2d498ea5100d9b52e356b46f2d594638b5b\n(cherry picked from commit 5bf03d8969fb86ffb2d33c905b35129d663bc9e2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65ceb99bd6f75a88c892f18c8c7ddf2f3463e94f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/65ceb99bd6f75a88c892f18c8c7ddf2f3463e94f"}]},"branch":"refs/heads/stable/train"},"3e54878f9c87266e1b9775f72af06de27a3dad65":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-12-23 17:46:07.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/76/700376/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/700376/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/700376/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/700376/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/700376/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/700376/3"}}},"commit":{"parents":[{"commit":"bfd4804c4763ef7b2990871eb5458288d54fc07f","subject":"mark regex as regex for pycodestyle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfd4804c4763ef7b2990871eb5458288d54fc07f"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-23 06:47:46.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-23 17:26:48.000000000","tz":540},"subject":"Add horizon to upper-constraints","message":"Add horizon to upper-constraints\n\nhorizon plugins use horizon as a library.\nHaving horizon in upper-constraints would be helpful\nto avoid unexpected breakage due to a newer horizon releases,\nparticularly in stable branches.\n\nThe reason we don\u0027t have horizon in upper-constraints so far is\nbecause most horizon plugins would like to work and test with horizon\nin a same branch, but Zuul v3 allows such tests.\nThis reasoning is no longer true.\n\nUpdated to stable/train maximum of 16.0.0.\n\nNote that this is proposed to stable branches directly.\nIn the master branch, it is required to split out common portions\nin horizon into a separate deliverable and it was rejected to\nadd horizon to upper-constraints.txt.\n\nChange-Id: I421ee2d498ea5100d9b52e356b46f2d594638b5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e54878f9c87266e1b9775f72af06de27a3dad65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e54878f9c87266e1b9775f72af06de27a3dad65"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
