)]}'
{"id":"openstack%2Frequirements~700377","triplet_id":"openstack%2Frequirements~stable%2Fstein~I421ee2d498ea5100d9b52e356b46f2d594638b5b","project":"openstack/requirements","branch":"stable/stein","topic":"u-c-horizon-stable/stein","hashtags":[],"change_id":"I421ee2d498ea5100d9b52e356b46f2d594638b5b","subject":"Add horizon to upper-constraints","status":"MERGED","created":"2019-12-23 07:05:48.000000000","updated":"2020-01-07 00:51:26.000000000","submitted":"2020-01-07 00:51:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700377-1578358286764-d3a5649f","meta_rev_id":"53f9edc384dfec208d6c82c8d2bdbcbc9eee6ba6","_number":700377,"virtual_id_number":700377,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-07 00:51:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-01-04 15:47:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-06 16:44:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-06 16:44:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-04 03:39:20.000000000","updated_by":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"reviewer":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"state":"REVIEWER"},{"updated":"2020-01-04 15:47:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-06 16:44:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-07 00:51:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c5c97afdca7ceef2517fbe57f62961cc7a29476c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-23 07:05:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7bd1a988b93d2c9b2e807a4af73a373eda1668cd","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-23 07:25:45.000000000","message":"Patch Set 1: Workflow-1\n\nNeeds to wait until the master patch and backports land.\nThe cherry-picked commit hash needs to be updated accordingly.","accounts_in_message":[],"_revision_number":1},{"id":"ad96c46a6fb5f3ff6bb04aac03d0e0b7987f9a90","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-23 07:50:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"470ecc7a5e5b07ee7a0c6d2c460e57edfb684f1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 09:49:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/cb8d1a751e4c4033bac1e26368b77c09 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82c02a39995848bd91d23286835d978a : FAILURE in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c0a8c421cc854001a2f18b96ca1fe799 : SUCCESS in 4m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/928f340f15a24889b4d1709354a5f0fc : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f09002313d8a4803821280de520e033e : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e221f83a0bdc45bd9b8339a1adc28931 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0783e8500f31491ca343fda13a001789 : SUCCESS in 3m 48s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f9cb37eee66241639cb406ed34ccb07f : SUCCESS in 7m 37s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/6d3d94ae63c94258a61e35182ada71ae : SUCCESS in 7m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/69f1b2bd93024b928891746c90265bde : SUCCESS in 7m 07s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/aeec479929c0450b93be9acafc655f25 : SUCCESS in 9m 27s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/f8db9c77e09c490aa8c91e7bc3971e11 : SUCCESS in 9m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4a17b776b58b476ca53c44e250292dcc : SUCCESS in 5m 08s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/38a1d529db8f4674a5ca3d77e2e555ba : SUCCESS in 4m 38s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/48535586dc6e442286e7ce544e8f7a69 : SUCCESS in 11m 13s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/24288b47c224425da8e1e349378ccb82 : SUCCESS in 9m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/083fae8672f54de8b413d6f4a23c5ff6 : SUCCESS in 6m 48s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ea6104f49f734770a59188b19f93a934 : SUCCESS in 20m 58s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/5144116edf574288be2938cffd44f06a : SUCCESS in 26m 02s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a23dd28b0b6e46b8848bdf3929ed66d8 : SUCCESS in 24m 18s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/c9276a58edf44c7282772a85479dc6bf : SUCCESS in 26m 21s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e5e1a2ff8fb54215b068a2e213d5f017 : SUCCESS in 12m 12s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/a939d2db383f4e74bd9d41b5fe421a99 : SUCCESS in 13m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cc2427cb4a504d728e7caa4533e5f9a6 : SUCCESS in 16m 13s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/8048dee628264c1e9cf3b1f5fb0394f1 : SUCCESS in 5m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5029c1957cb94522bc03a845471a34ee : SUCCESS in 8m 50s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/a1cd044fd4704a8f918b665846af14ec : SUCCESS in 14m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/587c806dd2dc46f1a38991841341b19e : SUCCESS in 1h 17m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/96fc0c2aa00c4018b907c6dc9bdd54f6 : SUCCESS in 1h 45m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4a6f1b46749c4db6b76c4de9926b6663 : SUCCESS in 1h 07m 16s","accounts_in_message":[],"_revision_number":2},{"id":"4cd49c52cb0d8656db0c79610eb23d6378d1978b","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-23 17:48:03.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"83a6e3c6e25406652270bd470697e7265e37fb15","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2020-01-04 03:39:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d571ad48a47b0d26cf1d7a982e4199af11ae2c09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 05:13:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6e2968e6d32742b1acb4031504edfdad : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e148c07ac10d4922884d1df5f01d5655 : SUCCESS in 3m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3048882cc95f4eda89a5994f7c55c239 : SUCCESS in 3m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f6b0bd99a73645a4a2325c761b10019a : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdd46b97523d4d9eb725d9b3ebf54cd8 : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8d44056292394a439ce8c229651cc0cc : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0782b69c49574868a1beb0e50b146445 : SUCCESS in 2m 45s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/60dd799487e74d62904b9d406ee51a60 : SUCCESS in 7m 15s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/965f0623321242a2933dd44fb83d0d56 : SUCCESS in 8m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9c9a26007ed94decabde45400a0fe71d : SUCCESS in 7m 05s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0a36857954a946008921602ea3925116 : SUCCESS in 9m 11s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/120a789cf23f4e56b96d7a92ab2e32d9 : SUCCESS in 8m 51s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d6e2466cb29340b2ab6fa1d0baae99b4 : SUCCESS in 4m 37s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/75c7297d2ff94800b46cceb13101e5af : SUCCESS in 4m 18s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/36a2dd88c212468dbb569e8c9e9bbbd0 : SUCCESS in 10m 38s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/c96c2fe7e70c48958e264e705e19484d : SUCCESS in 11m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/52a6b4a664c0451d84da2bf0f875eda9 : SUCCESS in 6m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2077da8a4ddf41629276340ddd5b6f73 : SUCCESS in 23m 01s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/d989fe6b5ec74ceca0bb14a92330b4b1 : SUCCESS in 26m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/455f6876c3f5483d99a18cce102891a3 : SUCCESS in 21m 51s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/ad2987b399144f428d703925e33aefc5 : SUCCESS in 25m 11s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/cec0a47eef4e4cb38c6ae87d2c252a8d : SUCCESS in 11m 59s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/cc8dc312f78d417cba07febc70afc8c9 : SUCCESS in 12m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b324accac0444c96b6b1dde9746ec995 : SUCCESS in 17m 16s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/a857ef3d66064faca8bfc46a0bb5f7db : SUCCESS in 4m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0c1d37da4fa14c18beab5e0336038c4a : SUCCESS in 7m 34s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/8f07e848b2f4410bafaa060377bf1429 : SUCCESS in 7m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/09c7b6c6919f44ba858bb1bc6a178e1e : SUCCESS in 1h 00m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/997dbf10fb5b46bda86fdc61999b31e7 : SUCCESS in 1h 28m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9b3ee48bc5fb4b5f859163c7a901001c : SUCCESS in 57m 10s","accounts_in_message":[],"_revision_number":2},{"id":"bae905e3cdfa53980a8c618fe51736b407ccb5cf","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-04 15:00:58.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"90cbe015cecf57c8fd4c6e303c8582db9d1d5dce","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-04 15:47:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ce912e4c79357092f130b90c81c4b7ea76402693","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 16:34:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/97894f699011433f809db694fe60d15b : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8dd7eadc11e43fbb02d98e9c3bbfa84 : SUCCESS in 3m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/961e8c14887a416096b262f0a1c0f2a5 : SUCCESS in 3m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/32df9d0ebfe44d7d92b3c20d11498d21 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a638cea0dae492e89959cb7cae8c484 : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2904083cc8cd40898235838adb0dbba6 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ad50d2da295b4c4f9f600653bebeb180 : SUCCESS in 3m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a9bf52206cb54f6aa1719fc5dc80dbb2 : SUCCESS in 6m 28s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/e3898605bb634178ab1055ad56da57dd : SUCCESS in 10m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/eed5050b745d4f77805d70907bab240a : SUCCESS in 7m 54s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ce0471893e774b4cbfba57f9f0816e9a : SUCCESS in 10m 02s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/be7d724028da4f228acbb55245ab8c50 : SUCCESS in 9m 33s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6d40fd9aae994fda88e706d02b09dcc9 : SUCCESS in 5m 22s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/68e4619d0b3f4031a9c3b0273fbdce26 : SUCCESS in 4m 54s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/658ec7d31d664a38b565d1c88309f37c : SUCCESS in 10m 06s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/81a402e31d334e859f5713ed99ea3543 : SUCCESS in 9m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f09a3158d6544faab63f88f315b73c7b : SUCCESS in 6m 54s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/71ddcf0723014a48b517e6d109ae002b : SUCCESS in 20m 10s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/b1a81df71fcb46c6a677e3c31945d348 : SUCCESS in 22m 41s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e48767df611344eb931ae9419765a06d : SUCCESS in 19m 01s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/2d33827d6cb24d9a8582885411cf0bd7 : SUCCESS in 20m 12s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/45c59e0714aa4c8dbf353d9f23431faa : SUCCESS in 12m 22s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/828a6bfc2f364817a556853d07d0d331 : SUCCESS in 12m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d49e2bc9861a4bb4803a9c1b6a84849b : SUCCESS in 15m 54s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/2124e9ec981649eca2119a83ffa0d737 : SUCCESS in 4m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e360ca13fbca4baca1957061bbfd8fc0 : SUCCESS in 8m 51s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/0b0a0616de5e4d5588ee2c5d9d9580a8 : SUCCESS in 8m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ebc04a0ef43549168ed2fdceb33054be : SUCCESS in 1h 02m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/af930f7c50194e3f9bb9b7a115da18ed : SUCCESS in 1h 20m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e3d70237db6249ff84e110bb1e9c860b : SUCCESS in 50m 28s","accounts_in_message":[],"_revision_number":3},{"id":"15f47f5afa55db402f0d80b55f94927978ae8558","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-06 16:44:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4f3c6463a6872b1c106e03028190c540b24bc4b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 16:44:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"22192aea2b309e0bee98624e080980d1d67b370e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 19:50:39.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a562186cf98f4101a3ba5b1b6319c7a2 : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd49fca94eb94f75a89acc617ceb98f4 : SUCCESS in 6m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/42d1bb539f224ff68e5f6ea1e26a817e : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfad4c34351c4cc29b8e5cb7f07df81f : SUCCESS in 7m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b3d02df72ec34396af855eacbbda2179 : SUCCESS in 6m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc2d2573e18e4646955c6cccac9e8cc7 : SUCCESS in 5m 37s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/21f7b9febe38445db226233b7194429d : SUCCESS in 9m 19s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/929a593f2bb24c0a9242c5df28137b66 : SUCCESS in 15m 41s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a3df09e943134acaa7ba1b7283c551e3 : SUCCESS in 11m 30s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/a2a15921ebca4977833d09f85460c577 : SUCCESS in 11m 02s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/888ad11deba04223a4f5ac9fb8c76886 : SUCCESS in 7m 17s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/7a917adab7504522b6c92b514ac3de1a : SUCCESS in 7m 18s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/27872bb7b33d46298e24dfd5f9bc8bf4 : SUCCESS in 12m 54s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/c48851a6dea54698b829fdbbdbecf5a8 : SUCCESS in 12m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6302bba7fb6a45b3b4d09222dd44f131 : SUCCESS in 9m 48s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5b0b5f9cd442442781bd67a2878063b3 : SUCCESS in 24m 38s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/bb785ff1b0574bba8aee64a787151399 : SUCCESS in 27m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7971851ccfd5410fad0f98e06709cb3e : SUCCESS in 24m 47s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/5a83c0fee6654cfe8a81b86863958457 : SUCCESS in 24m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/af726986afdc4f4d83c2803be29cfc2f : SUCCESS in 26m 55s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/296d9406d32f4c9ba3dc9431a7daa7b7 : SUCCESS in 15m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4563b423087d48d4a2fead573878b418 : SUCCESS in 20m 54s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/e77cb3b810ac4271b4637f9ed6dccc75 : SUCCESS in 7m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b8b9ba2bbb9c4b3393d3ad959b799e38 : SUCCESS in 10m 48s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/232a71843aed40f7b7cb945c6037c86f : FAILURE in 10m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/54616de478274385b02ff1d3203f70be : SUCCESS in 1h 32m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bafae6c9a5564951a5f314524a1431f1 : SUCCESS in 1h 44m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5a8b5e17e6674d80924b2a25ee53f803 : SUCCESS in 53m 59s","accounts_in_message":[],"_revision_number":3},{"id":"6383f019ae1dbe5aa397af2b6ce80a529bb939c2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-06 19:54:41.000000000","message":"Patch Set 3:\n\nrecheck\n\nSwift failure doesn\u0027t look related in any way.","accounts_in_message":[],"_revision_number":3},{"id":"6bb9c86acdfc204e538a84a62ce6e59c18c12a4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 21:23:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/69f059b38dd6425b96dbbc9aeff0b139 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa7af7d33a8a4824990634fbae18f3ab : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6969316b28d74d3c886c9a1bf1cb7691 : SUCCESS in 4m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/113f1928286d4fe8910efd2810363431 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5b37d371e3145ef9df26641fb806959 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aba4c885c37d41278ab8b1bb6cdaccd5 : SUCCESS in 4m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ec396cfe46846799c5d74d999aae987 : SUCCESS in 3m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3d7cb022947a4953b30244c7a925a157 : SUCCESS in 8m 54s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/d0bddaa2480b45f28d5fe03971f954e6 : SUCCESS in 8m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b6a824cdbfad4152ae13e31583ae1a9a : SUCCESS in 8m 13s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2593b22df4664aaeb5a9b091fc2a2b3e : SUCCESS in 9m 58s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/8b5f1d401fe544c1a8046fd634f788f2 : SUCCESS in 9m 59s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/24b69811006248e6b6eef33576fa029b : SUCCESS in 5m 26s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/2cc0163787dc48d6b2a4c97a81635917 : SUCCESS in 5m 19s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/98097928299045c4afe2ee3f4ecbf179 : SUCCESS in 11m 17s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/104984fdb9a344519fdf568a44d97750 : SUCCESS in 9m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/52d56ba0bee64fbd94d01ec03c34fe92 : SUCCESS in 6m 48s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/29bc3cccb3db467d92963b61f8e4f7e2 : SUCCESS in 20m 24s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/86e4faa7dff34a00925085680559f478 : SUCCESS in 28m 00s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c3d97b53f9274a029cb9d5b533bea1ea : SUCCESS in 21m 23s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/18eff1bf7d52412e86399535641922bf : SUCCESS in 23m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/db96538fb95d4d17af96a41902c8f704 : SUCCESS in 13m 54s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/c59c3c15d10b4517a5a6fc092192cb71 : SUCCESS in 15m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/740acb416b644f899e04c8ef037ea72c : SUCCESS in 22m 36s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/7391ef3003514a34a884a97069ed9961 : SUCCESS in 4m 15s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4d586d1b13b348e7a1ffce9c4da35a6d : SUCCESS in 8m 16s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/b0f8fd8773224686a64ecf47fbd0f0cb : SUCCESS in 9m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c99fc981152f45d787b5b0866771745c : SUCCESS in 1h 07m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4024cebd4233445e96eb4e6d56dc04de : SUCCESS in 1h 27m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/81e4eef726754ba4b5f1ae51897bd8f8 : SUCCESS in 51m 11s","accounts_in_message":[],"_revision_number":3},{"id":"42531735848e043e95f65cdb45ccd4eccb16e02b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 21:23:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"23d754de39652c177fa54d8ddaadb09a65906e2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 00:51:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c81994ecb404b35e2e5c39369749df20d2a3449c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 00:51:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dac8a37a21e4fec927b6c459cda6f90 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/608826292b27495aa084faec95ac25f4 : SUCCESS in 3m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/db1fcd5c4f264dfb912c7782e461828a : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f8864badfac4a31add419e2255cd7cc : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/10068b85b052403caad44c8ea57601c2 : SUCCESS in 4m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d9e523de5f7d4b53802b214a45c25b40 : SUCCESS in 4m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fb2ae8642c684f288179184195478224 : SUCCESS in 6m 52s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/8ff971aff03945a2acbe0658997624af : SUCCESS in 11m 32s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0f40785263ef4998864b22a9ba52080b : SUCCESS in 11m 13s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/4569c49ac9cd433db02176bd0151a047 : SUCCESS in 12m 23s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0564b011db9a475d9729b0e911fec369 : SUCCESS in 5m 12s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/d979fc5a6d584707a3db8c2446d7cb90 : SUCCESS in 5m 29s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4bc61e044ade4c55b24fe49518fabd74 : SUCCESS in 12m 25s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/f94cc6fd65854df9980da7e649a27cd6 : SUCCESS in 9m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce429ff974ae499aa0ee65e7cc9e232a : SUCCESS in 9m 50s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8d98092ba1854285ae384e8fdef9c7b5 : SUCCESS in 27m 01s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/be12aef29210430382b6919974a449bd : SUCCESS in 28m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/874a04d58ab3427d97ca0e91bea2f023 : SUCCESS in 23m 19s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/f521a5cc12c34c8c8c95948977c52ce6 : SUCCESS in 29m 31s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/84cf363b93784fb4918168758eeae333 : SUCCESS in 12m 05s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/178a0be006e04b0ab2261450ad241a6f : SUCCESS in 18m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9faec20dec7f4c8b9808534738f74d04 : SUCCESS in 14m 53s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/2b2be1d6409a4e899a03aa56a0104e84 : SUCCESS in 5m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8fe213432ad04bfc98e175930229c101 : SUCCESS in 10m 33s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/0862547f83d24b8c93a09d284f74e7a5 : SUCCESS in 10m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/54971ab0217d438c811425a7a4ea322e : SUCCESS in 1h 07m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/715d7becf03d4a45bf62cb7da97395e8 : SUCCESS in 1h 38m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7deb0033c4d048ef9454dc0a8e1244a3 : SUCCESS in 52m 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a9d91d58380250b21eb38e644079a7c6828271a8","revisions":{"747083ba9146e2420b7c2df08d679e3cbd447ee2":{"kind":"REWORK","_number":1,"created":"2019-12-23 07:05:48.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/77/700377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/700377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/700377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/700377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/700377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/700377/1"}}},"commit":{"parents":[{"commit":"4f9bf2f2d18784d5fc8fbae5f8e12b87aa482974","subject":"Merge \"update constraint for paunch to new release 4.5.2\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f9bf2f2d18784d5fc8fbae5f8e12b87aa482974"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-23 06:47:46.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-23 07:05:23.000000000","tz":540},"subject":"Add horizon to upper-constraints","message":"Add horizon to upper-constraints\n\nhorizon plugins use horizon as a library.\nHaving horizon in upper-constraints would be helpful\nto avoid unexpected breakage due to a newer horizon releases,\nparticularly in stable branches.\n\nThe reason we don\u0027t have horizon in upper-constraints so far is\nbecause most horizon plugins would like to work and test with horizon\nin a same branch, but Zuul v3 allows such tests.\nThis reasoning is no longer true.\n\nUpdated to stable/stein maximum of 15.1.1.\n\nChange-Id: I421ee2d498ea5100d9b52e356b46f2d594638b5b\n(cherry picked from commit a7a4d42b0e20bf000c6aa33aeda97c17089bf875)\n(cherry picked from commit 78997e6a3aafb6eebb48a7b30af6ba6644898264)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/747083ba9146e2420b7c2df08d679e3cbd447ee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/747083ba9146e2420b7c2df08d679e3cbd447ee2"}]},"branch":"refs/heads/stable/stein"},"e97195514bbc43bf455623f81667a150f6b6f10a":{"kind":"REWORK","_number":2,"created":"2019-12-23 07:50:26.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/77/700377/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/700377/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/700377/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/700377/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/700377/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/700377/2"}}},"commit":{"parents":[{"commit":"4f9bf2f2d18784d5fc8fbae5f8e12b87aa482974","subject":"Merge \"update constraint for paunch to new release 4.5.2\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f9bf2f2d18784d5fc8fbae5f8e12b87aa482974"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-23 06:47:46.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-23 07:29:39.000000000","tz":540},"subject":"Add horizon to upper-constraints","message":"Add horizon to upper-constraints\n\nhorizon plugins use horizon as a library.\nHaving horizon in upper-constraints would be helpful\nto avoid unexpected breakage due to a newer horizon releases,\nparticularly in stable branches.\n\nThe reason we don\u0027t have horizon in upper-constraints so far is\nbecause most horizon plugins would like to work and test with horizon\nin a same branch, but Zuul v3 allows such tests.\nThis reasoning is no longer true.\n\nUpdated to stable/stein maximum of 15.1.1.\n\nChange-Id: I421ee2d498ea5100d9b52e356b46f2d594638b5b\n(cherry picked from commit 5bf03d8969fb86ffb2d33c905b35129d663bc9e2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e97195514bbc43bf455623f81667a150f6b6f10a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e97195514bbc43bf455623f81667a150f6b6f10a"}]},"branch":"refs/heads/stable/stein"},"a9d91d58380250b21eb38e644079a7c6828271a8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-01-04 15:00:58.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/77/700377/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/700377/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/700377/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/700377/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/700377/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/700377/3"}}},"commit":{"parents":[{"commit":"4f9bf2f2d18784d5fc8fbae5f8e12b87aa482974","subject":"Merge \"update constraint for paunch to new release 4.5.2\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f9bf2f2d18784d5fc8fbae5f8e12b87aa482974"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-23 06:47:46.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-04 14:59:38.000000000","tz":540},"subject":"Add horizon to upper-constraints","message":"Add horizon to upper-constraints\n\nhorizon plugins use horizon as a library.\nHaving horizon in upper-constraints would be helpful\nto avoid unexpected breakage due to a newer horizon releases,\nparticularly in stable branches.\n\nThe reason we don\u0027t have horizon in upper-constraints so far is\nbecause most horizon plugins would like to work and test with horizon\nin a same branch, but Zuul v3 allows such tests.\nThis reasoning is no longer true.\n\nUpdated to stable/stein maximum of 15.1.1.\n\nNote that this is proposed to stable branches directly.\nIn the master branch, it is required to split out common portions\nin horizon into a separate deliverable and it was rejected to\nadd horizon to upper-constraints.txt.\n\nChange-Id: I421ee2d498ea5100d9b52e356b46f2d594638b5b\n(cherry picked from commit 3e54878f9c87266e1b9775f72af06de27a3dad65)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a9d91d58380250b21eb38e644079a7c6828271a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a9d91d58380250b21eb38e644079a7c6828271a8"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
