)]}'
{"id":"openstack%2Frequirements~700378","triplet_id":"openstack%2Frequirements~stable%2Frocky~I421ee2d498ea5100d9b52e356b46f2d594638b5b","project":"openstack/requirements","branch":"stable/rocky","topic":"u-c-horizon-stable/rocky","hashtags":[],"change_id":"I421ee2d498ea5100d9b52e356b46f2d594638b5b","subject":"Add horizon to upper-constraints","status":"MERGED","created":"2019-12-23 07:07:28.000000000","updated":"2020-01-07 21:12:36.000000000","submitted":"2020-01-07 21:12:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700378-1578431556838-af565c7f","meta_rev_id":"9e59c850473e546bf1101f6bf92f88168f250ab0","_number":700378,"virtual_id_number":700378,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"date":"2020-01-07 15:02:52.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2020-01-05 06:31:41.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-07 21:12:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-01-07 14:46:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-07 17:41:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-01-06 00:54:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2020-01-07 13:42:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-07 17:41:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-06 00:54:23.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-01-07 13:42:22.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-01-07 14:46:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-07 15:02:52.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-01-07 17:41:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-07 21:12:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b434578dd8033652d4e08b24f1218ac61ee7672c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-23 07:07:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff70bcca079777ed8759d21c616d19158c9c8197","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-23 07:25:25.000000000","message":"Patch Set 1: Workflow-1\n\nNeeds to wait until the master patch and backports land.\nThe cherry-picked commit hash needs to be updated accordingly.","accounts_in_message":[],"_revision_number":1},{"id":"7ce15891ae7059ce3adee0dd1e4095925f3f9b4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 08:49:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d47409b82abf46a5924a743b4b53cb3b : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e45eaae762a46d7b12c5d01c7c5cb6e : FAILURE in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/871d3df01f264445b280ce5136d0851c : SUCCESS in 5m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8033fd6de6974582abdcfc54212db77f : SUCCESS in 5m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c0de86721a6a421bb85304bd5b2ec5eb : FAILURE in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9923f70c9eff48e08ff17ac037bd0ab8 : SUCCESS in 4m 43s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/df49a60e7d1447988c46d33091a56951 : SUCCESS in 8m 06s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/73d38810d0bd4c0e9e8f5aaee2620118 : SUCCESS in 13m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4aaa7df7f79c469d940782b45df5e9ff : FAILURE in 6m 16s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/48b46540deda44ee90c5c48b2d8c8702 : SUCCESS in 9m 28s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/6f90da39c05c4725984aae185e426b8a : SUCCESS in 9m 14s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6db86d7846824b02acc68698e3fe7552 : SUCCESS in 5m 15s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/cb8ed116837f4b8da08eb718e381903e : SUCCESS in 5m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/67f31b1d3ffb418f8bcf9715c9690a35 : SUCCESS in 10m 31s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/e082a6e4c2024f31b33646adcd06d742 : SUCCESS in 10m 33s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/bc0dd7cc05f54387881063457716bf34 : SUCCESS in 13m 12s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/c4bfb07f040f4f118adfa0733c4cb829 : SUCCESS in 12m 41s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/504df95ff23a401b93446614d96aee06 : FAILURE in 4m 27s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/0a43ff7403404f0b97ea193f8b2d8cfe : FAILURE in 4m 46s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/89902dd5f359470a89369fe61eb87ec9 : SUCCESS in 21m 11s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/f97d7bc2187748f8bca8963c6a74619c : SUCCESS in 13m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/19aaa1543f874a6990f4cc1e5a92dd5c : SUCCESS in 17m 40s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/85f71493b34243ffb5f8b0ca5233ed1a : SUCCESS in 4m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/79e7a58a6c314bf29e2c76f06269001a : SUCCESS in 8m 45s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/535954fcc9654e78b8cdb80e3d157193 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7988ad0d24e48c595ed036b0e4f3a00 : SUCCESS in 3m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d2386799b5641b48ef54f45c9f909a9 : SUCCESS in 1h 01m 19s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/af50bfd49ae0401780442a171b6bbe54 : SUCCESS in 1h 18m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/557328ac4404478cb43b0d2a180a41e6 : SUCCESS in 1h 07m 54s","accounts_in_message":[],"_revision_number":1},{"id":"7d82538066314c6df2a0c482b2f92126212997dc","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-23 10:09:18.000000000","message":"Patch Set 1:\n\nAll failures look valid\n\n- openstack-tox-pep8: https://review.opendev.org/#/c/700383/ will fix the failure\n- openstack-tox-validate: this is caused by this change\n- requirements-tox-py36-check-uc: stable/rocky u-c does not support python 3.6, so the failure can be ignored\n- cross-neutron-*: https://review.opendev.org/#/c/699969/ will fix the failure","accounts_in_message":[],"_revision_number":1},{"id":"9fca8baa99353796d0ae18d1e8acdd062f8fac16","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-04 15:06:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5083e723df940601a5342ac407320157f7096d19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 16:58:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f1aa0d4f2a7e45ac887df879363098d7 : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd3a9d1c8fec4e7f91aad1162e4caf3f : FAILURE in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/78ec7952f9764a859e3e99ed8c1229cc : SUCCESS in 3m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a3b79460c56141d1aa04c2a4cbbacb91 : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b6353f74f619482fbf1e5a104eafdfb2 : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2acff3c6d3674c85976beca9b3f3c5fb : SUCCESS in 3m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c48ba9e7d6fb4a9c93f36434fdbd2f43 : SUCCESS in 7m 11s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/3dedd69651fe42849d07478bece42192 : SUCCESS in 7m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/da61d27635d948e0b24adec9c5b1ee12 : FAILURE in 7m 29s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0e7a5ea027af4e63bec1d29fd134f69c : SUCCESS in 9m 50s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/0648f52287514e1baa3908572dbf73a3 : SUCCESS in 9m 38s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/13084981821d4d2e9e48597aace7a13b : SUCCESS in 4m 45s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/c7dd6dad578844d3b14caa9ab532caff : SUCCESS in 4m 13s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/cc8ad81b25984295ae2043b3c7362117 : SUCCESS in 8m 59s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/b51efde84d4a49eba728c00e7fd559f6 : SUCCESS in 9m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/41ed9d6168af4341bee38a82e677b9f5 : SUCCESS in 18m 09s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/b7c786a7ec334c8c842cdcd29a9eb65e : SUCCESS in 10m 54s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e65a67600abc4d13b94d11edf7aeea50 : SUCCESS in 22m 58s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/d579abe262fc4071b3bc8e4699cf8ee2 : SUCCESS in 22m 34s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5713acef1ed5495ea4c7b84f9bcb4a8e : SUCCESS in 12m 43s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/b09833822e514582a75ce7342da8d4dc : SUCCESS in 12m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fb677883157a48f9bfefa1ca47905d08 : SUCCESS in 17m 19s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/0a665c5b85d04b16a77045b6606007b9 : SUCCESS in 4m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/40750b26519949a6a5fbec044fa53d2c : SUCCESS in 9m 29s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/443f6ae90800490cb81feaaf18630505 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed71791b048d4a1cbc6204f83fe7ac16 : SUCCESS in 3m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0001ce1dc3034901bde6467d33aade89 : SUCCESS in 1h 04m 39s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/0d1278c4e87f4c1ca93baa7234e9b748 : SUCCESS in 1h 24m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/274bece9000c4e25a80ac4e216d63591 : SUCCESS in 1h 42m 17s","accounts_in_message":[],"_revision_number":2},{"id":"df604c6fc92bc456ab9e5fdde1a897bcacb6a831","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-04 18:59:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"344a04b69eaeef7e04073a4902d457d862460818","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 20:38:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/29cbda5f738b4931b78830c57959da75 : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1a140fc85a44b69b912ab4c00f33616 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/115f5e2c464c4e54b32b8c12d00430b6 : SUCCESS in 3m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cf4bc6c4b9dd4a639713f945020b5b3c : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/07b547bef6fc4f1baeebccdfd8430353 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5b0212fff1be43afa5d2b914da19e209 : SUCCESS in 4m 24s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2b0b627ca8154ab6a94c5b8fa1545594 : SUCCESS in 8m 02s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/0df6edd5fce245e3a1ee458db3c69aec : SUCCESS in 13m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1e47c22f40934feab65f2e754eb0c77f : FAILURE in 8m 37s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1ca9d2534d5a44aaaf44a67d56fa2043 : SUCCESS in 9m 10s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/ac2029c526b54c6eac5a0ef3441cac5f : SUCCESS in 10m 08s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/222b127b813e4bd696a975fe2243df74 : SUCCESS in 5m 17s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/6e1530ce84d64730902faec7b5ec419c : SUCCESS in 4m 56s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/057615a4c81641c3aa33a0b724d9b668 : SUCCESS in 12m 25s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/f05a2d060a0941af96fd6e34037fefde : SUCCESS in 9m 50s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/80754ec455c14a139cae1ff1ced55bf3 : SUCCESS in 15m 24s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/bcc1c6eea7814760996e89681cce4daf : SUCCESS in 15m 41s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/755174b56da243a59ea4c0f75de19690 : SUCCESS in 31m 33s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/32cfa18c118247349a2dc1318790a9d1 : SUCCESS in 22m 05s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/58d5eec9ec254a7296a578d8319e1f06 : SUCCESS in 16m 37s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/64c74191b01544b690c586d40daf01cb : SUCCESS in 15m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0397697f90cd4de1804f4d457e36ceca : FAILURE in 22m 28s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/538ba16889254594ab49eb994d8d6653 : SUCCESS in 4m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/23f5ad8c9096491d9e626e5ff2b9277b : SUCCESS in 10m 13s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/8a011a56c65f46b188843c8faf893c42 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1916329ae6b44adb911e02e5305324a3 : SUCCESS in 5m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/01ec3d017036444e91c5cfd7767d96a5 : SUCCESS in 1h 01m 22s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/564205e4ca9545bfb008dc0b62eea76a : SUCCESS in 1h 37m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2c05e779659a4ce0942a0156e51c531f : SUCCESS in 50m 41s","accounts_in_message":[],"_revision_number":2},{"id":"c865439598825b8e808f37106311a676908ea37c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-05 06:31:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"06ac9f181c88a2b01fc20567289fb9a22fee940f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-05 08:32:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d747b8ad83d4433ea35fd164bf20f2a3 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b17dd65a07a488bb3f5f357f961c3a3 : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4e1b0214b99248f5b79f1c762fbc07f9 : SUCCESS in 4m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d86a6caa43484edc8463c152903ef7ba : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7ceb3835c0504f2c81d79422039459a3 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b4d2a020654f4aa28d236ae110318b05 : SUCCESS in 3m 19s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/09b0c0da9e6c4ee9bf7465bbffb35e73 : SUCCESS in 6m 57s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/efe049b756fc474ca0efc685abd64634 : SUCCESS in 7m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/38916450dc2b4470b220fe0237b48124 : FAILURE in 8m 52s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/acd0409030b94b53851a6459bb8de3de : SUCCESS in 9m 21s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/7f23f35cf0de4a49a034c5f2fc6bc9f2 : SUCCESS in 9m 02s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b2a51ffc2bf341ffa2b436deed8d963e : SUCCESS in 5m 00s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/335d8f04367b4664a2c52aa07d1b5d03 : SUCCESS in 5m 37s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e3187a896c5b4ceabc0716756aef4469 : SUCCESS in 10m 43s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/3a1272ffbb674574a2f801de535bb50f : SUCCESS in 11m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/44fd5704e9984227b6324554fe5bcea9 : SUCCESS in 12m 32s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/120ef9ff65cc48c3bdaeca53c1a2f5ae : SUCCESS in 12m 22s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ddf916b72f56437f903c4dac3489bb5d : SUCCESS in 26m 28s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/f6918c3636324dd7bbadfdfa9af20936 : SUCCESS in 27m 50s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6b059debe28141ac8d1686ce7c3f2a0e : SUCCESS in 14m 17s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/c52a459061e44e8895a4761a2711f466 : SUCCESS in 14m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/54c767075c764517ab72975d2baaf112 : SUCCESS in 19m 59s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/821acd7ee9c04f0ba56f670e20bc98a3 : SUCCESS in 4m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/89469558a42d47c08d2888c99b931cff : SUCCESS in 8m 18s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/8a26afde77664ceb83965a9005a9f174 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f49ce2932444149ae570b7ed96bad6b : SUCCESS in 4m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6083a3df430e4519a01f682a0b7c4ad2 : SUCCESS in 1h 18m 32s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/0a22c7982dbe4c5ca4c462b358b27db7 : SUCCESS in 1h 24m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/42e6e268ba39466990e65c2af99dc068 : SUCCESS in 1h 02m 56s","accounts_in_message":[],"_revision_number":2},{"id":"b1110e0c571ecc7864ae5e7a8f7b188812b2417a","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-06 00:54:23.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM.","accounts_in_message":[],"_revision_number":2},{"id":"2e8212d90d40c5d5e8c11f5a79774f0319948cbd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-01-07 13:42:22.000000000","message":"Patch Set 2: Code-Review+1\n\nIf I understand well this will fix stable/rocky jobs for networking-bgpvpn and networking-odl...","accounts_in_message":[],"_revision_number":2},{"id":"12032a9d5451944192980a2abdabcd7a79724994","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-07 14:46:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"78ae79aadb2d15334ef21fb593cde1cd7a64a2f9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-07 15:02:52.000000000","message":"Patch Set 2:\n\nARGH.\n\nanother branch in a need of patch in kolla :(","accounts_in_message":[],"_revision_number":2},{"id":"a5a73ca246debbd3c62fbb809e31f871e93f3729","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-07 17:41:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"516ce86f93fef4a667399a5a46cbcec7c45c394a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 17:48:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2e8ad3997d1c6f7956769450a07915e7faca41b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 21:12:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d86f12aa2d7771a5d9368a52d90bdb2f604496b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 21:12:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80acdb616f474f2692b287cd3b305f03 : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b012c345729a439b8bcb984b5f4f44bc : SUCCESS in 4m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f7e1fe57f5824da591176b35f32c81ed : SUCCESS in 8m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b96f3b942e7f480badff3749d3dd411a : SUCCESS in 5m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1d3ef240c5634f699d8e77097eef1f35 : SUCCESS in 3m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ab9f9343d7a04641869492b1ad1fd91d : SUCCESS in 9m 55s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/3afbc9be3dea4130a277edb79388c296 : SUCCESS in 7m 24s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e2d13dba2f574c7298f233fe6c448e34 : SUCCESS in 9m 32s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/cf028cbf49c64425944fa489c9b9537e : SUCCESS in 12m 04s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/bfa75a5fe46845739fd7db1bbcb00606 : SUCCESS in 7m 28s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/2e9af571268847b0b032fd6fe8de9242 : SUCCESS in 5m 34s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8f34520560964c40a71fd0c1ed8610ac : SUCCESS in 10m 37s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/ebeb3ced13d44421a2a1d91f04ef5802 : SUCCESS in 11m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d3143c6906e24f69b55de897a6dccc1a : SUCCESS in 13m 24s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/e1840e8b56024faea0dff66c32053123 : SUCCESS in 13m 37s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/518365091bcf4183a9353f8a6aa56e09 : SUCCESS in 25m 21s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/0458254a2b6649d5a2f8685ad235989e : SUCCESS in 30m 12s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c72b48779a5e4fa181c1626b38b7d155 : SUCCESS in 15m 51s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/f07987df08d74f11aeb95aee9ba0e16f : SUCCESS in 13m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2b2a787e6d6b47548c7b137831933393 : SUCCESS in 20m 19s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/e29336859c1d485fa7876efc7e873788 : SUCCESS in 6m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a78a87623cba42f895d0373987a4d34a : SUCCESS in 9m 41s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/e9768fb9cf6d4513ba226de17867be35 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa1f5560a8f7489d9ec6892ea6140b54 : SUCCESS in 4m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9427147fb51c49ecad16ec17846a518b : SUCCESS in 1h 13m 53s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/ed0b0351317346fa9fa598a24c0d6717 : SUCCESS in 1h 21m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/830b1291d48943c2bac24da509869a17 : SUCCESS in 51m 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6aa3b0ced466089cab5cf9eb529af2c1a4c67976","revisions":{"95212fa2ac94a7ad71524b5393585ebf288bab7c":{"kind":"REWORK","_number":1,"created":"2019-12-23 07:07:28.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/78/700378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/700378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/700378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/700378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/700378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/700378/1"}}},"commit":{"parents":[{"commit":"ae911320c1bdb2a2851f13589ce4751b62605109","subject":"Merge \"update constraint for os_vif to new release 1.11.2\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae911320c1bdb2a2851f13589ce4751b62605109"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-23 06:47:46.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-23 07:07:07.000000000","tz":540},"subject":"Add horizon to upper-constraints","message":"Add horizon to upper-constraints\n\nhorizon plugins use horizon as a library.\nHaving horizon in upper-constraints would be helpful\nto avoid unexpected breakage due to a newer horizon releases,\nparticularly in stable branches.\n\nThe reason we don\u0027t have horizon in upper-constraints so far is\nbecause most horizon plugins would like to work and test with horizon\nin a same branch, but Zuul v3 allows such tests.\nThis reasoning is no longer true.\n\nUpdated to stable/rocky maximum of 14.0.4.\n\nChange-Id: I421ee2d498ea5100d9b52e356b46f2d594638b5b\n(cherry picked from commit a7a4d42b0e20bf000c6aa33aeda97c17089bf875)\n(cherry picked from commit 78997e6a3aafb6eebb48a7b30af6ba6644898264)\n(cherry picked from commit 747083ba9146e2420b7c2df08d679e3cbd447ee2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/95212fa2ac94a7ad71524b5393585ebf288bab7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/95212fa2ac94a7ad71524b5393585ebf288bab7c"}]},"branch":"refs/heads/stable/rocky"},"6aa3b0ced466089cab5cf9eb529af2c1a4c67976":{"kind":"REWORK","_number":2,"created":"2020-01-04 15:06:37.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/78/700378/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/700378/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/700378/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/700378/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/700378/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/700378/2"}}},"commit":{"parents":[{"commit":"ae911320c1bdb2a2851f13589ce4751b62605109","subject":"Merge \"update constraint for os_vif to new release 1.11.2\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae911320c1bdb2a2851f13589ce4751b62605109"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-23 06:47:46.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-04 15:05:56.000000000","tz":540},"subject":"Add horizon to upper-constraints","message":"Add horizon to upper-constraints\n\nhorizon plugins use horizon as a library.\nHaving horizon in upper-constraints would be helpful\nto avoid unexpected breakage due to a newer horizon releases,\nparticularly in stable branches.\n\nThe reason we don\u0027t have horizon in upper-constraints so far is\nbecause most horizon plugins would like to work and test with horizon\nin a same branch, but Zuul v3 allows such tests.\nThis reasoning is no longer true.\n\nUpdated to stable/rocky maximum of 14.0.4.\n\nNote that this is proposed to stable branches directly.\nIn the master branch, it is required to split out common portions\nin horizon into a separate deliverable and it was rejected to\nadd horizon to upper-constraints.txt.\n\nChange-Id: I421ee2d498ea5100d9b52e356b46f2d594638b5b\n(cherry picked from commit 3e54878f9c87266e1b9775f72af06de27a3dad65)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6aa3b0ced466089cab5cf9eb529af2c1a4c67976"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6aa3b0ced466089cab5cf9eb529af2c1a4c67976"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
