)]}'
{"id":"openstack%2Frequirements~700383","triplet_id":"openstack%2Frequirements~stable%2Frocky~Id66a3e4d574fa4da5fbffe6a1aa24c5ea23ab8cc","project":"openstack/requirements","branch":"stable/rocky","topic":"fix-pep8-stable/rocky","hashtags":[],"change_id":"Id66a3e4d574fa4da5fbffe6a1aa24c5ea23ab8cc","subject":"mark regex as regex for pycodestyle","status":"MERGED","created":"2019-12-23 07:34:10.000000000","updated":"2020-01-04 18:47:56.000000000","submitted":"2020-01-04 18:47:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700383-1578163676847-0e1c537b","meta_rev_id":"7d9fbe8fea31ba564558236442229a836033ea27","_number":700383,"virtual_id_number":700383,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-04 18:47:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-01-04 15:46:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-01-04 15:46:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-23 23:30:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-04 15:46:30.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-04 18:47:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b57d75e517e45a8006bb31c6b627dc9e040f297f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-23 07:34:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"74643369d82f2d080e36ba99112d1d8af469e61e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 08:53:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9fa95662d3fb452c97c16c4d761812ea : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ac5c35cd2aa47f9969fb9cc1ec07ae6 : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0ebdfdc2b1ba420f936328d94b099a7f : SUCCESS in 3m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b6fe21a451044831b774df5a40b4c22d : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/67d26cede2344488909b40f1a4330621 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47d23c26938d4a08b97ea8acca21468f : SUCCESS in 3m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e9d83951c18c41fba0009068790cc1ec : SUCCESS in 1h 05m 57s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/b57fbc11cf2c4781bb0e156330a98922 : SUCCESS in 1h 17m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1587cc7c88d84020b6b9c680776edaee : SUCCESS in 55m 15s","accounts_in_message":[],"_revision_number":1},{"id":"98e14ae2794f44162bebc7cc07ec86dd47bf3165","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-12-23 23:30:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"de87e6fa7d53701d84640f80e056d363ade5cfa7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-04 15:46:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"795151522330039ecaa57426bf899a4e96ddb34f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-04 15:46:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nJust fixed cherry-pick info in commit message.","accounts_in_message":[],"_revision_number":2},{"id":"77c5c7e00c4319a03c05483421396f4d907d442b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 17:17:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ace25c7c88fa4afc9bc5f145e1552540 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84cdc4bc96b14ad1a9d6a42920a0a63d : SUCCESS in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5fd0a488bbc40e182edc3765b5e367d : SUCCESS in 3m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/319367ffafd3487491b8b7d86b2b4c09 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c9126f616a5a4a5a985f7901e974514b : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d7a8d6b6d534ea6bed5d16b4bde17a9 : SUCCESS in 3m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bc845c67eb1542c3b6ee7b92bb683ba7 : SUCCESS in 1h 04m 13s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/f4095597196345c5819c850aff0c2490 : SUCCESS in 1h 25m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/51f10564b2304c81b5e6725e7b20aff5 : SUCCESS in 57m 50s","accounts_in_message":[],"_revision_number":2},{"id":"b2fc0836cc4a058b70a1f5b833df3bd089878964","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 17:17:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"56a269c894087be5e27c7a37477bdda7fd55b540","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 18:47:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5be51213d4964394a4afccb64a1c0386 : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/090c52b5e2064bd6b92022ee49617b41 : SUCCESS in 3m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/be0c80ff7b004106a41b010fd560f553 : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/258904688631432ea7d0bcfb1b8d4a7a : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cbaa9995cad485ca91db55e2c10884d : SUCCESS in 4m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/177df76894814346a3b4a5b238d1424a : SUCCESS in 1h 05m 20s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/e52e439cca654ef1af5a23a19625a98d : SUCCESS in 1h 28m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0b96109d48cc4774a3ac2fd6ecfbaa2b : SUCCESS in 55m 44s","accounts_in_message":[],"_revision_number":2},{"id":"999c86c9b9e9fc82c2f8a52cdd6cdc4d116b1452","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 18:47:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"58eb070abc027913e0e4d054af030989cf6fd8a5","revisions":{"a7e66652c735b19f337062720d260cf61beb9977":{"kind":"REWORK","_number":1,"created":"2019-12-23 07:34:10.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/83/700383/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/700383/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/700383/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/700383/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/700383/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/700383/1"}}},"commit":{"parents":[{"commit":"ae911320c1bdb2a2851f13589ce4751b62605109","subject":"Merge \"update constraint for os_vif to new release 1.11.2\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae911320c1bdb2a2851f13589ce4751b62605109"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-12-14 05:15:45.000000000","tz":-360},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-23 07:34:10.000000000","tz":0},"subject":"mark regex as regex for pycodestyle","message":"mark regex as regex for pycodestyle\n\nChange-Id: Id66a3e4d574fa4da5fbffe6a1aa24c5ea23ab8cc\n(cherry picked from commit 698b571749982f61e65814ca3de489177c14d30e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7e66652c735b19f337062720d260cf61beb9977"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7e66652c735b19f337062720d260cf61beb9977"}]},"branch":"refs/heads/stable/rocky"},"58eb070abc027913e0e4d054af030989cf6fd8a5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-01-04 15:46:09.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/83/700383/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/700383/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/700383/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/700383/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/700383/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/700383/2"}}},"commit":{"parents":[{"commit":"ae911320c1bdb2a2851f13589ce4751b62605109","subject":"Merge \"update constraint for os_vif to new release 1.11.2\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae911320c1bdb2a2851f13589ce4751b62605109"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-12-14 05:15:45.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-01-04 15:46:09.000000000","tz":0},"subject":"mark regex as regex for pycodestyle","message":"mark regex as regex for pycodestyle\n\nChange-Id: Id66a3e4d574fa4da5fbffe6a1aa24c5ea23ab8cc\n(cherry picked from commit 698b571749982f61e65814ca3de489177c14d30e)\n(cherry picked from commit bfd4804c4763ef7b2990871eb5458288d54fc07f)\n(cherry picked from commit 86d8ca9eb507196351353abfeab73c6375ac4bbc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58eb070abc027913e0e4d054af030989cf6fd8a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58eb070abc027913e0e4d054af030989cf6fd8a5"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
