)]}'
{"id":"openstack%2Frequirements~701104","triplet_id":"openstack%2Frequirements~master~I19e72d3ddd74c0cbd64f420b4e42f1842078f545","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I19e72d3ddd74c0cbd64f420b4e42f1842078f545","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-01-05 06:38:07.000000000","updated":"2020-01-11 06:02:08.000000000","submitted":"2020-01-11 06:00:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701104-1578722411644-c94174cc","meta_rev_id":"285e150d2cceaa4aa54ac793f923e2e98a41eea3","_number":701104,"virtual_id_number":701104,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-11 06:00:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-11 04:25:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-11 04:25:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-10 19:49:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-11 04:25:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-11 06:00:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"427b1e5513637bb75a26a7ab60621d378dfd2ee2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-05 06:38:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93599f78ee3da5624bc6ec24c42624456691ea6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-05 08:17:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71b61c7c07e24b84bfc81fc57bae799f : SUCCESS in 3m 37s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/2b28474c493c4517ae8fd69cb0b54a8f : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/689f358ffe824394ab5598eaae0d6a02 : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9030afe212744becbdd8cbcef0132b50 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8aaf6689ddf74165b3168024c67ad5c7 : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6577f185ba6048c399a58be0636f00a8 : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4082f84e4a5847aaaa7a7a80fca88e79 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/411987fff9a545d3a6b6f33dd93b4f7b : SUCCESS in 4m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e424e2bd12304c33839a30c5daf9da1a : SUCCESS in 7m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f991d10ea9a94ebdbb3bf564dd0b92b7 : FAILURE in 9m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d99e7a754f884bca8bd31bf5e38ce5bd : SUCCESS in 10m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3911ee13ba104759b9b1ab92037ceaaf : SUCCESS in 5m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a3caa049bff441228a2a3831ce390d8e : FAILURE in 3m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2cc29ab511d647f18d7ac2b911c22829 : SUCCESS in 9m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/883fb8fed70740eda285e49bd2a6fdf1 : SUCCESS in 15m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/85f2c2b6ebd043c6b115c7274d575477 : SUCCESS in 24m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/466d6c258c174832892fc7a7ee135a02 : SUCCESS in 13m 17s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a489f51acd7a40489b43890c1f22e05b : SUCCESS in 17m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a844d3492b6b41ad8901c3fef8c07ff8 : SUCCESS in 21m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b4a6def9ecd445e1b3504de647de255a : SUCCESS in 4m 42s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/dfd9a2d97cbe4b8c937a4950bf610947 : SUCCESS in 7m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/90f223064ad7451dba5dfe63fc3ea8f1 : FAILURE in 4m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/93e99f0bd2d04546bd833ef721b72c05 : SUCCESS in 10m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d36bd10d9e624b48b38efe90c3a85507 : SUCCESS in 10m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2c785a5c04c479bb35281ba4799fe04 : SUCCESS in 55m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/36b02a36650541bbbfe115d776140aa5 : SUCCESS in 1h 09m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/31ca3a1aa2ce4d2a9b2390ff078bb76e : SUCCESS in 1h 32m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f7ab8635960a4a459684e7fd6e3af176 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e51fbabe3bec93862954bd7212b4ab0178139675","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-06 06:17:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ab7a8909ff78bb8fc9bc717432fa24b444773c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 08:54:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b66c916d20f842eab5701691538b2995 : SUCCESS in 6m 49s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5057a1fadd85480993980214b68bc7d4 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5b9c5ff1daa4497a11948499a5ecaac : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b791273644c1469c9a837e35ca23a49e : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5a34224b09e4d2988e55054cb7c78a6 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/748a8c36c9dd49cc9091c3d1cf8d15da : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f53b80a414de42a3a597ceb876371a6e : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a9c97b9c1f1f498d95f25a716503ed96 : SUCCESS in 4m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/10ab7ed18c704a60b62676a4b1c86376 : SUCCESS in 12m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/01437aa86a204921b1eb1fb0811965b1 : FAILURE in 18m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0457e175bbe04aacb19c99b8fe84f02f : SUCCESS in 8m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6ba6c8268581495980359bbe81bdb9b3 : SUCCESS in 6m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cae095fcd6454ae89f8961d30074c789 : FAILURE in 4m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff1d55be3e1b47acb4914732b61aa2c9 : SUCCESS in 10m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4dcc01db559f46f699f461e86d7d6b8f : SUCCESS in 16m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9bba35be21a94662bd63b11e5f9521e7 : SUCCESS in 27m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f8bad98351ed4d209dc9f9f4713d2727 : SUCCESS in 13m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/df626e2a390847d8a320bad86630dec7 : SUCCESS in 16m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/15040b6774594963a51809952b5c9c3a : SUCCESS in 18m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c1ff8965b86e4569b6bd1833c495ce08 : SUCCESS in 7m 40s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d2f2add7e7724f739c1156eb5ad28c7f : SUCCESS in 5m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/82c48164faca42a78e705a5bd1986b88 : FAILURE in 4m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e80af8588f9a4e28a88416bf6283edc0 : FAILURE in 10m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ba271cb3368f400ba1c837d538c1e5d1 : FAILURE in 16m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9dc33dafa8e845babfc43a1ef5ad4e69 : SUCCESS in 1h 05m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/642d8eb2086e41e3b56d64a0832502b7 : SUCCESS in 1h 09m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/15558fb0e7da434e95be5d7ed6103e97 : SUCCESS in 1h 40m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f586d21f0bad41b5bbacdf606651f9e9 : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":2},{"id":"b467fa7cd6093e43974dea5cd60c2bcc139a6b63","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-07 06:16:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"41afc4fec33045b6ffd346e105c6b1d498a6dad2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 08:36:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7821985f444f4b52b511e734b3531222 : SUCCESS in 6m 16s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/63ddf1edf8cd49c0adf3caf70aeb56b2 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/780bc466df4b47168ae7c4f609a0f821 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e281902e2b744a24821ee4bcf7ddac08 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fc2f847a01b4285982ebc1869907d61 : SUCCESS in 7m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a82a1fb1048343e188952f72021cb292 : SUCCESS in 8m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eaf9c12f3503404d926e5817214a8678 : SUCCESS in 7m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3e0b55e1ff4d400ba37a11ba7c1bf196 : SUCCESS in 5m 43s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a5bec6ca293d4141b95c178023176aff : SUCCESS in 19m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7bd2b92688c74bdcb157b006f68ca85f : SUCCESS in 19m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0c82b48ea7b64d298231ac6420f7dc61 : SUCCESS in 12m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cd2e517ce44a422db3cbd8f94f08727b : FAILURE in 5m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3b74b652a22940dc8f1aed4843a6620f : SUCCESS in 9m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/638277176c6a42cabe9705df574d099b : SUCCESS in 8m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/350639d8f67a49ecb307a936c3262aa1 : SUCCESS in 13m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f1c8127b597e49248014a7b0f5e2776b : SUCCESS in 25m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1924ec11eac24ba5840a41947d16fcca : SUCCESS in 13m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/82abf559102f4fdea7e4d4e70867219a : SUCCESS in 27m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4fb1aede06d9454eaf3a15777f433801 : SUCCESS in 19m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/49dc461462ed4e67af3ff95fedd74e7e : SUCCESS in 6m 24s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1fd1fc514b494500a2ffd7ed9dd2352f : SUCCESS in 6m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a00eaf1be9164d56b61876dca851142a : FAILURE in 8m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ec9acfe2f0554ae889966d3e18db56e0 : FAILURE in 12m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/547fbb65abe34524bf83e1b0a2ca04d6 : FAILURE in 12m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f53d44a5f214e5bbba832ba2e96c6ea : SUCCESS in 1h 01m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9b6b2f020cd54e4494879657783c7823 : SUCCESS in 1h 19m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a0b863b7542644e0a71af5ea555bd537 : SUCCESS in 1h 27m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8b0dfc29d71b42dabf3ea95244a4051c : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":3},{"id":"7d5d66f13843fbfaed0899dab481db0a371d61f8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-08 02:21:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0936a42c1a6018a29e8b3348b3c51ced86da3f07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 04:39:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0648f75e43e4037866eaab8e49c379f : SUCCESS in 3m 41s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a4fecc4593f443baa40da85ac4783bd5 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2043fd637f0425bad1ccd57e24a2582 : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d33d3be33d2847449375de41d78a408d : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/305edae309a84db2b941d451ee89f7ed : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/668c01a6bda345b8a62711ef91bd14de : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f3330483409f4e4aaceacd0793ab4606 : FAILURE in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d88dd56ab31341faa4ddfbb66d0ed141 : SUCCESS in 4m 50s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3cae52ff4562458e97d7a0f4edbb8f30 : SUCCESS in 6m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3a462780cd774aa2911d8b2a409398e8 : SUCCESS in 14m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c006b3c90d5b4623bf9857ec9b67fbcf : SUCCESS in 9m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/df0bfa87bfe7470692086b0309646814 : FAILURE in 4m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5f814fe72630464a98d85762f1b29856 : SUCCESS in 9m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bad9d9f948fc4b288fe39b260fd0ebe8 : SUCCESS in 6m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1ae520f27bd341a4a5303cbd0c2b2d1c : SUCCESS in 11m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1d2fac9bd6e9428a920eeb54c03a14fa : SUCCESS in 24m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/06ed917ebd3c464c99233e056d8617a9 : SUCCESS in 12m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/774dc2aff48f4e1db6b09c8304588afa : SUCCESS in 22m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b680a33b5df42d3b1a2c9bc00f4c50d : SUCCESS in 23m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/dd8d4ad36f6847b38c3843a451dbf2fa : SUCCESS in 4m 37s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a41669740ee44bb38d92243c2e7952fe : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/271a96b6be2b4fc38324f7c914942686 : FAILURE in 4m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/df880740e71843c388e0942b18e69559 : FAILURE in 7m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9682b21b26c24d0aa7007bfaf8f24586 : FAILURE in 14m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64076ee712284d9daed3a26d847f7a23 : SUCCESS in 55m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d49fd6aa2e034d73bc9469b88e4e0817 : SUCCESS in 1h 05m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6e3cc66a5fd242f48c4f1c80cf7d16e8 : SUCCESS in 1h 46m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6c4fec69a7c648cb965e5e927a6259f9 : SUCCESS in 1h 16m 23s","accounts_in_message":[],"_revision_number":3},{"id":"f06919c43f3d75a46f393e2bdbe5a572ac46867b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-08 06:15:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a5faf5fb7dfbef1e30a8ecae9db21dfbdce23f60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 08:30:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f71f25d8553542c687158dc57b030b8d : SUCCESS in 5m 21s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/2790b6a79e2f42e8a9216d999e6db8f5 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34287fbccb7c4adbbeec78417d29a5f4 : SUCCESS in 6m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/30d8dbba6e1040d6ac9be54b878fc440 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8189a87be694fc18f4be97910b469ad : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20eae69008fe4d1985e389406cf317a3 : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/828cf51a9d4344acbc4aa75466664a66 : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7a1c7e4dd8c347059ac7bc6aca3c9aac : SUCCESS in 8m 36s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b37b5e790d9340c4bb4ca37c81a38cb7 : SUCCESS in 23m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9a0cc75b5f50477199c990680645c85c : SUCCESS in 7m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6db33ecdc8484b26bcdb4d54496c6c1a : SUCCESS in 11m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e1b4c8e5773643b182344b010a469cbb : FAILURE in 7m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/de92fce7a35f4b05b44e90116908c148 : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cd999d21e82f473e9b3d777e98d8291f : SUCCESS in 7m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/110c82c566814cbba24fc6ccb1265609 : SUCCESS in 12m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f7f4e73c45254b35a20815df3a43fc5b : SUCCESS in 21m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e40a98c77d2c41a9859c97a3ebb8aff0 : SUCCESS in 16m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4c888f0ae19b419e9014dbe46a593658 : SUCCESS in 22m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd165d5b824a479b98785c2277ea7995 : SUCCESS in 19m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5c0a92f2491a4b69ba1c015c92319864 : SUCCESS in 8m 00s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9c4eb88731f746bf9ddbc5d678b3c217 : SUCCESS in 6m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55a6e1c10a4f42f59b1527997eaefcdd : SUCCESS in 9m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/82488e469c9b48fc8c76430c7c120ae9 : FAILURE in 14m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/263299a3dde44c6a806431c413d53d6b : FAILURE in 9m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/394e43eb8b004f4a86a4fa39177d98bf : SUCCESS in 1h 06m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b12e1658ea3241b1843b26b5021403bb : FAILURE in 1h 16m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1c1b52b4a2c7424a811c91e689be4282 : SUCCESS in 1h 28m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/10634d1011ea4fee94888ef09dc3d6ec : SUCCESS in 1h 07m 26s","accounts_in_message":[],"_revision_number":4},{"id":"67eec5ffd7fba7b40674f8c3165529b48286442b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-08 14:42:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"80920f4eb204f08f6920f555c5216e92d5795bcc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-08 17:24:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fd636619ee7293072ec38ed5c0eb154171b307ac","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-08 17:25:06.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":5},{"id":"6ae449c0123fffdcadcca56a3fbe64bb85397286","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-08 17:26:13.000000000","message":"Patch Set 5:\n\npulled in due to django-babel being a child dep of some other random thing, we don\u0027t really need to worry about it as long as it\u0027s not in global-reqs\nremoval of enmarker should be fixed (by adding it to global-reqs)","accounts_in_message":[],"_revision_number":5},{"id":"79d4a5bf44f8d7dc485e9478af1452aedbd6cfeb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-08 17:32:48.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b5a1bcc1c4beafd1b50da45f338f60121b414dc5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-08 17:54:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"091a2a6cce87b15f39badbb3ceb029be77f260f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 20:29:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4da9aa029fa64e9eb00824f83b335f23 : SUCCESS in 3m 46s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c0b4d09fc21a43848c8a842ae6327339 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bb32b5b0a764ec7a00dd5cbbcf2fa21 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ccefec704cbc41ec8caf4fcff5685079 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb4433984fbd4834bd49f42d6cd37bd7 : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a6064612b6e47908ff9322a71062c6d : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/56db3a11ef6f4918870b79c17adfb299 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8b68908e47224d3993a4782402c15f25 : SUCCESS in 3m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0bf0e8ad62454f018b521e5a83c21c61 : SUCCESS in 7m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7089cc3fc4254a8ba80ae12209230f0c : SUCCESS in 9m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e4075ba5af1445158c6c26f0e5553c6c : SUCCESS in 9m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4ea8169235e947ec9526100d584cc998 : SUCCESS in 4m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d57185494f3248cc98c9b7efe36d4188 : FAILURE in 4m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b6ac8b738f441f59e675e49e0fe5a0f : SUCCESS in 7m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/824009944ccb4b448906ac5fa56033d9 : SUCCESS in 12m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d11c5a4f79fd4c11b3a10a27f98f19ad : SUCCESS in 25m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5a7c9bff51c1414eaecf42f17055a270 : SUCCESS in 14m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/410af3907d8546298343978d01b1aebf : SUCCESS in 12m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/67336229a1144dfb8f56c35277c0902d : SUCCESS in 18m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6e1530710dcc479bb2c9a60d61e0cb85 : FAILURE in 4m 52s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ac35fcd66fa94a908334ba39f146691c : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2b7a53e055274c3dbed22c50f0036413 : SUCCESS in 7m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f2bc4de177f0405980a9c745e68bbf0c : FAILURE in 13m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/485108d373f34fed8670c3a836ca7c9d : FAILURE in 9m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19a24cba1bb442d9854ab55095a87464 : SUCCESS in 1h 06m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d18de00b54514ec086bd172385dbb6e9 : FAILURE in 1h 39m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a9f2e97ea312494f94d7d3cd066c0ffc : SUCCESS in 1h 26m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/12ade2b9be3c4d3095624a4cae7f8ff5 : FAILURE in 46m 45s","accounts_in_message":[],"_revision_number":6},{"id":"e6f91d3f42ed0f50120b47460e8d00a1893af5a8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-08 21:01:03.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"d29280251684ee7cb9c9f17b174caef28c20055e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 22:47:55.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8de971b7bb84711b246f9374ccfb725 : SUCCESS in 5m 09s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8444b0b7323742828ee97ef1fddae64c : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c52ada61f70d44ad9db7a77be801fb4b : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b9641733928b4f1694ce486daef5d7ef : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c93c37ab209412dbcd0a938a659c425 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8175a2977ef84cc8bbc4dfe19b902de5 : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e44416a2fbf3420097a8c63052655fd4 : SUCCESS in 6m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fabc338ba26d43d8a20698b6980d5f56 : SUCCESS in 3m 43s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1ccef5c2054b4967a173310cccc9b9b2 : SUCCESS in 9m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/af3babb90ab34a1d88c81c1277bc8ff8 : SUCCESS in 7m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b08f9a52c4f343ea8d948624afab9986 : SUCCESS in 11m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3c64c49d6e604ca7a7b47e82c86c21aa : SUCCESS in 4m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2b559768c2004c0ea64f5393d1a3e774 : SUCCESS in 11m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5e0e0a8cee4e48cb9e6792ed9ff821c1 : SUCCESS in 6m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1472a80761ee45cda2abd3324485c571 : SUCCESS in 14m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/076738126e914a9a9403917f13f992e3 : SUCCESS in 25m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/953e3782c9714950949f3397bc7c3243 : SUCCESS in 11m 12s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e8291bb446c5454dbd041767e1c49473 : SUCCESS in 15m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82e645c25f3448a5a5d5eef4800b37ea : SUCCESS in 18m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e321dc370d364d7da15e5abcf2f36b33 : SUCCESS in 5m 57s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ef61fd0ffd5a472cb992fde9824d2738 : SUCCESS in 7m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/db5031832aa94a7ea9c8c34e89b29d3d : SUCCESS in 7m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7250657ec2a44e4585e7f90fcad7da62 : FAILURE in 8m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cde72c7cd4174d2e9a57ecd2ab5f7ff7 : FAILURE in 13m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14ce789eee5643c18f84403b9e97a4b6 : SUCCESS in 57m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1144f96ccad54fa3b0a8d0de57021194 : SUCCESS in 1h 13m 29s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bf23a5d8814142faa188cacadd1f2760 : SUCCESS in 1h 40m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/aa314a11538a401fa6b7d79b425b10fc : SUCCESS in 1h 00m 57s","accounts_in_message":[],"_revision_number":6},{"id":"81ae9b6c422de74f8678f2402199047fd6f0456f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-08 23:05:37.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"bf3d847df127e6fb1fd3f08d177856964a5bfbbb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-08 23:06:37.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"c8a7904c8dc02e637a4305a8c718ecdaa66926ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 00:37:53.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05b68e2942b64b0cab53e1ea2d3f9be2 : SUCCESS in 3m 15s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d53c0a5352de40f888ed14b0ccd83f64 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06833159ed8b446483b36580a59457c0 : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/690380557b7147058d697933c4440e5f : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/154cc53a8a4d473588765710ddec55b1 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b40f8251ad8841f6870b68b219ebd254 : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/adbd552148de44ccb5d44817378ff70f : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8f3c57b667e84d15a1c7ab9b596ddec1 : SUCCESS in 3m 12s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3bb2fa0e2592457c8dcbfca4141face3 : SUCCESS in 9m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6ca2c70c1c7b416d960366f00b5a5a9c : SUCCESS in 6m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9f118897b5874112b11a60e547fb4506 : SUCCESS in 7m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/37005e58f0974336b3a0cd7d00d7a134 : SUCCESS in 6m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/58bebb36fc9742128364b4a18a73053a : SUCCESS in 9m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d434a94f02304693b57b0d43c02a3928 : SUCCESS in 6m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1c319021fb8d4815a2e62003a4d5b2f3 : SUCCESS in 18m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4c6e35eccca84bf0be9680d4b4dc73b3 : SUCCESS in 27m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e06c9e6723be440a82c87a79a7bea3ba : SUCCESS in 17m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b7265bc0474f41cf8328d6f9262d705e : SUCCESS in 13m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/101563f4c4c94e30a69c202fa3647e9e : SUCCESS in 16m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/56ba08e8322242b7b74407994d96525b : SUCCESS in 5m 43s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/fdabd441aa514cda953fabbc9c544733 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/87ad2dab31e9427cbdc67e95946f896c : SUCCESS in 5m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b64fa610bf4d4899ae0b159b65b025cd : FAILURE in 10m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5e4686d382814c5d926eb0887761efe9 : FAILURE in 10m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97ee6101916e43c69fb659750b7a87e8 : SUCCESS in 57m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/00d74a4203e4465b9034d2e91e5a453a : SUCCESS in 1h 10m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/71b05bd25357486cab4503e612245270 : SUCCESS in 1h 26m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7a51c598d3a44b9dbc8c3b19f7aa0a1d : FAILURE in 1h 01m 19s","accounts_in_message":[],"_revision_number":6},{"id":"c19be2df7140a203ecd5c791097b8a58be79807d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-09 17:37:27.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"883ffa189c9b8f427929a8fcb7e78808405b0869","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 20:23:08.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19b35930d2b54a04abf78b460cbd6dbf : SUCCESS in 5m 08s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9cbeb986f2e6482c838e3eb5c765635d : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc5476fe0e2b4a5f9b794e4f3340bfe1 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/763afd99e6f74f928aa1b8e23300489f : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35929a2bca124c50a09145f5eab3c371 : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aaedcf904ef34ca0b29fc96e2c8f64ba : SUCCESS in 5m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9962245b9db946f3967953fe2c76d2f2 : SUCCESS in 4m 32s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/23a8d323c99c40c58807fd517e6066e5 : SUCCESS in 6m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0e61ea85f6f5416dbb304a171a2c658d : SUCCESS in 8m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9868078bbfe740ef9a72d4d538280931 : SUCCESS in 10m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e437a42cf5bd49b5a823131b58d92659 : FAILURE in 6m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ea0cea55475b4f2e9a3632c61f7a7cdb : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b47b0bf8656440b19139e24125dda9aa : SUCCESS in 9m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/14c55665fff5466081dd5ab6ca5a27b0 : SUCCESS in 19m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/248da3475546497da7806b32f40712c4 : SUCCESS in 29m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4ef08586f65446d083b4ba82085f8194 : SUCCESS in 12m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/53877327595e4aae9c809c066d12265d : SUCCESS in 15m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e055435067e542d2b4e7e1b8ad38d3c8 : SUCCESS in 18m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ea3671511b084148af50692c85e94c95 : SUCCESS in 7m 19s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4036a221e2244b79a88571fade409f47 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/10cd793a55374a3cae0a814a77b6029c : SUCCESS in 6m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e5fcc33b06a741c19741bb88389e8f87 : FAILURE in 10m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f816643458524f8592ee947c711f59db : FAILURE in 9m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05fc5e174681476bb36cba3167620c10 : SUCCESS in 59m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca1d81f894cb46f6862b7a99bede4a47 : SUCCESS in 1h 10m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1239f472c38740f1be50d10ddd3cab2e : SUCCESS in 1h 33m 09s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c9d01b6ed5394b69bd4359e120d62f55 : FAILURE in 1h 33m 25s","accounts_in_message":[],"_revision_number":6},{"id":"3d514eda88bccb87ab6fcfa17a9f0d2ceb09701f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-10 16:42:06.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6926e4d8724621bee331cead6fbc261d5181782b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-10 16:43:01.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"60e3b7fe394f8edeaec231b33aff6af327b8b341","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 18:25:13.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baca1a6a9f3c4945849468afc925105e : SUCCESS in 5m 00s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d734dc4ba4044b69ab18071cacbe1088 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7767116d446e46968396e56063f0cd06 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3de2814352c4bd5be167c35b4a6602e : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7d88df481134e94ad5c4bf05d857af0 : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d244a20d8bd24f188557d3c866fcff28 : SUCCESS in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d953f52110bf42b89a45262be9e89a49 : SUCCESS in 3m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f2676689958744c5bc10d02ccd1e500d : SUCCESS in 8m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/db0d50e3c88b4502bbbb4f80991ff42e : SUCCESS in 7m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dc33be57cd7a4f22b912484d29d4ef25 : SUCCESS in 10m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/32945e7c01c745939b6d2b55e470826b : SUCCESS in 6m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c00df12421854113ac87ae0dfb322418 : SUCCESS in 11m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cca66fda5ca74baeb812c913c98a48fe : SUCCESS in 10m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6905e367b7344bfba1cbc30644f9d61d : SUCCESS in 23m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/501c2ea6450549809a9bd8f742a54e37 : SUCCESS in 24m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ed328e09bb8049338f26386467cb4a1b : SUCCESS in 13m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/114851dea002429c987e18daa0077f3a : SUCCESS in 13m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3b15bc4e058c4c0c8e25ffba8686293f : SUCCESS in 16m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5119ac19fa714404a1aa6b43270d99ac : SUCCESS in 4m 18s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/02bb9d75df9443afa14e816895ff6936 : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d7ce1bf26d1f464abbf0da4893682f8d : SUCCESS in 7m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6fab4841cfe942dd9cbab4138af98643 : FAILURE in 12m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5e987ac52b9c490aaa75d4990b62a461 : FAILURE in 15m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f602541e99764bef9ce15a779c0c5499 : SUCCESS in 58m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ffbd1a729a294ee0a42f55044d4b0e72 : SUCCESS in 1h 14m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6137734fd28f48d3958e0a661abfbb4f : SUCCESS in 1h 23m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/11d335da8b7f4a4599c2e59559f9e44f : SUCCESS in 1h 12m 12s","accounts_in_message":[],"_revision_number":7},{"id":"1008feaa72195453162f099c4467247201e71b0c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-10 19:48:20.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b31b661a2969e804d7d015789939d997c73dac5b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-10 19:49:57.000000000","message":"Patch Set 8:\n\nAh, boto seems like a likely culprit for swift failures. Hopefully this makes the tests happier.","accounts_in_message":[],"_revision_number":8},{"id":"bc865f4ccb92d87f7ac119da35d1004271f71576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 21:24:49.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20cc80974e42441492208e258cebf609 : SUCCESS in 3m 48s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/495737b4e7234a5aa692ba680ac53be5 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ad5579b5edc4c2b8bcffead14110152 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f90f520d5594051b65dfe7ea0f50bd6 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad74db03faad44a080e9ad41b9524bc7 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b62c297041f4780a7950b17d7d16e7e : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4c802fa6d2434f7ea3358ca6b70f1ff4 : SUCCESS in 3m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/dd932f68e4de4623a2290a9ae5e8d86b : SUCCESS in 8m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4c13acc565864e068bff7ae78eaf2fae : SUCCESS in 7m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e8cfbc4e409c4f16bf39b983233b876b : SUCCESS in 8m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/91c6f4ace3fd4ffcbcc9aed30dd17025 : FAILURE in 6m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/aed674eb6d0e4abfb0b518a5ca3e2152 : SUCCESS in 9m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8953455ca1884225ab4d231e41cdf086 : SUCCESS in 6m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/96f7025f94dd496abde5eea437847e83 : SUCCESS in 24m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1d0fc977c5864e8ebfb621405cdd4c65 : SUCCESS in 25m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ea2aeead14d84616b75e004922420ff6 : SUCCESS in 16m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/31df1dc5870346918917c63fb7594d33 : SUCCESS in 13m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/99c40ef40c414d67b40d44efebfed3ca : SUCCESS in 18m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/97c5705ea2b44178bcb4393c5dfd6b9b : SUCCESS in 4m 11s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/37d420710b63432281df032f28f68200 : SUCCESS in 4m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4dbee99f271b467a90fea568ad3a56bf : SUCCESS in 7m 15s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8e1d24e201b44764bfbe2ac1e570ff21 : FAILURE in 10m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/82f1a28fc3844849a63f1615a3988b27 : FAILURE in 13m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/511143870b094c1e9a043c69d3f243ea : SUCCESS in 54m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/51ca40f0b132405d958c132a62abc8d5 : SUCCESS in 1h 10m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b893742c6c7b4d0b9e592eb5f3f44dfa : SUCCESS in 1h 25m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/20b3f18954b3404681bc09a226ef2733 : SUCCESS in 1h 02m 27s","accounts_in_message":[],"_revision_number":8},{"id":"a5413c50ff24b7651079a88f1b0be2bfa0f096bf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-10 21:34:45.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"673360c92686ad65721486c3d0e5718c66be0676","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 00:17:07.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f54f1b8927f7403bbe90091816b67ed2 : SUCCESS in 4m 01s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/372e721e564d437d9cdf9ecdac7b0e42 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4409619d35848cda3016dc5929c15a0 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bd85fede9094332a3677007b4d30f26 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab767a98dc3840479cc6fce2b2c3950e : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b7e07f17540b484a9c7e16d82e7915a8 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b3d9346398aa4edd88f2d238c5980dbb : SUCCESS in 3m 18s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/293e714f47794c5b86bc134fed8bdf8f : SUCCESS in 7m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c6dc5329243041caa58ca12c29a62ef4 : SUCCESS in 7m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d8bdeb9343cf4facba42aa448a519aa3 : SUCCESS in 8m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8d19e0331b8b47dd9baeee8af0e683b3 : SUCCESS in 4m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7745a579164448249b6db31c9918daae : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6d071bed41345b2b56a682aba712299 : SUCCESS in 6m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4af49e3600a94e6f92a6ae9c98395d84 : SUCCESS in 13m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b1be9f5c71b04ca296f2975163db7c85 : SUCCESS in 24m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6a1ed5c98adf47b8affeba7aaf889495 : SUCCESS in 12m 04s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/473e403cbb594adc9c2791a779883a27 : SUCCESS in 13m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a424953b31204dc48cbfc0ed3705b002 : SUCCESS in 15m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/638ee626a4cc4c138057fb03ac652a52 : SUCCESS in 4m 26s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e19de42e79744bfba1a72e3d45cca592 : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/78392f4e3f0c4d7bbd21cd448fcbc800 : SUCCESS in 6m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0f1bb928ec9848acae095bbd90676082 : SUCCESS in 8m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5bdb41b54eab4343a5d0eb21f9ee8edc : SUCCESS in 10m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55c3d174a6584fb587b2dff69f0eb051 : SUCCESS in 58m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/119f8b60cb7e4fba934599e73e8b7a13 : FAILURE in 2h 09m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3b0c8b5a41fd4af689f46794c077667d : SUCCESS in 1h 25m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/87f7cff55eb345a2ac39f600c53b4712 : SUCCESS in 50m 50s","accounts_in_message":[],"_revision_number":9},{"id":"1d6c88a0cf52ee3fafac61d2c984a57048316d61","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-11 00:19:03.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"e78e76e4627f2c623b9f029ecd7df2e5e8f65308","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 02:02:11.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d3c17b479174216b81740b79fe927f2 : SUCCESS in 3m 57s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/05e81f809dcd4e8ebd29704922a9fb17 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/599a2057ac8942448c8fa28fb29b7c8c : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8e35ebf2d1845208656ae3305f05cb3 : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ccaa647a90e4f258092f21cf1ede2e0 : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9fa9becb3ea24d6c86089950a2a9914d : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eddccf57203a4447ad18be78105ca161 : SUCCESS in 4m 45s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9afa3311fb264bb4871b744b0eea9df1 : SUCCESS in 9m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dba84e519b7749aeaabcf4b8fe40215f : SUCCESS in 10m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0ee33b40de60427b8866e5774255bf2c : SUCCESS in 9m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a1a1698003dc4b5b8100881afdb93236 : SUCCESS in 4m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e6a4fe49190648438ffed2b0ae685ce9 : SUCCESS in 9m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/405d29fcd341495bb7e462bef6a711b5 : SUCCESS in 6m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f7edceacfe2e47f782f91a23610e84c9 : SUCCESS in 12m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/114cf489b7e341c4a0483488879b60bc : SUCCESS in 27m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2c2d9de0fc0441fcb5dd9f0de99b58d0 : SUCCESS in 12m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b2a4b820bcb04fddbd5657e533c5f3c2 : SUCCESS in 15m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c95094a044364c15a6669c59eb2bcc99 : SUCCESS in 16m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0afef77bab6047d3889a25c59616bf1a : SUCCESS in 4m 27s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3c14ff83fb034eb086fa262c03b1189f : SUCCESS in 8m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e25d6a73e03149638f2fdf7c346fe08b : SUCCESS in 5m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0123b601a1784b41b01da8be31ec6df5 : SUCCESS in 11m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e9fbd4a1f5dd4a0f9f4a9ff025041dd6 : SUCCESS in 10m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e0b3f61a0ac4ca3a7e39df8bd7fc2e2 : SUCCESS in 49m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd9e10fdef8c40ccb1cc4097fd42657f : FAILURE in 1h 22m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2ac25c829aff4a9285157b0c833b72f8 : SUCCESS in 1h 41m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5062df10e3634af2b929e12c5eb72e5b : SUCCESS in 57m 15s","accounts_in_message":[],"_revision_number":9},{"id":"fafca590db8e107879ab2ff15d38fe96bfdbb769","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-11 02:10:56.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"5f1d2799bd49b7c5dc57e3671bc5fcd558a026ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 04:01:09.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d6ff576cf9d46798f100b496bbcb7b8 : SUCCESS in 3m 33s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/3cd2f3dcf82e4e62b35b686bc47b450d : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47d273411814434284c542efd8b0911c : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6dfee886d76a450fb7fc2d2c1f68f88b : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d52e9c43505048ac8c5e610e74c1012f : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/383468e105784fae8259ac0e8a51f907 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6561ab3981bc4aeb8e75dfa0e4312340 : SUCCESS in 4m 00s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a082746678574ec192fa2e500183820b : SUCCESS in 8m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1e9ee526efa64542a4cda4bcc2d38a6c : SUCCESS in 7m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2aa8ffebad1f4dd4959034d790bcb181 : SUCCESS in 8m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cd1ecc2fb5a34a3ea744bea240cffaaf : SUCCESS in 4m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0474e8afd1994fbbafd0c238ff3fa556 : SUCCESS in 14m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1cff2212861045c389bfc69e98537bb1 : SUCCESS in 8m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7df1ff421d6e4c6db0d32432a06322fc : SUCCESS in 18m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7ccf8bd48e604036a1888a359ab309b3 : SUCCESS in 22m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ab2e79dec9a14641b95d0cd01138f18f : SUCCESS in 13m 23s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7d6b712e2376497abc972c70e2809758 : SUCCESS in 13m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/acbdb8ca3f2d4cf9878349c2b7e589ce : SUCCESS in 14m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/df1a6ddaf759469a87e592dd2dac06a2 : SUCCESS in 4m 23s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0d3f8e9897894d60b0891ca3709a49df : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6ce5614579046d484ebcef2265bd159 : SUCCESS in 5m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/96dadf265f3048c5a0e5dac4502eefa7 : SUCCESS in 11m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/430c24a9095d43afaa1959d1d56efdb7 : SUCCESS in 10m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/512962bc1143491e850421064552905a : SUCCESS in 50m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/912082dd56df49cb9b41fc0bec2406ef : SUCCESS in 1h 10m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/39dfce11702c471488943f8459856c15 : SUCCESS in 1h 32m 18s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a112d33c3bdb4f3c8ddd4c8086d3caf2 : SUCCESS in 56m 50s","accounts_in_message":[],"_revision_number":9},{"id":"c680b49b1fd50502b49ce7c5ba41f0f3b9e33712","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-11 04:25:12.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"0acbbf3c5edb0a3f3f3cd7b2b760f07a528c7c81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 04:25:24.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"8ef7d08cb687d8f41a77cc0e97241fcc87ce3555","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 06:00:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"c49edec71b5ef5129577af79be6433d448eb2c5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 06:00:11.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffe9e67858ca42a5b5362230e94942b9 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1d78ee96ece453b9ee0f9e07e1aa0b1 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c585db99b3a8404f887631045de3361c : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d33af5c57424a0ba29ecec0222c5180 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1a1c85abab4463db094ec2c945d7fe5 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/09c302af81364bf5905c4a282de1e4c0 : SUCCESS in 3m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1aec18fd5706469795dda869900a461c : SUCCESS in 6m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a119a2e0f73046839ea28bf21a27bfb1 : SUCCESS in 7m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8a10bbed47d4490fa1c878bf25ca7745 : SUCCESS in 8m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/21d414aabb734810b1a962b6b6a27002 : SUCCESS in 4m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1958d2814ef04f08baec83a89548b35d : SUCCESS in 8m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cd9adc8ad15f4751b8d62b25d18c688d : SUCCESS in 7m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/06f2ccc661e541c58e68501b41ebe266 : SUCCESS in 18m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2056661cc1f04fddb50ddb27531756f1 : SUCCESS in 23m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2487135d43ed4dd6a8eab4715a3487d8 : SUCCESS in 13m 09s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/12ecc2d259304ffe92dcaf6ad2a9efbc : SUCCESS in 13m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/beb460f842634870997df36cd47c554a : SUCCESS in 14m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4b62082ac9d74c35a80d5a3fffed0bf0 : SUCCESS in 4m 51s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f9b8c0e25609430fb373807191e1b0cf : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4ebe573c3710402e8d11fae5a22baf5f : SUCCESS in 5m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5aded0e9153a4f998ab5f62e7df1524e : SUCCESS in 11m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9bbc16942e614a3c9e16e0553e106b75 : SUCCESS in 10m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12f31b3038f04fbaa18bfc009a59acfa : SUCCESS in 52m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bd01aaec63454cb9a5cec64d1efe540d : SUCCESS in 1h 09m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cd0539e8ae1b45a1859fbb51bff0b826 : SUCCESS in 1h 19m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b4270205d493486f8911fbc9f2bc3848 : SUCCESS in 1h 04m 53s","accounts_in_message":[],"_revision_number":9},{"id":"285e150d2cceaa4aa54ac793f923e2e98a41eea3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 06:02:08.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1647ee77b48a4968831fb697edc43ee2 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"7ac99875fa8e85db6984c9e178a7c8b8f0d83a00","revisions":{"4b08e358dbf3c3c19899e2791979b7fe86cc8e4a":{"kind":"REWORK","_number":1,"created":"2020-01-05 06:38:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/701104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/701104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/701104/1"}}},"commit":{"parents":[{"commit":"1bd226f4bbf79454e439ff91fbd032feda55d414","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1bd226f4bbf79454e439ff91fbd032feda55d414"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-05 06:38:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-05 06:38:02.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I19e72d3ddd74c0cbd64f420b4e42f1842078f545\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b08e358dbf3c3c19899e2791979b7fe86cc8e4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b08e358dbf3c3c19899e2791979b7fe86cc8e4a"}]},"branch":"refs/heads/master"},"8c966ed9288db53cf011eb4f528b78b94cacc829":{"kind":"REWORK","_number":2,"created":"2020-01-06 06:17:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/701104/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/701104/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/701104/2"}}},"commit":{"parents":[{"commit":"1bd226f4bbf79454e439ff91fbd032feda55d414","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1bd226f4bbf79454e439ff91fbd032feda55d414"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-06 06:17:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-06 06:17:53.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I19e72d3ddd74c0cbd64f420b4e42f1842078f545\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8c966ed9288db53cf011eb4f528b78b94cacc829"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8c966ed9288db53cf011eb4f528b78b94cacc829"}]},"branch":"refs/heads/master"},"aa23eacb0582822ef3410e0e093f24e2673dec81":{"kind":"REWORK","_number":3,"created":"2020-01-07 06:16:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/701104/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/701104/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/701104/3"}}},"commit":{"parents":[{"commit":"c285569b069af39f306571ba21b7ea251f9ce722","subject":"Merge \"update constraint for keystonemiddleware to new release 8.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c285569b069af39f306571ba21b7ea251f9ce722"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-07 06:16:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-07 06:16:56.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I19e72d3ddd74c0cbd64f420b4e42f1842078f545\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aa23eacb0582822ef3410e0e093f24e2673dec81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aa23eacb0582822ef3410e0e093f24e2673dec81"}]},"branch":"refs/heads/master"},"fd8c8b74a30dd62d666e2ae2742ac95efff0f6d3":{"kind":"REWORK","_number":4,"created":"2020-01-08 06:15:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/701104/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/701104/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/701104/4"}}},"commit":{"parents":[{"commit":"1a6430f8785b7f7d1f9b6fc7beb6694727c00a73","subject":"Merge \"update constraint for neutron-lib to new release 1.31.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a6430f8785b7f7d1f9b6fc7beb6694727c00a73"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-08 06:15:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-08 06:15:33.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I19e72d3ddd74c0cbd64f420b4e42f1842078f545\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd8c8b74a30dd62d666e2ae2742ac95efff0f6d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd8c8b74a30dd62d666e2ae2742ac95efff0f6d3"}]},"branch":"refs/heads/master"},"45d3e2bf86c1679fdab377e261a5c57062cc9cdc":{"kind":"REWORK","_number":5,"created":"2020-01-08 17:24:39.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/04/701104/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/701104/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/701104/5"}}},"commit":{"parents":[{"commit":"1a6430f8785b7f7d1f9b6fc7beb6694727c00a73","subject":"Merge \"update constraint for neutron-lib to new release 1.31.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a6430f8785b7f7d1f9b6fc7beb6694727c00a73"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-08 06:15:33.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-01-08 17:23:44.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I19e72d3ddd74c0cbd64f420b4e42f1842078f545\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45d3e2bf86c1679fdab377e261a5c57062cc9cdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45d3e2bf86c1679fdab377e261a5c57062cc9cdc"}]},"branch":"refs/heads/master"},"ca557d50a49c2d020b6481807c8243a3d62030c7":{"kind":"REWORK","_number":6,"created":"2020-01-08 17:54:23.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/04/701104/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/701104/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/701104/6"}}},"commit":{"parents":[{"commit":"1a6430f8785b7f7d1f9b6fc7beb6694727c00a73","subject":"Merge \"update constraint for neutron-lib to new release 1.31.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a6430f8785b7f7d1f9b6fc7beb6694727c00a73"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-08 06:15:33.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-01-08 17:53:53.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I19e72d3ddd74c0cbd64f420b4e42f1842078f545\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca557d50a49c2d020b6481807c8243a3d62030c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca557d50a49c2d020b6481807c8243a3d62030c7"}]},"branch":"refs/heads/master"},"ff671476824e46bf83c07b1d4c696509d5659298":{"kind":"REWORK","_number":7,"created":"2020-01-10 16:42:06.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/04/701104/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/701104/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/701104/7"}}},"commit":{"parents":[{"commit":"1a6430f8785b7f7d1f9b6fc7beb6694727c00a73","subject":"Merge \"update constraint for neutron-lib to new release 1.31.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a6430f8785b7f7d1f9b6fc7beb6694727c00a73"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-08 06:15:33.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-01-10 16:41:42.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nheld back pyyaml and packaging\n\nChange-Id: I19e72d3ddd74c0cbd64f420b4e42f1842078f545\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff671476824e46bf83c07b1d4c696509d5659298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff671476824e46bf83c07b1d4c696509d5659298"}]},"branch":"refs/heads/master"},"3de60d0f8b15a42fb07ea01fd322f17ed2a25e1d":{"kind":"REWORK","_number":8,"created":"2020-01-10 19:48:20.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/04/701104/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/701104/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/701104/8"}}},"commit":{"parents":[{"commit":"1a6430f8785b7f7d1f9b6fc7beb6694727c00a73","subject":"Merge \"update constraint for neutron-lib to new release 1.31.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a6430f8785b7f7d1f9b6fc7beb6694727c00a73"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-08 06:15:33.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-01-10 19:48:02.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nheld back pyyaml and packaging and botocore and boto3\n\nChange-Id: I19e72d3ddd74c0cbd64f420b4e42f1842078f545\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3de60d0f8b15a42fb07ea01fd322f17ed2a25e1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3de60d0f8b15a42fb07ea01fd322f17ed2a25e1d"}]},"branch":"refs/heads/master"},"7ac99875fa8e85db6984c9e178a7c8b8f0d83a00":{"kind":"REWORK","_number":9,"created":"2020-01-10 21:34:45.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/04/701104/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/701104/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/701104/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/701104/9"}}},"commit":{"parents":[{"commit":"1a6430f8785b7f7d1f9b6fc7beb6694727c00a73","subject":"Merge \"update constraint for neutron-lib to new release 1.31.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a6430f8785b7f7d1f9b6fc7beb6694727c00a73"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-08 06:15:33.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-01-10 21:34:16.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nheld back pyyaml and packaging and botocore and boto3 and coverage and pytest\n\nChange-Id: I19e72d3ddd74c0cbd64f420b4e42f1842078f545\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ac99875fa8e85db6984c9e178a7c8b8f0d83a00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ac99875fa8e85db6984c9e178a7c8b8f0d83a00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
