)]}'
{"id":"openstack%2Frequirements~701191","triplet_id":"openstack%2Frequirements~master~Ica66bb3a1e9c63f5bb09fd1d3af798a5f26cc1d7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ica66bb3a1e9c63f5bb09fd1d3af798a5f26cc1d7","subject":"update constraint for oslo.i18n to new release 3.25.1","status":"MERGED","created":"2020-01-06 10:32:19.000000000","updated":"2020-01-06 18:07:30.000000000","submitted":"2020-01-06 18:05:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701191-1578333902347-af565c7f","meta_rev_id":"88bb566814c95af5314ef00795dc3f0488fbc863","_number":701191,"virtual_id_number":701191,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-06 18:05:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-06 15:07:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-01-06 15:07:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-06 15:07:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-06 18:05:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d1651f60c2481186599045dce19175954649500","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-06 10:32:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a79f9c9ee0adb35c6489c9c2836c35480556cab1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 12:12:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2f7551fafd0456ba4de98f7800edf31 : SUCCESS in 4m 31s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/03376214f265456bb839a67858fddea2 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee358d3f70cd4a67b32dd6ced464e51a : SUCCESS in 6m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6d497721f7f443ebe4a33ce48d00c55 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9ae58f214bc46308fc111ebd3761a6a : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d27c220c533443089f612f31bcb3707 : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3f2764900f2142d19fc9cf3308bc0a89 : SUCCESS in 4m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f15fa0063b13442db6f02b2341686ed0 : SUCCESS in 4m 13s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/156162b375e14622807a74d12868aa32 : SUCCESS in 9m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/da4c8e534a89464b98fbafbe3418959b : SUCCESS in 8m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/863a6dc5758f4bb5af4225929adb0d98 : SUCCESS in 9m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8a77c4aaa26947f09ba234c42b4df25f : SUCCESS in 5m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d8e2300b308943278b22c2dd1f15eec5 : SUCCESS in 12m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ed92d6eb52a9404ba34f1b9c0c883de2 : SUCCESS in 9m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/015cb635e265401c9cf47ce1e2b48b9f : SUCCESS in 17m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3ec1f077a1894d49a7aced59b6d0a939 : SUCCESS in 25m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5a84c56797384ca5a00363901a9cf3c5 : SUCCESS in 18m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/61c042ebd0e840d4855c6e5fe3294f6b : SUCCESS in 14m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/16a4a11ba35e451cbc013b9206cbe314 : SUCCESS in 17m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5604975460b64e85b9f2608e7c17ab31 : SUCCESS in 6m 01s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/929f74e096cb47369e9ef879011f64bc : SUCCESS in 6m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4c4201a0190e42a191642b56616b3658 : SUCCESS in 6m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8dc4190e5ebe431eb97881f350f8dcce : SUCCESS in 10m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4ec3a570d755484aac731a8061a6baa2 : SUCCESS in 9m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70e9b0afc8ad4161b67b92eaa3a38ad7 : SUCCESS in 55m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8e70e0c13b3142878a213f183ec05544 : SUCCESS in 1h 15m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/376ad07c2c6b47e595d60d597263461c : SUCCESS in 1h 33m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/83d7e70e9ccf47bc806f616b68588c7a : SUCCESS in 1h 01m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a3ca852189e78a351827800227c01c44e713a9e1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-06 15:07:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"81c33d9aabe918596173fa24315e6ce531b54ad0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 15:07:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1a44143af28b6e3e45c96ad8bd79e6b9305ac8c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 18:05:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61987d70c96848af8cf6d810ab8e0c65 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5002fe2aaf8343faa1bca255a5c605d8 : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/44215e2b3fae498a851df8a46e6a9ea9 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fd1149e362c43228cdaafa9f2a9b48d : SUCCESS in 7m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e48822f75dd467a867797ba14b84d35 : SUCCESS in 6m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a487b0533cc6477b9f5e645ae4bc285f : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/03381d0024a448dc97bfe2193b6fb4c5 : SUCCESS in 7m 22s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a7146108aaa14c2e9af84ad25d4a70aa : SUCCESS in 8m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/98d0ec1da0d14fafa32134ea3aca391d : SUCCESS in 9m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c36b969a86c044ffa05578d70ca7fff0 : SUCCESS in 11m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/db7e49cef9c8415996a97672d416906d : SUCCESS in 8m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/28314cfe63e74c8ead241312d5f091ac : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f491d53df74b4b09917ac615cffd0d31 : SUCCESS in 9m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0f88ebd3363c4860bba872dd499ceb02 : SUCCESS in 18m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8d216de2e7a84bf8942482d9790d2dfc : SUCCESS in 25m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eef5fd17151e4f65a8a82a1f8e1bbd56 : SUCCESS in 16m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c2fed5da27b044aa9d2b1f5d84d1bd9d : SUCCESS in 17m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/32d756e8673b40c49217542197500afc : SUCCESS in 18m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4ce7d99e2d5b4a839f198d9c324585c6 : SUCCESS in 6m 21s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4dadbb3e57b044a9a9f5ec953dae0379 : SUCCESS in 7m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5cc988472c3f465bb86a07c064510655 : SUCCESS in 9m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/367ffc857c32471ab43652626ae565b5 : SUCCESS in 17m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/15dcc0f7b6714d2caee4c9c2a2fcb819 : SUCCESS in 12m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d9d37625b61482da8ced729c7524d95 : SUCCESS in 58m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b6bcd28709e42d39017def6110c3195 : SUCCESS in 1h 08m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/864087fcdcde429ca00c2f392c1ce57e : SUCCESS in 1h 30m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b52d4ef8e6ab4f399d750ad8ee156540 : SUCCESS in 55m 34s","accounts_in_message":[],"_revision_number":1},{"id":"517e9ccf711f742219a69d5a4d863c0e6e5b4795","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 18:05:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6b2b2ee4f3cf75e6a6640fcc38b4b9adaa2ad989","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 18:07:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1c197a237644709a8e6f7f235dde8aa : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9df6880cbc5fbbe3969ca29a46a9aaca3d6e1aec","revisions":{"9df6880cbc5fbbe3969ca29a46a9aaca3d6e1aec":{"kind":"REWORK","_number":1,"created":"2020-01-06 10:32:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/701191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/701191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/701191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/701191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/701191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/701191/1"}}},"commit":{"parents":[{"commit":"1bd226f4bbf79454e439ff91fbd032feda55d414","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1bd226f4bbf79454e439ff91fbd032feda55d414"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-06 10:32:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-06 10:32:18.000000000","tz":0},"subject":"update constraint for oslo.i18n to new release 3.25.1","message":"update constraint for oslo.i18n to new release 3.25.1\n\nChange-Id: Ica66bb3a1e9c63f5bb09fd1d3af798a5f26cc1d7\nmeta:version: 3.25.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Ia018071ac72f652b29929c1f23fd09e3e61ae3da\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9df6880cbc5fbbe3969ca29a46a9aaca3d6e1aec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9df6880cbc5fbbe3969ca29a46a9aaca3d6e1aec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
