)]}'
{"id":"openstack%2Frequirements~701192","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Id2c11d36215c3d6cc93526a80ce1bda265ccc2f7","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"Id2c11d36215c3d6cc93526a80ce1bda265ccc2f7","subject":"update constraint for os-brick to new release 2.10.1","status":"MERGED","created":"2020-01-06 10:32:28.000000000","updated":"2020-01-07 02:33:17.000000000","submitted":"2020-01-07 02:33:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701192-1578364397927-17819b5d","meta_rev_id":"bc5cc567cd0d3d77ac4e672cb8ac2099401bcc96","_number":701192,"virtual_id_number":701192,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-07 02:33:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-06 16:56:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-01-06 19:56:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-01-06 19:56:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-06 16:56:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-06 19:56:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-07 02:33:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b195a5200c025ed7163ac73847333095daf6cc5e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-06 10:32:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d248b4bda1ba4ac04d600b996184b71ee291863","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 12:17:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ba20424286dc49598942cb000399e557 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7768707c58664d9caccddceb3644f535 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/071874f3235e40fcb093fa4c381558e6 : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7d1f99964ce4522ae25353a8232b92d : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32c9daa931794582b32bf9f5c7f18614 : SUCCESS in 6m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a6b9c28497064a9ca7bda67b0fb05773 : SUCCESS in 5m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f54d69198974e46bf4c7b3a05026615 : SUCCESS in 5m 14s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6a3e71a75a5f42ec8ac982c20f117587 : SUCCESS in 8m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b89ea9c6dae14bf28cb7eb2d22fecff2 : SUCCESS in 9m 12s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b545d5558872462c8bffb12a7d4e8f27 : FAILURE in 12m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/47319c3523ea42f1a6d154da8792e373 : SUCCESS in 9m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/11c0effc9f59413bb4f43a50349925cb : SUCCESS in 6m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/39b1c4c7277b49baa2d893b56f68b138 : SUCCESS in 6m 37s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e29a1085fe51420d9da344ee8eb530d6 : SUCCESS in 10m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7569c95e05974a82acf0678d5ab57e70 : SUCCESS in 13m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4762cbd6b1c74464a25c6a415d806f5a : SUCCESS in 8m 24s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ec41429705574fd99357347f20a6ba5d : SUCCESS in 13m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4d6980e39d894011a554eb202d9b29bd : SUCCESS in 16m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/346a0a50f6c648f59eff1baf4e26d24b : SUCCESS in 24m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f9222347b1fb4ba68735c302bf707426 : SUCCESS in 26m 52s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/89f7c8e0f4d149d0a9823907533c4e16 : SUCCESS in 13m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/20554d1c84004daebf5d3476c6876cc8 : SUCCESS in 13m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/623118529bbe4c9483a76afe9d58a145 : SUCCESS in 17m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/78ab803068a74896ad9ae0e1d06ecd59 : SUCCESS in 22m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7478040831184232ab927e7e86eaa2d0 : SUCCESS in 5m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a51b5c3a14b74d6188f5ca1cca35fd71 : SUCCESS in 9m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fb8c222faab342618ba17180c05a6bd4 : SUCCESS in 10m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f1eba96f503b47ad9480e88365e33cab : SUCCESS in 1h 09m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/10880b974fae486ea16cbaae474030db : SUCCESS in 1h 38m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3f142bb2f2b247dd88cb6caae3ccdab5 : SUCCESS in 1h 09m 07s","accounts_in_message":[],"_revision_number":1},{"id":"0fb0b93ad0487282f8eb8e93b1858d4d66d23406","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-06 15:05:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"49bad0dcc0c4fb9893ce896ab3e7a9e9baaa4510","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 16:52:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ade3913e175a4ea9b37e465deb6068b4 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/818bd6f1852d4ca8ba886981f6092efb : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/861504ec28aa4a2fbac13fe56e4c8dbd : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97cd6623a67140fea1457d8bf4168577 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1286242c06164165974487166ba8db6d : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ab9560ca20641d9a509f609d888ee16 : SUCCESS in 5m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3ea0b8be5801443ea53432832a6003a6 : SUCCESS in 5m 07s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6a198a365cf7453d88908396cc163152 : SUCCESS in 10m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ac3725e9bc494a0dba3238cd4ad9589e : SUCCESS in 10m 14s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/34e5d450d121471593acd0aa1097d998 : SUCCESS in 11m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/12d21ebda7ab4fd1b40f35bc17af9be7 : SUCCESS in 12m 29s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/54fdc32646e245af8f5cb9034df4b34c : SUCCESS in 6m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/56f8e78444024a29b08b5290e058b4c6 : SUCCESS in 9m 14s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1dbca72257c645508ce75961a2dd17a8 : SUCCESS in 10m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8c98134459b24a399855f759ffd36393 : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/869760d7a5be4b289f6d4f1af05b8465 : SUCCESS in 8m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0977629792f34283a8b169170f5ce994 : SUCCESS in 22m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5d7cf438ce2c49c3894cbffb2e4dccc2 : SUCCESS in 22m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0ef163f5c4b4455d98344e1221ac9ef3 : SUCCESS in 32m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/097d32b124c64fe3bc60040d1f667d23 : SUCCESS in 22m 09s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c548fb82384a41b2b3df4e165f9c701f : SUCCESS in 14m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bbb1dc02e7944effb8fbf3826f5c368c : SUCCESS in 19m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/40b46e5ce5724aa083d6192d85f6c9a7 : SUCCESS in 15m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/456f66ebddbe449b92d89a3b2a014071 : SUCCESS in 26m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1998e61fb4624074b5fed1c4f7066fd6 : SUCCESS in 6m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0e1f1dd3c6474d7a82d00d7d7c20b3b4 : SUCCESS in 13m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/51c3234755224a14a2d8d881541452a2 : SUCCESS in 9m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d3929b62f7d4325a7826035c8262e2a : SUCCESS in 1h 10m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/83fc1ea4ead2411e8e92257f525a3438 : FAILURE in 1h 25m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e0a845d535d94f0ea34c1fa7d46e7495 : SUCCESS in 1h 03m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3a9dc1d091db4b6c78f55880afad08aaa0b1beac","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-06 16:56:15.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3a6dd801f46c0a7befcd8fc8640887bd836d27a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 18:29:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/abaf40f16e77412bb15371045c8502f4 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f45c81db3734b16847d588ba6227521 : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5b69a2943639400fbb4652806b631123 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bfd429a006b4640aefe05e8ee52c624 : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c3d9df300efe48019eaceea17ce0e496 : SUCCESS in 4m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ec4564ae52c40eda1b3f5473e40c16c : SUCCESS in 5m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eb19d369e7f74c05af47fbad46be273f : SUCCESS in 7m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/81a882f3440f4e19a90f9d93f662eb9f : SUCCESS in 11m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/28a010a726974ee9997edf7b5d7b9455 : SUCCESS in 8m 57s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/bbf8012bcf864f74807f10a484349d9c : SUCCESS in 12m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e21b741149934404ae4b861d6b80015b : SUCCESS in 9m 51s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5c878c6bb33e4d6ea5f396a27584b3cb : SUCCESS in 6m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9196bd4632624414a772d359b80c422b : SUCCESS in 6m 06s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0fd0a2fd3f5643f398d06236b65fd7b4 : SUCCESS in 10m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fdc1469eedc240bba8be5abd3c0dbaf8 : SUCCESS in 10m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dddee779e34545e9a176111aee260474 : SUCCESS in 10m 20s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e1a0c18f603e4671a7f90b42e86a3a40 : SUCCESS in 17m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3a5f370199354601ae91d497930f73f2 : SUCCESS in 14m 41s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/baa4ed30f181469097be3356e837087d : SUCCESS in 23m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/be064feb03844349a1ed8672f3852154 : SUCCESS in 28m 05s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b6ecb2226a994464a0f07568eb4af22f : SUCCESS in 14m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9c656a5a7d274bccbae61f8459ab8487 : SUCCESS in 13m 52s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ba6bbb67000c4c10813e079c2f7e80d3 : SUCCESS in 14m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/64fdb9951cf24fcab845f3ddd7c7bca1 : SUCCESS in 19m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/870d74d2aeae47728046f6d3f7e7a07a : SUCCESS in 7m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2088718cec2a4397b4ca8b596bb36b82 : SUCCESS in 10m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/924e7643b69b4672b411276a3a2a19c5 : SUCCESS in 10m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca290f8b90424ae8bf9143053f1294d7 : SUCCESS in 1h 13m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7509ba7a5d9642fd97b17efe314ee507 : SUCCESS in 1h 31m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dd4aebc4325b45d09ba01a3cbc3cff29 : SUCCESS in 1h 04m 44s","accounts_in_message":[],"_revision_number":1},{"id":"fe8fcef9d7cf836e5f8adc49afe87d680394b3e1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-06 19:56:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9cc0bee849f7bcf61d3a48163c7dd0e18c1563e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 19:57:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aabff5130572317ef8d5c8f53870252a97ec7b53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 23:13:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a00768b183544469a1fbc8fb528c04ea : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/72cc4cdab7a94f518a15ae7ebb6bd8ea : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8c8313f732641c5a6f2ce4277e4536e : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2533ed6ecf024204802c93444b082e1d : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/33a62d5eb4f142af85a324cd1570f81e : SUCCESS in 5m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1b4938d35be94a40afd2b1d6997567a1 : SUCCESS in 3m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1f8b12c0f1da4c19be5648f7a887148e : SUCCESS in 9m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3758fbf42e914aee9e0d978ab478037c : SUCCESS in 9m 06s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/592daf643acf45a5a1c434770836517b : SUCCESS in 9m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/03e4019784994004969c404da966419a : FAILURE in 8m 22s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f039396395b449c680b53d2aec17a9bc : SUCCESS in 7m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/01f564ec4a944eec87a8e409c22cc550 : SUCCESS in 5m 36s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e34583bbc07a470ebfe6ca5c4dfde25b : SUCCESS in 10m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dd43150c1c6d4250ae9e17bf83a375e2 : SUCCESS in 8m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a414c32561544e9fb0d9fc661d2b5196 : SUCCESS in 8m 57s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1fb4fa4f082841ee959c361051cc108a : SUCCESS in 15m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6a1c7cc13f2549099175bef20fd1043b : SUCCESS in 19m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/72f2afc82632468bb9e1c0e98375d52b : SUCCESS in 18m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7bdc4b1167294385bf90f4cd51679763 : SUCCESS in 28m 11s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/95f9daeb64a04cafbd95f8215f2aee4e : SUCCESS in 11m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d79cbf7b966a42368b0dc366e0ef660b : SUCCESS in 13m 59s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d2c044a7fe45414c9e6213029e16fdf6 : SUCCESS in 15m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d221e45de5854ec69e538d5d022c80af : SUCCESS in 25m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3e82ae9554ae485b80f0b6e9a3fdff66 : SUCCESS in 6m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c8e3455ca37a4968aa9f5eef7b4426ae : SUCCESS in 9m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a4aacba90b21433aaf0ff32f4af152dd : SUCCESS in 11m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca9891e6d388467692b6faf3005b4060 : SUCCESS in 58m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9e0ddded247843c6bc60de0f66508c7e : SUCCESS in 1h 23m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6d86aaf288aa4ca1acf4558275bf21a5 : SUCCESS in 59m 32s","accounts_in_message":[],"_revision_number":1},{"id":"26206d74d672b9c064eb4a9407c3728bb17a115f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-06 23:15:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"733186fc6291e45f63ec37419c158a032f40c960","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 00:37:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/64e27e5f2a884725bffb44006676376d : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b217ec7d5744065b1061305cae55c31 : SUCCESS in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/035bf4d15df3463aaec2a7767c17e33f : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5d93cb3b0444343802da7fdc5d058e1 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6d8b89ebbd24c60be6eb8269d3b5dcf : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2e00f90ae984ec6b8922469927941f2 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6bf97d896d5044a1a42a5e560fd69da3 : SUCCESS in 4m 12s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/27280e55372c402497af0ff154af9316 : SUCCESS in 7m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dda1ffeb553a42ea90fd727a4fa1f578 : SUCCESS in 7m 14s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/309aa3523a2647b5b803b29646a25a9c : SUCCESS in 10m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/abdb03663db44b7ba304833f887a4ecd : SUCCESS in 9m 35s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b1e3f8db8cec42a79f7ac21ecfc42000 : SUCCESS in 5m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/17b89244954f43349cb4e0cae6ffb40e : SUCCESS in 5m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/574fc7833d9f4a9781b2aabb53a2c757 : SUCCESS in 9m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d68d5c6bf6c246cebb3b2bfa6564a7b0 : SUCCESS in 7m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ad27b24741354271ac8ad149cad82932 : SUCCESS in 7m 01s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f480c00a0ab540ee94d6af9ed6a24d00 : SUCCESS in 44m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e88e98e7e0f5448f90cd4411dc7cc9c4 : SUCCESS in 13m 37s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6c9091d839104ba39e5c9e5a18d6863d : SUCCESS in 20m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cfe9d40842ed480e81ecf34c16293d58 : SUCCESS in 31m 16s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/14a18c15f20b479a876f732edf4bcf29 : SUCCESS in 16m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cde5ff3a08a3404281aedb9589049572 : SUCCESS in 18m 42s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7948f43d5e25452ab23f7e4528f9f0d1 : SUCCESS in 16m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bde57389158243bf9f15c366298433e4 : SUCCESS in 29m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fbcd74cb9de942f39c93323d6054fed9 : SUCCESS in 5m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bf68b30952694c5eb08fcf6865b3bd81 : SUCCESS in 11m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/328d84195b734cbc80a7c4bbe92f7c8d : SUCCESS in 11m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f6c6026b3040417e88b98a4904983e6a : SUCCESS in 1h 11m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9d427d4aa37b45208d91a53ab8489dcc : SUCCESS in 1h 20m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/40b8d13f1eec4072bdc46027116a8cdb : SUCCESS in 52m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a756b85662b716bbcb950b54638f2d36b1aaba7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 00:37:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5deea5a97dde862f995f50387687bc5289af10aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 02:33:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"25d18732685d30b792f832123183737a9f304443","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 02:33:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c59db2eeb34d412686f07eec9414ff77 : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c739d80da38944e1b8d67a4e84824397 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2dbe5abc5824e5f918fb1fe50b788ca : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de5cd1ec121f47bab1fffdf98d776539 : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3fac3ffd452648719f532754c39ef689 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/140700a0109541cb88db29ad736cb8ae : SUCCESS in 3m 30s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/bcb00e9955be44b8b887c221edf9c83d : SUCCESS in 8m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2dab84d7105048e08913bff5dce940f7 : SUCCESS in 7m 02s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7fe67e0fd38c4abda058123be48568e6 : SUCCESS in 13m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/89a02838a86f496798807271b55c2d56 : SUCCESS in 8m 38s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9833a7c097ef457583d5468daa39fa16 : SUCCESS in 4m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ec129848c03f417dbeb23f76f8dfa65e : SUCCESS in 4m 46s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b8c2d2643b5d4c4881b0a2b3e16d222f : SUCCESS in 13m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8065263b73214405b36e164a83faaad7 : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0ef615168f4c4eaa8d748c1ba9af459b : SUCCESS in 7m 13s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/cbb80c3e7c8b4907b59fbe4f5cf3a4df : SUCCESS in 15m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c3e53217b13f46de9ba8fcb4d91356af : SUCCESS in 17m 02s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0bf3fe788f6a4edfbed0b6987a2efd23 : SUCCESS in 27m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/193fa6b6c84049aebc6e5742d3a7fbf3 : SUCCESS in 34m 35s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/fe8e0f99db73446aa19dc9070b65d82d : SUCCESS in 21m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6e46ba2adae643338484098abd259c01 : SUCCESS in 11m 52s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c801efea990f4a148f959085dc3b7e38 : SUCCESS in 14m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4dd34689f97f48f7b061cdfd75d2a605 : SUCCESS in 20m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6fd47289f355444995e74a8fd1272f7a : SUCCESS in 5m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f385c894ec844f20a4721faf8df5dbd7 : SUCCESS in 10m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cc1f1f592353409aac598c8b1a14f7ef : SUCCESS in 10m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f615ccc09fdd4016afbc66344bcfc724 : SUCCESS in 1h 19m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/361fa8ced7cf47de97f06fe3fd4edebe : SUCCESS in 1h 52m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/aab6f8aaa21545d4b4eee89b0e2fb4a9 : SUCCESS in 1h 01m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f460537c0f476eba13e121eb974038190f5169b0","revisions":{"f460537c0f476eba13e121eb974038190f5169b0":{"kind":"REWORK","_number":1,"created":"2020-01-06 10:32:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/701192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/701192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/701192/1"}}},"commit":{"parents":[{"commit":"3e54878f9c87266e1b9775f72af06de27a3dad65","subject":"Add horizon to upper-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e54878f9c87266e1b9775f72af06de27a3dad65"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-06 10:32:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-06 10:32:26.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.10.1","message":"update constraint for os-brick to new release 2.10.1\n\nChange-Id: Id2c11d36215c3d6cc93526a80ce1bda265ccc2f7\nmeta:version: 2.10.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: If83444293af19ecac1aea3b22672728a325a58dd\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f460537c0f476eba13e121eb974038190f5169b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f460537c0f476eba13e121eb974038190f5169b0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
