)]}'
{"id":"openstack%2Frequirements~701199","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I98c14f9ca629e8cba4b45bece3a17a26439894b7","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I98c14f9ca629e8cba4b45bece3a17a26439894b7","subject":"update constraint for os-collect-config to new release 10.5.1","status":"MERGED","created":"2020-01-06 10:42:59.000000000","updated":"2020-01-06 23:51:05.000000000","submitted":"2020-01-06 23:51:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701199-1578354665758-d3fc192b","meta_rev_id":"72c9497e7dde530fa5962d99455372d425598eb8","_number":701199,"virtual_id_number":701199,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-06 23:51:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-06 19:00:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-01-06 20:15:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-01-06 20:15:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-06 19:00:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-06 20:15:47.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-06 23:51:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c9184b3169536bd8a15e944db4dbba557b5911c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-06 10:42:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a170ae96a423683cec09d5184cdb118079690b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 12:35:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ec4aa7e8c1dd4b2c875cbeba95269c2f : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a4e2c2d2ea046c3ac9119d1d48a8211 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3cf686fbc04d46c5ae42c28e9082bb83 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/860fd1efc78140ac8f046900b85d5444 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/458221839ce5483f96691ebe362e2b5b : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/517888ebe75d477e8fe2339536c4f39b : SUCCESS in 4m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b2bb6fe293894a839ad1bf757a3e63d2 : SUCCESS in 5m 28s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/cfb2c9bacb224d519d50b6841aba201b : SUCCESS in 9m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a37bf4918abc4284a625b19de9614f09 : SUCCESS in 7m 12s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6008842fe3114efb8bdbae36432f4354 : SUCCESS in 11m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e524fe2779e04b7c8d0d7aa1a78e9041 : SUCCESS in 11m 00s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/781c5a671e6448c9b024712672d4d955 : SUCCESS in 5m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f214063aa00a4eb09e0731faac1a2ba7 : SUCCESS in 5m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5e7e596a092949e7b7bc8dc719be34b0 : SUCCESS in 12m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6b36bc092bc342d78177be2b0cfa360e : SUCCESS in 10m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09227d61191440dbb1140817de5d3145 : SUCCESS in 6m 18s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f96c31fb13aa4c1fa920b79d17052aaa : SUCCESS in 14m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c9321a71f3284e268e4af6b5f7b7fded : SUCCESS in 17m 25s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c3ddabf20a5a4ceda70a73d46fa08ad3 : SUCCESS in 22m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/85a4b79e544f41188abc1bf97562d076 : SUCCESS in 32m 16s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/432295201c6e45acbb766ba3a7f797ca : SUCCESS in 15m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6da22a57423845949886ba556a4371c3 : SUCCESS in 12m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5a778483ada74a02b750bf274074f202 : SUCCESS in 18m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c06d5e0030cf44ccaa42539d9bd8fcb7 : SUCCESS in 22m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5ded534c089c4ae9b355be488f217a35 : SUCCESS in 7m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2f22e2af7dda48598ef22df820da3c1c : FAILURE in 11m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1706db7eece34f11a1d381e56286a721 : SUCCESS in 10m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/29241d214ba34041ad1439c9dd644441 : SUCCESS in 1h 07m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/44fbcdd9c25e45a281319a860cb422a9 : SUCCESS in 1h 45m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/515c4193109744d59ae9603ecf8659a1 : SUCCESS in 1h 02m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6f2ad17de020dddb9a85df88431928a0c447a2ca","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-06 15:05:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f6a8b71c8e570c3ff1c7324e4e05363a1413b1ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 17:01:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c0f81ada1dd2476abd8e1d8181bc7ec5 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbf273c468d5464ca02c6193d82c36ac : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ed27914050dc46cabb64372c7df0942e : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e71ee80d06a34953bbe93d175a887fc6 : SUCCESS in 7m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa480ede97a54cab87d4ba52ae782f00 : SUCCESS in 6m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be4098b99d494482a8c9c07503befe25 : SUCCESS in 5m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1de660c97dde4f7e852a72677156b2bc : SUCCESS in 6m 02s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/21ab4f8b785043eeb45889ffb8ee875b : SUCCESS in 9m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3be988980e9a4a11b8f8bd6a005823e9 : SUCCESS in 11m 30s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4391763a6a1440a3817ed0bc194f4248 : SUCCESS in 11m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1c73761ae44246d3bbe0598224f287e9 : SUCCESS in 13m 20s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4db4cf047b2b43ff9f7706c3f6ca0af8 : SUCCESS in 6m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/81fb6df8d7b744be914975604c6bafa1 : SUCCESS in 6m 18s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b762dc13a6f84dc790999b312627118d : SUCCESS in 12m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/23a9d78a95784bd9add8c05ff7094cb0 : SUCCESS in 11m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/811d091dcf344731b8ffcd3f6a25abee : SUCCESS in 9m 12s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2afcdfa85d5145ef8ffb9b5d5122aad6 : SUCCESS in 15m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/02c697cdf0024e7cb22d339b914fc6f5 : SUCCESS in 13m 45s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8efa62f9bf5a4f009dc2de17519d16b4 : SUCCESS in 27m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a734e59674584ab1b9af8327c2d0cf48 : SUCCESS in 21m 28s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/452ba068c734458ea12ade6b88ed0a6a : SUCCESS in 13m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b1075022f9274c318edd4401f4a887c6 : SUCCESS in 16m 42s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5c90198d9788469e943153755cbbe30c : SUCCESS in 15m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3b3d70a6a3bc418e9a25f48a4753e8b1 : SUCCESS in 20m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4f519f9ff97b44ea83d2d97d838e8719 : SUCCESS in 8m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c4de5e40d0d24bdbaaca85d212d948fa : SUCCESS in 12m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cce46b830b3847d7b6fce0ba7a77cad2 : SUCCESS in 13m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bde4d847fe2544d08d1ed1177561ebda : SUCCESS in 1h 07m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b91f659717b64b26b178277277e515ff : SUCCESS in 1h 50m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e4568041c4f14a588e4a23cf505ab72b : SUCCESS in 1h 08m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7dacc2582109071964882528ca52cebf7199193f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-06 19:00:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7dfb427270e1a33157fbd356645fcae1568c68a0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-06 20:15:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"162e998558b673f4b985231779886e1c7295d1c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 20:16:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f80f3800b6826a1e53945068daff21b91bffd5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 23:51:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4322f8448ce95ab40de92a69d09ab0feec8945f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 23:51:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43f1ab847c4f4391bf752c61120dfe40 : SUCCESS in 3m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c1629a3c35e4a7fb0f558fe6cbeeadd : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/728a0083f7df4c8db605894b43cbed92 : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4225d3d84d8649538bf184f7e9a82f23 : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/86046555ea7b4b3f97f7c4d6c43aa7b9 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/da6989e9d6f448b1a054340f73b27158 : SUCCESS in 4m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e6904f1c47834a34b793f00421b41f93 : SUCCESS in 7m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/026bd88985df4cd99114e6d55d157ae8 : SUCCESS in 7m 25s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8cb626275b434764829385af3bb89110 : SUCCESS in 10m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9702acab52414e0f9b569c16df7c066f : SUCCESS in 10m 32s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f00f332fcb92406fbb9a2fd4bfc5758e : SUCCESS in 5m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d3c0d83992ce402baf2ef01af2e498bd : SUCCESS in 5m 35s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9f69d3bb4fc84a4ea8daa60fe6d4eb49 : SUCCESS in 11m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1ca00912996a4c6ca31cc7b593d88b09 : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fd4149cee53b498f835c418c365c974f : SUCCESS in 8m 39s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e200894bd2dc4aadaec7f5eefa5c20ea : SUCCESS in 14m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8a0136a24dbd431387d19e438b9b0baf : SUCCESS in 16m 30s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/525478ad36164d3e8281338ccab4bc87 : SUCCESS in 27m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ef8d02a0d89e4127bfb25e8b8729994e : SUCCESS in 28m 08s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8470c080b97343bdbf718e7d1ceca68b : SUCCESS in 13m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e53a69b0fa0f4d6db6b401f944a1cb9e : SUCCESS in 13m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3628c8f7636a4173886a3a1dad9543f8 : SUCCESS in 14m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/74f9d6e408244ff580e935d0e640f8a8 : SUCCESS in 19m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/133f33ee27ab43aa94f4235622522895 : SUCCESS in 6m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6f991a1268b34ede8cce42b37311f9ac : SUCCESS in 9m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f73d5f879bcd4522be58bd4c979163c8 : SUCCESS in 10m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4153785db3ed4e17a17fcfbb0e225c69 : SUCCESS in 1h 13m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ddc26137e3eb435193315ff34c8c900a : SUCCESS in 1h 31m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8915a90cefe94806b5f8cfb60d2dcb20 : SUCCESS in 56m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1c2fe48d3d2f8663162b83669cc8219093881bd","revisions":{"c1c2fe48d3d2f8663162b83669cc8219093881bd":{"kind":"REWORK","_number":1,"created":"2020-01-06 10:42:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/701199/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/701199/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/701199/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/701199/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/701199/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/701199/1"}}},"commit":{"parents":[{"commit":"3e54878f9c87266e1b9775f72af06de27a3dad65","subject":"Add horizon to upper-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e54878f9c87266e1b9775f72af06de27a3dad65"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-06 10:42:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-06 10:42:57.000000000","tz":0},"subject":"update constraint for os-collect-config to new release 10.5.1","message":"update constraint for os-collect-config to new release 10.5.1\n\nChange-Id: I98c14f9ca629e8cba4b45bece3a17a26439894b7\nmeta:version: 10.5.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Commit: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Change-Id: I353fd64275c496380ccd850991015a5c885bf94d\nmeta:release:Code-Review+1: Kevin Carter (cloudnull) \u003ckevin@cloudnull.com\u003e\nmeta:release:Code-Review+1: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Code-Review+1: Ronelle Landy \u003crlandy@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c1c2fe48d3d2f8663162b83669cc8219093881bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c1c2fe48d3d2f8663162b83669cc8219093881bd"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
