)]}'
{"id":"openstack%2Frequirements~701201","triplet_id":"openstack%2Frequirements~master~I288e71b70eb111d948da1ecfc046188f8e11df25","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I288e71b70eb111d948da1ecfc046188f8e11df25","subject":"update constraint for oslo.messaging to new release 10.5.0","status":"MERGED","created":"2020-01-06 10:43:23.000000000","updated":"2020-01-06 19:34:34.000000000","submitted":"2020-01-06 19:32:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701201-1578339125974-716aec10","meta_rev_id":"2f10480ca72c1ef86a973675df99bd2c063dbc28","_number":701201,"virtual_id_number":701201,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-06 19:32:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-06 15:07:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-01-06 15:07:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-06 15:07:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-06 19:32:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"89cabb2b80ffdf2bbcdf84ee2a04fa231c4ab2aa","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-06 10:43:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b42ba021cce2b2c0ddfe788044e23af169833ea8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 12:12:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdf2e106ceab4df9b0db87eef63e4c38 : SUCCESS in 5m 38s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6733c8a2705b4043b9e7cf3e0ed99674 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcbf68343b364295a6e7c09e5c21b3a1 : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/765b3f01af85487eb9dcbaddb8ac293d : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5d25aa6d8ba40c4a51e1a61fc497570 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b2af11281564ff89dcb9304dd4471a6 : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a661645677840beb6b9d5371bfae1c2 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dfa6388b5010431c99ad6b2e4387f282 : SUCCESS in 4m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/79210102573c4f508238614c5e339312 : SUCCESS in 7m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2fc10da7307c408c87c1014c020d310c : SUCCESS in 7m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/796e602d2d6c42b391195cfaa0077426 : SUCCESS in 8m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/aa86c3db2cd746569b19c31d7c51e642 : SUCCESS in 4m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1011070940b2480ab81babb5e72d06a1 : SUCCESS in 13m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/676013cf07da42549a3350a8b6e1f6b0 : SUCCESS in 7m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/42a601e84cb44979a8593227bddc60c0 : SUCCESS in 11m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c25dc6ba1746410eb9135873bab840ca : SUCCESS in 25m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b1df99da7242447dba582844db0e35aa : SUCCESS in 17m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/91ff4cb77ce444078a7b4953efd85a0c : SUCCESS in 14m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a2bdfdeda63b439da45de11fe8e6f914 : SUCCESS in 20m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7c57e5b9469b4dc5bdd090e770346a23 : SUCCESS in 4m 17s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0a02d749bc8b40cd872326c2bcbc00b7 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/890a3f598fe649ddac922c604a1764d8 : SUCCESS in 5m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/eb1d8bb87fcb449aa26d2911154f0a31 : SUCCESS in 8m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/415210d1de20444593c0e88ef3213880 : SUCCESS in 9m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46f0802f1f0f485f9c0a3c99fe4192e5 : SUCCESS in 1h 01m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fbeac94bf7c543bd964ae4bd4ad18cfc : SUCCESS in 1h 12m 29s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c3729673e5bf4599b141a3aaa82ad254 : SUCCESS in 1h 20m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c3cc51abf08c45938338dc4cf6a53d42 : SUCCESS in 1h 06m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5334c3293e29ffd0479e4a5ba887014ca736f22a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-06 15:07:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4299fc3a84d08af9e2ee2be2e85ec8bd81aa078a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 15:07:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4760d29ce921ba76896853cf22ac1741dda09864","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 19:32:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf5b35d376354a448accdd596849d6c1 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75e68534ac364929b79ca26a68ce9fd9 : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19403f73f8c84753b53b5043682b84d2 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4abfda08ba8e4f0092cb0f7c4d1dccac : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0643bcd3a08d4386a0c37d88ceed620f : SUCCESS in 5m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8ba4dbf5adeb41de98379ed05ef12df6 : SUCCESS in 5m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4242cfc281ca4fa7b26dd3074e08b38d : SUCCESS in 4m 28s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/135803baaf6040d2acec3b4db98d39eb : SUCCESS in 9m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/faddb20a4f3b428e8f2595c37d8d0e9b : SUCCESS in 8m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/377e3ba06e534c8385e61dafbc286e58 : SUCCESS in 11m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bb3740810be644c8ae14a6861bfbd228 : SUCCESS in 5m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d7226712b8134025988087d24c85118c : SUCCESS in 11m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9b7e3c6ba8154ea6878a13806e525c62 : SUCCESS in 9m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9cb341b1c8f041538483b5024f4cf409 : SUCCESS in 13m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b217c8f32af845c4b430fff5973bda44 : SUCCESS in 26m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9fd7a268c29f4f75a4ead7589a0ea2f0 : SUCCESS in 16m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/458622dcc93d45edb6968e9143daef3c : SUCCESS in 16m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8c058578f96b4b4585b2bbd3adf467d3 : SUCCESS in 22m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/18ff08d6cb054f45a62d8b77639a4c59 : SUCCESS in 7m 15s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ce77bf2f50074b218591aa0aba59ee5d : SUCCESS in 6m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e0c109e84abd4182922de89d9c4bbe43 : SUCCESS in 6m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/258a317ab8584d94af81f843d93f13c7 : SUCCESS in 11m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/75cad55b63e241b2a6354cf2ab42dd82 : SUCCESS in 13m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ac90924999a44439bc1609953f876a4 : SUCCESS in 1h 01m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bcfaa449df454c90b3b06940780bf2c5 : SUCCESS in 1h 12m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a5bba77765cd413fb82bfd75a38cac1c : SUCCESS in 1h 34m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/830e50c0cd81467eaa4b9c28d95559a4 : SUCCESS in 57m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8d7b637efdc10127b0f5626a4631167f437740d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 19:32:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"91561a112f5a2a102457ed1fc1f00c14c8922dbe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 19:34:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fd29e3f484d41bca839a0a3958a259a : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d25801a8ab90d171c311b8cf234ca024104d140e","revisions":{"d25801a8ab90d171c311b8cf234ca024104d140e":{"kind":"REWORK","_number":1,"created":"2020-01-06 10:43:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/701201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/701201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/701201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/701201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/701201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/701201/1"}}},"commit":{"parents":[{"commit":"1bd226f4bbf79454e439ff91fbd032feda55d414","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1bd226f4bbf79454e439ff91fbd032feda55d414"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-06 10:43:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-06 10:43:21.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 10.5.0","message":"update constraint for oslo.messaging to new release 10.5.0\n\nChange-Id: I288e71b70eb111d948da1ecfc046188f8e11df25\nmeta:version: 10.5.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Ia018071ac72f652b29929c1f23fd09e3e61ae3da\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d25801a8ab90d171c311b8cf234ca024104d140e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d25801a8ab90d171c311b8cf234ca024104d140e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
