)]}'
{"id":"openstack%2Frequirements~701202","triplet_id":"openstack%2Frequirements~stable%2Frocky~I9deaef1f9d4b79f2cfa82d675415b18b7793a4f5","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"I9deaef1f9d4b79f2cfa82d675415b18b7793a4f5","subject":"update constraint for os-brick to new release 2.5.9","status":"MERGED","created":"2020-01-06 10:47:21.000000000","updated":"2020-01-06 23:51:01.000000000","submitted":"2020-01-06 23:51:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701202-1578354661477-af565c7f","meta_rev_id":"2ed2e0b01167e58274754f346e587b55ac4d6564","_number":701202,"virtual_id_number":701202,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-06 23:51:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-06 15:05:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-01-06 20:15:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-01-06 20:15:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-06 15:05:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-06 20:15:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-06 23:51:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8d4adcec27e6f84ffffed563851b93fb2adb3a4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-06 10:47:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46361de4be102af749cdc6f59ea95bb03fa63a29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 12:28:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5ba7cf06857448579262a08451a08eb5 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f07457229964605bc4343292090273d : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7fb9cbf234f7433397fcb7550705a187 : SUCCESS in 4m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8ae6e98f6a4744fb86de59add95a8849 : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/461ab78872394a7ca562dae8c550ba11 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e38037c69bf6442eb56dc5e40aef3d0a : SUCCESS in 3m 55s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a641b371c665469985005211a3c5ecc9 : SUCCESS in 8m 05s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/502b3d5192ca4d3c82929e88c6027178 : SUCCESS in 14m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e721f042aa344e32aaa57a911344116a : FAILURE in 20m 02s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3cf50dd996c043208379c9a41cac3a31 : SUCCESS in 9m 20s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/8eda83d6f1944b4d93f7e4cc06555fe7 : SUCCESS in 9m 38s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a69f37ad0fcc4d26a78d58f291e2c47f : SUCCESS in 4m 54s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/7dfc4b5a84a5499483f898317c9bb513 : SUCCESS in 4m 41s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e77335800546446ca77d9ecc8e5853d7 : SUCCESS in 10m 34s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/ea82bc91c7cf43e5b4f56986ffa2151f : SUCCESS in 9m 47s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c2e25b4eb8e1406a8defb0831d0bf698 : SUCCESS in 12m 30s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/570232a207ca487a8a3e7d1d285506eb : SUCCESS in 12m 28s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0719fc177b274d52906af2d48d35430b : SUCCESS in 23m 13s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/cc91d2dad2a246b796b9430b5cc242b3 : SUCCESS in 25m 46s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/54e1f5d67f654176a57f85d345b6c113 : SUCCESS in 14m 07s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/ebfe65fcdad7472fbc30d624a5ae0328 : SUCCESS in 13m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2561af34286d45e382053d794750975b : SUCCESS in 17m 29s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/91253a10d49a45fabcf0fdde15a2370b : SUCCESS in 4m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/227c0f782dbe4547b902f77b4f80fe98 : SUCCESS in 8m 31s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/372b97ebccdb4e6090cc382c51bbf32e : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4fce93564b849178a2a049493d53852 : SUCCESS in 4m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/864928734ca04e7ba1f22ef4c14b0a54 : SUCCESS in 1h 05m 26s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/d21599f358274515b1807d1e0efdfcd2 : SUCCESS in 1h 35m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bca9cbae041f4937a5174ab200e8b4ff : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":1},{"id":"435e569919ef30a953928677dfa7ff725c793314","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-06 15:05:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b81dc4cee0b443032fe9d74357d4833e59692e4c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-06 20:15:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8863bc6419758756000ea7a35f292de3266cba2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 20:15:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"90ad2d49a0646ed431c71ec6cdf065b16fb2de8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 23:51:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0fccf994829ec3957fcea9311ff9ee8633112746","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 23:51:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e1c062bf74646cdb95827f801eb0b20 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/48dcdc0fc83d4058a2d26908d653e4df : SUCCESS in 5m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/258d7823fc3b48019fc9dd90addb00c9 : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d0aef0a6ab774642a5840610f2a29c96 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/40f2b90e917148c6ad62a82c43f1bce8 : SUCCESS in 3m 36s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1e9d84f86c1e434bb394b6a180c339ef : SUCCESS in 7m 59s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/4d1cf1b72324477e9c3f08dc7fe33d08 : SUCCESS in 9m 41s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/027c9382a6e840f58c34ff8cd0ae389c : SUCCESS in 10m 41s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/7de9995ff06c4e21868d26fdc4c2b1c9 : SUCCESS in 10m 08s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/868e918c5f614896aaea701ab11e5d8f : SUCCESS in 5m 46s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/b9290ab4853a4632ab5fde6c2d0da1c5 : SUCCESS in 4m 41s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4d6cde6cc1aa4ddda62ce9df40264ecf : SUCCESS in 12m 13s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/2a5c6eeb0e6e434f812d6f4e0c8b97b6 : SUCCESS in 11m 29s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b09c4d4a657a480b89e2168cb898e154 : SUCCESS in 14m 26s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/7f7b3aaced0e4428bdc9f39946e95696 : SUCCESS in 15m 41s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c6587513f8504d2aac53e5f6d176ef06 : SUCCESS in 24m 46s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/26e35939b3234b9a8408e557c37fd76e : SUCCESS in 25m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/82ba183ea2eb44a28a79643200a23c47 : SUCCESS in 13m 39s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/9c9d009d991d4e4aa2cabd9fe9e86860 : SUCCESS in 14m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3efe377b7a784d3a8eccac7f536aac57 : SUCCESS in 17m 32s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/2289b40569d349c5b1a9b160de1ccab9 : SUCCESS in 5m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1a488aaa7e524349ac6a4aa4405f304a : SUCCESS in 8m 58s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/03825cd85ef44c2b943b4ba0369f1813 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a95a4edfb7da457ab6aa8190ad0b74e4 : SUCCESS in 4m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d15515b8a2b545b5838a6f7d630bef90 : SUCCESS in 1h 04m 34s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/be0889a662f749829993ee56e028acc4 : SUCCESS in 1h 28m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ea9bc8cfc28640d0aebe8cf1dc3edd54 : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"270fc1c5c6e0d582472b117f34779d64fede2fa6","revisions":{"270fc1c5c6e0d582472b117f34779d64fede2fa6":{"kind":"REWORK","_number":1,"created":"2020-01-06 10:47:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/02/701202/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/701202/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/701202/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/701202/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/701202/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/701202/1"}}},"commit":{"parents":[{"commit":"58eb070abc027913e0e4d054af030989cf6fd8a5","subject":"mark regex as regex for pycodestyle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58eb070abc027913e0e4d054af030989cf6fd8a5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-06 10:47:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-06 10:47:19.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.5.9","message":"update constraint for os-brick to new release 2.5.9\n\nChange-Id: I9deaef1f9d4b79f2cfa82d675415b18b7793a4f5\nmeta:version: 2.5.9\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: I4ed8fc7a1cf48c6e0a0a56eb0b1a2fb57329c073\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/270fc1c5c6e0d582472b117f34779d64fede2fa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/270fc1c5c6e0d582472b117f34779d64fede2fa6"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
