)]}'
{"id":"openstack%2Frequirements~701406","triplet_id":"openstack%2Frequirements~master~I7ad1bed355734445c722f2001113b0d1a3a628cc","project":"openstack/requirements","branch":"master","topic":"cliff-2.18.0","hashtags":[],"change_id":"I7ad1bed355734445c722f2001113b0d1a3a628cc","subject":"Blacklist cliff 2.17.0","status":"MERGED","created":"2020-01-07 17:21:26.000000000","updated":"2020-01-07 21:15:17.000000000","submitted":"2020-01-07 21:12:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701406-1578431559549-716aec10","meta_rev_id":"8061341abe18e2d6b6d983a6b0079a464f05dfe8","_number":701406,"virtual_id_number":701406,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-07 21:12:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-01-07 17:21:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-07 17:57:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-07 17:57:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-07 17:21:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-07 17:57:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-07 21:12:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cddfdaff71a0963b2a636a96665355224a1813c5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-07 17:21:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46c811751837f07d4f56d2617d01325c5aa825e1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-07 17:21:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fd77ae8a6fceb37d50ee5ef704c62c97fd3571c1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-07 17:57:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f6f05b94720cc9da7911ca44f903300b4031bcbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 18:52:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00be8a2f373b4eb59b8bddd1b971da1c : SUCCESS in 4m 40s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/15941d427a534f7f88644d5436bd359d : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7929fcb71b9240959f9a7961ae456068 : SUCCESS in 7m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c15fc853f174ac5b3f85b94178144cd : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e0e11b3196745ba9156d58cec405735 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c79055d70b40472ea8be59b2fe60de22 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/964a2bfb0969487a976916031211d4a6 : SUCCESS in 3m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81de5c91080f469bbaf4dd08814faf33 : SUCCESS in 57m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/89de388d4b4b4665974d9105a1483388 : SUCCESS in 1h 14m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/19d568ddb1de4df3aef1d9531b6ad20c : SUCCESS in 1h 25m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0477b08a22544f90bc62431137944aeb : SUCCESS in 58m 09s","accounts_in_message":[],"_revision_number":1},{"id":"3c66a280013d8f62e496dba4a983315c2006f645","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 18:53:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"90adc3e53ba196b38e8c46f50a40d4c6d7d38ef3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 21:12:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77011d35e30b49f1873b50ed723f86ce : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e9b9b605cf5437aa597741353048a5d : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cb511e30e3094f1e80458c65ba89ac00 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8d86856927749a7bc22a36b6548f691 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/574cb8984a4540b09b49d7dac4fca286 : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/945777bd4a554c3980ea19e76525b65f : SUCCESS in 4m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/269a78713c404dbc92603ff00607c2af : SUCCESS in 57m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c17fdd52db6c433b9301b75bb50dcde8 : SUCCESS in 1h 04m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/68cb7a0a0efe4488ad2fb6a87d629933 : SUCCESS in 1h 17m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/26b27f7861a84d0e9ec8da22e2e4dcf0 : SUCCESS in 56m 21s","accounts_in_message":[],"_revision_number":1},{"id":"c07ec2f9179c8a30a71a6577e10db736af79eb3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 21:12:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b6f60f9190f273a0b51ffe1cf38defc195a07909","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 21:15:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe9610937984686993484fdb754aff2 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0dd8a4ecce508f7626be12f5316b9c242cec7bcd","revisions":{"0dd8a4ecce508f7626be12f5316b9c242cec7bcd":{"kind":"REWORK","_number":1,"created":"2020-01-07 17:21:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/06/701406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/701406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/701406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/701406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/701406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/701406/1"}}},"commit":{"parents":[{"commit":"c285569b069af39f306571ba21b7ea251f9ce722","subject":"Merge \"update constraint for keystonemiddleware to new release 8.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c285569b069af39f306571ba21b7ea251f9ce722"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-01-07 17:20:19.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-01-07 17:20:19.000000000","tz":0},"subject":"Blacklist cliff 2.17.0","message":"Blacklist cliff 2.17.0\n\nA mistake was made when releasing cliff 2.17.0. Instead of releasing a\ncommit on master, a commit from an arbitrary review branch was released.\nThis means 2.17.0 is actually 2.11.0 plus some commits instead of 2.16.0\nplus some commits. Blacklist this release.\n\nChange-Id: I7ad1bed355734445c722f2001113b0d1a3a628cc\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0dd8a4ecce508f7626be12f5316b9c242cec7bcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0dd8a4ecce508f7626be12f5316b9c242cec7bcd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
