)]}'
{"id":"openstack%2Frequirements~701418","triplet_id":"openstack%2Frequirements~master~I91fcf9c405f07cf849bcdbf6cc6263c4d21f1fb2","project":"openstack/requirements","branch":"master","topic":"drop-py2-package-tracking","hashtags":[],"change_id":"I91fcf9c405f07cf849bcdbf6cc6263c4d21f1fb2","subject":"WIP: Test dropping tracking of py2 packages","status":"ABANDONED","created":"2020-01-07 18:15:11.000000000","updated":"2020-02-10 16:27:23.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"be92e3e6bdecd5f0c83015f63f45647d2a9937c0","_number":701418,"virtual_id_number":701418,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-01-07 18:16:00.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-01-07 19:57:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-07 19:57:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6ce9f7bc6e4f04b1101f136956247fd86a7a5243","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-07 18:15:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c2fe9675b8266e0458ae2a55df6e0422374ff77","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-07 18:16:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d657b59465950c0edfd6800bc84fbc90cea3038d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 19:57:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0843f1819a874191969769c4168ef27e : SUCCESS in 4m 51s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/1820ac06de6b49bdb8cd1a383cb13be6 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7cf5bdba65340b495a95b493e0aece5 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/434eb4c04d61465fbe8da4456f5fbc93 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/798c86b32f6b46778b637ecca4d10cd4 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/737e8ebe0cfb48af9e541da11bb7ce65 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f8393a041dab44e389ac0f316693c085 : SUCCESS in 3m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecb4f282d1ac46f0a5f1f110d2616f99 : SUCCESS in 55m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5341c69ea163478792d3a9943222d1ca : SUCCESS in 1h 20m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5cd1d30ff5d448a5bd15256f04104634 : SUCCESS in 1h 36m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/160291067d5e4fb9bd7d40c9f0a7a1f0 : SUCCESS in 1h 09m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9345f1e256f76984cfe795ec9547700cba4a2bf9","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-08 09:15:02.000000000","message":"Topic changed from django-babel-to-enmerkar to drop-py2-package-tracking","accounts_in_message":[],"_revision_number":1},{"id":"be92e3e6bdecd5f0c83015f63f45647d2a9937c0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-10 16:27:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"afda2436a4cb76247115e3b177cb1b3f5854a473","revisions":{"afda2436a4cb76247115e3b177cb1b3f5854a473":{"kind":"REWORK","_number":1,"created":"2020-01-07 18:15:11.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/18/701418/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/701418/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/701418/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/701418/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/701418/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/701418/1"}}},"commit":{"parents":[{"commit":"5f4d1a64fd16c53d49d4d26796569d8b1a4e509c","subject":"use enmerkar instead of django-babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5f4d1a64fd16c53d49d4d26796569d8b1a4e509c"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-01-07 18:14:51.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-01-07 18:14:51.000000000","tz":-360},"subject":"WIP: Test dropping tracking of py2 packages","message":"WIP: Test dropping tracking of py2 packages\n\nChange-Id: I91fcf9c405f07cf849bcdbf6cc6263c4d21f1fb2\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/afda2436a4cb76247115e3b177cb1b3f5854a473"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/afda2436a4cb76247115e3b177cb1b3f5854a473"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
