)]}'
{"id":"openstack%2Frequirements~701692","triplet_id":"openstack%2Frequirements~master~I53701224298eb3abc6c2e261b1a7bcd51b8e6b43","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I53701224298eb3abc6c2e261b1a7bcd51b8e6b43","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-01-09 09:01:10.000000000","updated":"2020-01-19 07:21:58.000000000","submitted":"2020-01-19 07:20:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701692-1579418429095-9981773a","meta_rev_id":"710d8e56642481458e2b942e5694898bb2632613","_number":701692,"virtual_id_number":701692,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-19 07:20:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-19 02:06:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-01-19 02:06:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-19 02:06:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-19 07:20:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e77ccfbcddaed932a236653d642a131d17a5cace","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-09 09:01:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f2915b7b64fde29fe57e967594e5483cff32032","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 10:17:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80639fa1b3a741b7b726606f30ca72fe : SUCCESS in 6m 55s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/a1b80fb1c4114569be504687e98a0dc8 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63a6a835d4c249c8b8fe507dfdd0d4f1 : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b6bc5e369f04c3ab9409cddd992053b : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1924bdcce40949839cf4cb7e59369102 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87b6b3a5c8b04c1eb7c8808285b1c8e7 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1fa423f3b4794aaea9df84ab68de48ce : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0ec71d842c7145ba8f25618230c802e4 : SUCCESS in 5m 26s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d148d6b0a7314ac7a123d825acd10bc3 : SUCCESS in 11m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e5d9258d50224a93b40ddec19fbbe87d : SUCCESS in 10m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3bdba2b7075f47dc875dfbd3373f71eb : SUCCESS in 9m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4e72aba8e02545b79a48bec0c8c2e1ac : FAILURE in 6m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/140457341df4417c98a20ae51b95bb56 : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1d4283eb5b9747a686896a7aca711d1f : SUCCESS in 10m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4bb7802747c341ef95cf4cb8eaf321c5 : SUCCESS in 22m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7421f44e81c740dd967977b98115dbb9 : SUCCESS in 22m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f0e335c2c418446f8101769b25ffe373 : SUCCESS in 15m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cdc5b53be05a4755a1f73b368410c6cb : SUCCESS in 16m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e232f2356e524e2b9c998225b6bc8660 : SUCCESS in 16m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ba3ffdab4b6e431e9dab719250f319e7 : SUCCESS in 5m 22s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/cb71ffe1447e49b4a4a504a781e7b91d : SUCCESS in 6m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eaacb37a48994ad09e5c53cd6a45a8d0 : FAILURE in 6m 15s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6ee767d93c2c463c883554c080e98c1e : FAILURE in 10m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/442948ab4ee84c80ac30f9ce7b827d92 : FAILURE in 12m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b3e0dbfce104236a873fb80673bcf73 : SUCCESS in 56m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1426ae14852f47e4b953402c673e6dc6 : SUCCESS in 1h 10m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8256b687344f44cc81bd7685aadd6ae5 : FAILURE in 1h 13m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2ca3cee8460f472e97dba69437c767ea : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":1},{"id":"24825be497f4de2a3ba31c48c65d5ceb48daaa48","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-10 06:19:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e388da8430d437dc6597f7406a0afa23078beca3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 08:06:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1364edcb8d6414a889b381ff5d1800b : SUCCESS in 4m 29s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/074df705576d46ba8e07b4b53749bbc3 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0febfb229f10448f9796d84bff93a1e3 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8259f333a6a141ecb806e4e7e610eca5 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/812df9e08cf74210bdb7cfe9b961e41a : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0aa6b9d861004e42a6b20e54afaa411f : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b58dade8079d4890b7411230741cc805 : SUCCESS in 3m 52s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1c87668d61f44698a276532ea2ca356b : FAILURE in 10m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/49e311e42a014d188f8a7df3c88f5ad8 : SUCCESS in 9m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e19fb31b2db64f8a9d7edc51d2853b29 : SUCCESS in 7m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e8845c287d684e7a90143b4dcb0a5920 : FAILURE in 5m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dc7925d1d88f4041aaca19feea3e9b1e : SUCCESS in 12m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff99521bc9c5417dba11813788547dbc : SUCCESS in 8m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bfe278bf14fb4826bdff7882791fe2f6 : SUCCESS in 16m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/09b4938dc731405693355dc87568b612 : SUCCESS in 29m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/83e1d5f236684da5ae35ce2a667c3447 : SUCCESS in 14m 34s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fc8863b1e504437abf3561ddaf8b486b : SUCCESS in 13m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9fb9cd529c2f4fb9a29673c1c0129d57 : SUCCESS in 19m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/be2b09d83d404b4398477f0154a253c0 : SUCCESS in 5m 12s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/95e63c2fd2374dd7bbfd9991ae6d4c10 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0cd16b27c9074135bf80410c06c7f71b : FAILURE in 5m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/faac8156509f4071bf88be3ce5a7648c : FAILURE in 8m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3d26bad04e664d749b19428e94350ad6 : FAILURE in 16m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/20c580d567944f36951bca426e8b5a94 : SUCCESS in 1h 14m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4b9b149c6bac4249a53df504f5163d3a : SUCCESS in 1h 27m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dc15aa65ff644467bc44269aa11bbdae : SUCCESS in 1h 25m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/94bb4b1f929e4c1fbb67cef26442ddee : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":2},{"id":"4f841afc1ca3ed0e024a45be67fa74471d26548d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-11 21:05:19.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"c4579a36ca47ba3e68d63735a98cc0d2ea14e902","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-11 21:06:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"84a482e7a743b7cc54d44741ec109031c4b391a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 22:29:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5272a7b285ff4175b67e2148842f1886 : SUCCESS in 5m 59s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/38ada8be4d1f4a5ca2b074f7a55be0eb : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47a354469c634a268d947643a9aeb378 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88e9f386d60e4b888a94b781ee5b06fd : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4719f9f468c64d1899925c633ac27dae : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/23ad785e43724c05bf8da5849f944f0b : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b2ffc36fe4494268af21c5b14bf7103c : SUCCESS in 4m 48s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c2532dcb286d4041bba22634ffb0f0f2 : FAILURE in 7m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e8dd0e2c32f04bc0878761bf862599d5 : SUCCESS in 8m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7a19eacef9d6430981e442a7cccea214 : SUCCESS in 9m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/697912f6790b4037afe3825443af6aeb : SUCCESS in 4m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a4d2a7a2fb304699b3e4414fbb1baf9b : SUCCESS in 8m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/24b68af894bd44b397733bd0af9d8c08 : SUCCESS in 7m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/82412373395b4a21a8c39be07a16836e : SUCCESS in 22m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c8b4caede14f44829a7fcee4e91885a0 : SUCCESS in 24m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fef83e9f0793483a98d3d3c8ef02076f : SUCCESS in 16m 10s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d7edd56a3c0443b3bd065c9dd77842e6 : SUCCESS in 13m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4676dd7245924ee39ef4b56a4d291399 : SUCCESS in 16m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3b587047aeea46d6850fb14f28c77a47 : SUCCESS in 4m 03s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/93e590c68af047cfb0b54a244e458e0c : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e95744d03cec428dae86f4c9db3600d6 : SUCCESS in 5m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ac555b5e63d04c0e96593cbe7446aa0f : FAILURE in 9m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6cfa73c435fd495e8fefe23111d50bc1 : FAILURE in 7m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0dd4d3e6d8a41cfa28c34962ad69058 : SUCCESS in 52m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/aae60d2cb4b7440f85e5ae945b6b16ac : SUCCESS in 1h 07m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f7068f9d77c84d9992887f4db09e94ae : SUCCESS in 1h 21m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bb87dd7b339040aab7e3b4050a1a139f : SUCCESS in 58m 39s","accounts_in_message":[],"_revision_number":3},{"id":"85783d2b04e321ba163ffdf3a98ff468cc5e78d1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-12 06:17:08.000000000","message":"Patch Set 3:\n\nrecheck\n\nmaybe it was coverage that did it","accounts_in_message":[],"_revision_number":3},{"id":"d4f6c282aea16109372c6eff19ab16bb08e12fe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 08:05:21.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8469fd3e42f4dfbab2da855c5c480cc : SUCCESS in 7m 04s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/4a06c3ba0f284feb9b4d48a9318003ea : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca42bcf2243e41a1bef73f42ca5a22fa : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf8e9c909ed44a06bbdb9631c020efe9 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52e47a86036e4570af86c587f4740bc1 : SUCCESS in 5m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f5786ca77d3a4256bd78d78e0b486060 : SUCCESS in 7m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a1e7d81fa8d24334bc5a8d3013a6e2fe : SUCCESS in 4m 50s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c71cc720d2a542c7a3316fb8e32eac7a : FAILURE in 13m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/07dcab6ba8774d2d9ff17a2075844a67 : SUCCESS in 13m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/312890e627e24d06b17653fc96f6e5e5 : SUCCESS in 8m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ecdf2a92ad3d4f25ae7ad02cccb3fba3 : SUCCESS in 5m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3e8eed3e678943bd80563939a28bc13b : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/870b43a229f04dcd9a949695b5899ef0 : SUCCESS in 7m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c601e80b82074a42af84521c8e6181e8 : SUCCESS in 15m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1cd7ffe4c6fe459dbf3d09bc53880071 : SUCCESS in 26m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/37bec89b33fc4596a64143398834f6aa : SUCCESS in 14m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ac9476ca02424e799e48aa5533e851ab : SUCCESS in 13m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1f8e48373174488fa0bf7593b82c265e : SUCCESS in 18m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/762c9de1bdf2416898b07db473fb6dc6 : SUCCESS in 5m 48s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b1ba1c91091b47819aeac874bd407871 : SUCCESS in 9m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/15caa1cc9080498a96c45be303d7f171 : SUCCESS in 7m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e8e7707946584a70b6bd1d9eb78cc181 : FAILURE in 11m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/956fc146511c483d813d4426fba81175 : FAILURE in 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f64678f7cb4d4da99587c324f6a7dadb : SUCCESS in 1h 04m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6ff9d39f0b47438ebaecb680f69796fd : SUCCESS in 1h 14m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1365f5bdbc0c463290649fb97c65061b : SUCCESS in 1h 41m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a6ee0fd975b64c4b8b3175e34ffce1e5 : SUCCESS in 1h 13m 16s","accounts_in_message":[],"_revision_number":3},{"id":"35bc0b9dc5f2659a2f8d55fa04b4c2f4f2f18607","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-12 08:29:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1b9336c089f9b63f7cb8175c7e587146be275755","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-12 08:31:06.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3570135e0d2493981e2b1d71268ff589528f9c89","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-12 08:32:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4be1bb7e94c1aa483a9427bf619318940ae9fa98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 09:08:39.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cf89b4465564d02ac691d49be43d091 : SUCCESS in 3m 28s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/551a6de402b448889251f676c3c77e2c : FAILURE in 3m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2a5bf77910542c7a48f7b94cb8139c8 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8376141607a4f00b058ce00641edfeb : FAILURE in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/02bab7a5b8354f1198d16fce33550719 : FAILURE in 3m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/10a3d81cedb142b0a5478510a5f6b5f1 : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8dca1f3d70f84f5795c6c98f458cdd3c : SUCCESS in 2m 56s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f4389cb1054445ee859608f4c24ee021 : SUCCESS in 7m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7afe1c1f238d4d24a2b97085c74c0b1b : SUCCESS in 7m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9daa665d5502440aa64aa1e8aac4c609 : SUCCESS in 12m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9242f6f9896840e1bda1bc6d5edb8c79 : SUCCESS in 4m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/421de44d68a04769a39acda1d76bc197 : SUCCESS in 8m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5f3c1675df484a1ca6d20f1704d482aa : SUCCESS in 8m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/339a6ee139c44f6dafc806c7d72a4d54 : SUCCESS in 13m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/300c8bf76fd9472aaf0308cdbfe6790f : SUCCESS in 21m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/80aa3cfa3f724432af3f83d294dbc563 : SUCCESS in 11m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/666550b917e84649adb9b7954b4ba426 : SUCCESS in 12m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/88a1305c8ae84d6086bc9b7165c9990e : SUCCESS in 16m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6a38bea341b14bb0b0d6a7eb83f48c9b : SUCCESS in 4m 02s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7084724023b74e27b236928e3bbf0a09 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/709e20ee492147548f571a24436f6a86 : SUCCESS in 5m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/295a9c8409e94d9aa845c90ee446aad6 : SUCCESS in 10m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d3cf297e58284a8c9a6e853cee3cc612 : SUCCESS in 8m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bfc25287dabd4de482952449fa4e7d00 : FAILURE in 12m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3c0ad0b1348a49ecb95681729f10b75a : FAILURE in 4m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1aa60d88f28f4cdbb84c6e5546550b99 : FAILURE in 13m 07s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cf795bcddba6408ebe2add0934518e6f : FAILURE in 3m 31s","accounts_in_message":[],"_revision_number":6},{"id":"094d9c791b5e6433017220a895ac72a1b7214caa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-12 18:59:58.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"6ed819fb96f6902882a174b1b7d2bd857d302eee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 19:29:58.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77211b884e9b4c9183ea9941b3a9ef7a : SUCCESS in 4m 18s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7eeb6ce76d0b483e943d1b92b5be1ec2 : FAILURE in 4m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98486b103829474fb87b49bce3afc579 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f726b064fc814dbf9559a23b0709a037 : FAILURE in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ca48845ed7648199f59debdb43d1379 : FAILURE in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2daf24a9f9b46e39c596c1a92ca5122 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/28857683bfa1400fbfee81ef3ee716e7 : SUCCESS in 3m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3a86a5211f0747e69a3ffde321f8f0bf : SUCCESS in 12m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/06acab9baa2a46429a8a8f7e090fa568 : SUCCESS in 7m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/56ffc30c3e3b49b59a49ebbca86cb70e : SUCCESS in 9m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/344c6493150f4710b38999688c5143e4 : SUCCESS in 4m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7ade314a9eac4d888501421f63eb65e8 : SUCCESS in 10m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c25edff776a34eba9f19e25b9c4319c4 : SUCCESS in 6m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e632ff60706d4f2182b7895b75ca1d03 : SUCCESS in 20m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3ed89124186146b194c032c6471416f3 : SUCCESS in 25m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1daef76a24dd41cdb86ec3a11dd7d746 : SUCCESS in 11m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4e47dcd5aab749e3a68e74abf581ddd7 : SUCCESS in 13m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/71e3567ac80048e797ab70fcbe4ffde2 : SUCCESS in 17m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/249930312bb340a68d49abb8b87b0320 : SUCCESS in 4m 16s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e0bf54adea304757a8693f5ec56c8242 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3d67310bcf994fab8dc543807e2e0e62 : SUCCESS in 5m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/597d184d47a545fe9e55f3c2aa771c2e : SUCCESS in 11m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7ad83327cc6c4af086ca1d69500b0d13 : SUCCESS in 9m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23c8d1556498437eadc5f27ad4329295 : FAILURE in 10m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f5c1cc9f3e694f51b34ef2e2aab91a21 : FAILURE in 5m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d657cacc85814d2384a78fc61ad8635b : FAILURE in 9m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ca6a48527d694bad9632c3fb25af81b6 : FAILURE in 4m 16s","accounts_in_message":[],"_revision_number":6},{"id":"f293596406c898e01942ce51df0b90c8f9237c25","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-12 19:57:46.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"b281c58da9eec3d44ef1bd997ebaeb3b1adfdcef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 21:09:33.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8512eba03ea242ff9a3c3f7a79418e8a : SUCCESS in 3m 30s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0d6ed41203c148d287998072def1dbe4 : FAILURE in 3m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7280cd37591547f4bf7756550851b3cc : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99f9c9f5479c47fb9e7c802e131d7961 : FAILURE in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f28539809b4b4399818ac216e0fe9aa3 : FAILURE in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/09b309a135a44dcd9622b1938d252998 : SUCCESS in 5m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7f50e47dcec3460d9cd39ab5e0c78c03 : SUCCESS in 3m 18s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7911287412de4675b6b6cd4ac17311e2 : SUCCESS in 6m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0a21ea692c0d4933a81431f6c2220906 : SUCCESS in 7m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/261639105e4e4d3088142875fc87afca : SUCCESS in 8m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c27ff7d31e1341f69b95adb5de5212d8 : SUCCESS in 7m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/70c06127fb8847d68a37c1d11ea33c8e : SUCCESS in 8m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/da4dffbfee4c4b67ad45611de12417c6 : SUCCESS in 6m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/16b1d9445a1b45138e945a884266d483 : SUCCESS in 13m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/92b790b19a61499e9b9780d7af55ab13 : SUCCESS in 24m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5b67e5356db640c48996458c17ef34d8 : SUCCESS in 18m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7146eeab7f014a95ab3cc04ebb775fc7 : SUCCESS in 14m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c91aadf49ea743569285adfe3c9f2a32 : SUCCESS in 19m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5980df217d47434c9af18813c3780ebe : SUCCESS in 4m 01s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/44276741ebc645e3b777944643048091 : SUCCESS in 7m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/03714aa4ac784836a39792905a63da4c : SUCCESS in 5m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c438b020c0a44b1b9304d2f70ed7fcfd : SUCCESS in 13m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4153c394b70e49b2af0ce189717d1891 : SUCCESS in 10m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e13238480cc44a92bd2ca17763d57711 : FAILURE in 11m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9d68bd2ca8664d51903124a603912889 : FAILURE in 5m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6324be1b7ebf4fa9974976c001976dff : FAILURE in 11m 20s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0c0470d0eb3648fbab4bc5bdc6f511cf : FAILURE in 3m 41s","accounts_in_message":[],"_revision_number":6},{"id":"060f7f35ae29ca09814d998b0979cf64673ee687","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-18 22:38:59.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"ef2d77dd6b148cc6a261c5c8c2f7bf5d795b10b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 00:12:39.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37ada9876f534a1d906d2636e2a4acea : SUCCESS in 4m 21s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/253a8c27d8ff4d76b96131d158a50952 : FAILURE in 4m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33f07ca8039b44a9960abea21590fb4d : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e31b7b8dbfd94cb7aa8cbd408eae4966 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b49387ff1d8457da44570072483378c : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cfdfda854c76461485062481fa07abc2 : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f4e449384e2404592e761935ad57c16 : SUCCESS in 3m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/686da491505d4331a65597db8b701076 : SUCCESS in 7m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9b479441f5ec45d098a690961e66ee47 : SUCCESS in 8m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7f5081e759a84486b89e51c7568a2e71 : SUCCESS in 4m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3e9ce7a144e14b3c9b088d881ea5f935 : SUCCESS in 8m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c0c95412302c49b39e2bc6d89e122a9f : SUCCESS in 6m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3b35313b9a864c8db939c9bab2db9ec5 : SUCCESS in 27m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/351217b14dcd4a8587c4881c609e8a05 : SUCCESS in 22m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e54a3f40935845188ae9aeb8d2a64d3f : SUCCESS in 21m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/87d6591a922c4a9eaa66d55273084562 : SUCCESS in 12m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8c6255af62084bd1bb3ec522debcb402 : SUCCESS in 21m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d4020b56d7104017932fe9220963fcd6 : SUCCESS in 4m 27s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f8a66f5d70424d77945c6b77af439a1d : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/42e914013c75476ab47ce89c5deeb681 : SUCCESS in 6m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0301d5b9dfb94160ad5807f85b7ea94c : SUCCESS in 8m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a7200141365e4a43b641dfad4db67a99 : SUCCESS in 11m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdc600fc06ab4cd5ae50509deacce81f : SUCCESS in 53m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3835cae51c724a5db000077227bd7a65 : SUCCESS in 1h 16m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8d036103c3e04c19a220b7ae613ba4a6 : SUCCESS in 1h 32m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/acdd0f85a74f41e987bf33c05dd4b8a7 : FAILURE in 1h 01m 36s","accounts_in_message":[],"_revision_number":6},{"id":"b39a33bdb85fa9cdc249b7a060bf72e53ad3b5d2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-19 00:26:24.000000000","message":"Patch Set 6:\n\nrecheck\n\n4da6e47b8f7e2912de062d9e70f4fb3c06090b32 may have broken pypy support...","accounts_in_message":[],"_revision_number":6},{"id":"7bdc7ebd6d1474e8546e431dae9ca747810d1bb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 02:01:04.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bc452b03c27420497db183a53835467 : SUCCESS in 4m 05s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0cceea6ca0e74a66a65d5d6195b8d44b : FAILURE in 3m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65a21c71b7094be981672d353e3731b8 : SUCCESS in 2m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c5702a194fe4f5797e749e455e3e087 : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8bbace05508f4ba3b9498cda5236c4d4 : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a349d70282f477ca57d8ef78cf2186c : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c80573799ff4bc7aaafe280a26208f9 : SUCCESS in 2m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cb00f000753e461cb852b6627365aebd : SUCCESS in 7m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f267e85628e7431883a7fbcb5fe68e9b : SUCCESS in 7m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b97032f6a9584746958af8f22c2e0f09 : SUCCESS in 4m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/09c533e77cb2473eae5cee2cabaf94e7 : SUCCESS in 7m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b410acaafbba4119988d5a476131581e : SUCCESS in 9m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b1fd071fa9ba4734a1f9392cd356e886 : SUCCESS in 59m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cc817bc04685419089a7335f08bdfaa9 : SUCCESS in 27m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a4312ece84ef4a6e8871857dd6a36a73 : SUCCESS in 15m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fea918f8a0884b4ea7428bf936a4ef43 : SUCCESS in 14m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a24823a22caf418fbd92bc3a0b8cd969 : SUCCESS in 14m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ee6399d877294277b1062324c01cba3a : SUCCESS in 5m 28s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/fdecd3cb17cf4cf6be0c225c615da644 : SUCCESS in 4m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4a72ece616704e0abf46b190b58bb119 : SUCCESS in 6m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d47190299a4f4fc096e53429a2b64fcf : SUCCESS in 20m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/28836f569d9b4e41a142dc24dcc4ed23 : SUCCESS in 8m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/487e23b0eeef45d59b190b5ba6810c00 : SUCCESS in 53m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2fbd42cf6f7b44778390e22285a2835e : SUCCESS in 1h 14m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/44a500183ca549a8a86f700e29d78600 : SUCCESS in 1h 29m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5ef339a75f2c42fea13559f5c0a58cc9 : SUCCESS in 1h 00m 03s","accounts_in_message":[],"_revision_number":6},{"id":"b82c37068f7dfe886784be31d11874a52825601f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-19 02:06:57.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"65de45102200d0629d1100778ece1fa6d1639568","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 02:07:09.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"206fa3c3162ac9498d41192f9260efa628bf56e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 04:10:42.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65ee8a22be2f4877b555fc0c5497cd01 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c9bfc8aed9846b5b8c61a3c69b527cf : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b0ea6810a35429abd89abef5e16158a : SUCCESS in 7m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87d884ea126e40a8954b0566826f30c2 : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c32af1178c384e0996da54a1a3e68d26 : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/27cc9bfdd85a4934ba7dc7ebbf6872a2 : SUCCESS in 6m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c76cef3ced764ea1b1eac95d6d549f56 : SUCCESS in 14m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0499441758da491f9918f2035d079fe5 : SUCCESS in 7m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f3723d8840ab4381bdc5c519ce2a019a : FAILURE in 7m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/65fac4627f7a4ee1b06d7bcb1c88999b : SUCCESS in 9m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/53e3d2c4943a493fbf2790cf075778c0 : SUCCESS in 11m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c6ec81f51dec442aa405495cdb17a778 : SUCCESS in 13m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/03ef6476473c4377b2c1dd734f02a36b : SUCCESS in 34m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/82fd2bf3e9174d72932f15a10fe1cd22 : SUCCESS in 14m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2c900b1d3e514fa0bc7f8f9b78f50e29 : SUCCESS in 18m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/feda11c4a62e45508c59ffbb9f5afd97 : SUCCESS in 18m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f42d8697b66b47b6a5007a6cf845b43a : SUCCESS in 6m 33s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c702fcde0e9146ff98916afec8d85a55 : SUCCESS in 7m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fae95ec524d843458031dac58cb06517 : SUCCESS in 8m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c051a2277f4249688e3d02717cd4e821 : SUCCESS in 10m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4490f29c597047e28be3e5d63d2f66f7 : SUCCESS in 10m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be09550830ae4ca084404d9e0b4a2714 : SUCCESS in 59m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5d6977eda0b4ffeb87a38504134bfad : SUCCESS in 1h 06m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c789174cc7064e6386194118ef6fb9c2 : SUCCESS in 1h 58m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5cd7a574c89f4c1abc18366060e31e61 : SUCCESS in 50m 49s","accounts_in_message":[],"_revision_number":6},{"id":"72d5d5465998fa224eca7cb5d6c1e181907dea36","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-19 04:13:56.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"ed30b8b1a061fae3891defe79ebe490a6bafbc4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 05:42:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b42fb0d469d4d0a80fe606f2443c248 : SUCCESS in 4m 55s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/da7b37d8e7584a97b520beae79cf18a6 : FAILURE in 4m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f94251afa4d14c488633119c0f04b619 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4be82d6b76844f1f9feff74c4c72d9f5 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e448689b843747098b0bd6b9970cf302 : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7ce00ab58aa427f928e33acf5cc3355 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/298e4ebd7097496f9ce08fabc26385da : SUCCESS in 3m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1a4ea728763e4e3ab7023f07d99347a3 : SUCCESS in 6m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6e1ab0a2ff50451588cb95576004245a : SUCCESS in 7m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f5921ca5cbc04b60be99d9e0ccdd15be : SUCCESS in 4m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a936ebcd3975451194178fb2aedb6936 : SUCCESS in 9m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/93a66a6f56aa40bea083505854e8f69c : SUCCESS in 8m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bc2b18a77ed047b48e1076e508b076f9 : SUCCESS in 17m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4fede9d7875e46ecba1d54eff6724fe7 : SUCCESS in 21m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b861c20e14fc464aa7961392de3d9426 : SUCCESS in 11m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7766e06759584313bf3c94651193dc37 : SUCCESS in 13m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d45cc8194612495cbd24d190d063af0d : SUCCESS in 15m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/aff5686838954aa99b4d33932ab87c18 : SUCCESS in 4m 19s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/491c1eab908a4f4a8055927b31b7eb49 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a310e137d7b643e3bc7dd2e36da57ca4 : SUCCESS in 8m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3ec47a2451354ddcb1f5b4c8f34803cd : SUCCESS in 10m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/88af716c1a0e420b8f05df085c8bcb27 : SUCCESS in 10m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1161c07dbf5f4c1fa77ab2a05b5bbdc1 : SUCCESS in 51m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b1c45ad0cace49adb8db98ca7d149ce8 : SUCCESS in 1h 07m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/93f9477d579a45b4b5b115a055e2307c : SUCCESS in 1h 26m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/aaa6f3d35c0341cfb8daf4bd5c3927a1 : SUCCESS in 1h 00m 03s","accounts_in_message":[],"_revision_number":6},{"id":"d62549c90237f8c6b48ef5110d4a58b236c51879","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 05:42:37.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d035ff145b78bdb46d304e3c2765960fdbaf3d5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 07:20:28.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1245954d410f4c839cfb1fd7c2a3de9f : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4578f76e410645bdbfac0717f2c34fc9 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2952f3b71c624d48ae245d4402a44814 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2aee4d7f0e654b0c8a44501813966572 : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/70ab35781a114b399ceeac49744e250f : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8df68fff615e40969c233f5922fd859b : SUCCESS in 3m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f94cd786823948848363b348bb204b35 : SUCCESS in 6m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5d0acf22d0624a2180ebdcbf96d9b012 : SUCCESS in 7m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0555af0d4fc441dd9610334d837071d9 : SUCCESS in 4m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e9b36e909ec5445db9539ead6d51007e : SUCCESS in 8m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09b83582c46d46979ec54328aa043e34 : SUCCESS in 6m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/443209e7428a4c29b313a9d122a72b62 : SUCCESS in 18m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cb71ff6806024691b7dfd06ed0215ca3 : SUCCESS in 22m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a8c870ba457942dfae0476f31ae64943 : SUCCESS in 11m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f6b10c454c5e41a6a95857ce0acd0f1a : SUCCESS in 14m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/77c6d1c9cd144ee58fe18adff7310578 : SUCCESS in 14m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bc84ea1c08c04ffbb00d5128695fda2c : SUCCESS in 4m 53s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/97268d1acc73476ea28844230c64b07e : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/40082ba4a0434b8c8241e82b70ae4ce4 : SUCCESS in 6m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e1163f2fe00744059fd218961326b0be : SUCCESS in 10m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d7be07eb92b746bd8b35be0588582b94 : SUCCESS in 9m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/935ae35049e94d939d8741855108cff6 : SUCCESS in 58m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/feb0d563daeb45058d4fb648ea83d50f : SUCCESS in 1h 06m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7428b82602a94eb2bd19644de56afec3 : SUCCESS in 1h 37m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6c41498cc68744f6add52760013ead68 : SUCCESS in 59m 24s","accounts_in_message":[],"_revision_number":6},{"id":"41020fed7fc21a40f24b86e890ab22fd37a102dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 07:20:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"710d8e56642481458e2b942e5694898bb2632613","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 07:21:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beb75cf6015e48d6873f659ad2c8f337 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"88825ffa4716d9fde812eae9edaea49843552e69","revisions":{"cfea28e93dda04ccd8a4e0391ce956178e47fbbe":{"kind":"REWORK","_number":1,"created":"2020-01-09 09:01:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/701692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/701692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/701692/1"}}},"commit":{"parents":[{"commit":"1a6430f8785b7f7d1f9b6fc7beb6694727c00a73","subject":"Merge \"update constraint for neutron-lib to new release 1.31.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a6430f8785b7f7d1f9b6fc7beb6694727c00a73"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 09:01:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 09:01:08.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I53701224298eb3abc6c2e261b1a7bcd51b8e6b43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cfea28e93dda04ccd8a4e0391ce956178e47fbbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cfea28e93dda04ccd8a4e0391ce956178e47fbbe"}]},"branch":"refs/heads/master"},"cb122dc6db12010b0d56ad9e7a02eb92aa990db3":{"kind":"REWORK","_number":2,"created":"2020-01-10 06:19:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/701692/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/701692/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701692/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701692/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701692/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/701692/2"}}},"commit":{"parents":[{"commit":"5ab33e985c3197cfe327c486c4b365f142843ebd","subject":"Merge \"Add flake8-logging-format linter to blacklist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ab33e985c3197cfe327c486c4b365f142843ebd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-10 06:19:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-10 06:19:45.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I53701224298eb3abc6c2e261b1a7bcd51b8e6b43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cb122dc6db12010b0d56ad9e7a02eb92aa990db3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cb122dc6db12010b0d56ad9e7a02eb92aa990db3"}]},"branch":"refs/heads/master"},"dab5ee20d78d83f4d07b5bcf222da389a5f6c499":{"kind":"REWORK","_number":3,"created":"2020-01-11 21:06:58.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/92/701692/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/701692/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701692/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701692/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701692/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/701692/3"}}},"commit":{"parents":[{"commit":"6f4854db3687bebf34a6a62ed1cd92e5257bc36c","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f4854db3687bebf34a6a62ed1cd92e5257bc36c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-10 06:19:45.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-01-11 21:06:29.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I53701224298eb3abc6c2e261b1a7bcd51b8e6b43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dab5ee20d78d83f4d07b5bcf222da389a5f6c499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dab5ee20d78d83f4d07b5bcf222da389a5f6c499"}]},"branch":"refs/heads/master"},"fcd39b5ec013ebc5f0b6ba6a9abd7fcccdcfae1b":{"kind":"REWORK","_number":4,"created":"2020-01-12 08:29:32.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/92/701692/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/701692/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701692/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701692/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701692/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/701692/4"}}},"commit":{"parents":[{"commit":"6f4854db3687bebf34a6a62ed1cd92e5257bc36c","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f4854db3687bebf34a6a62ed1cd92e5257bc36c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-10 06:19:45.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-01-12 08:29:20.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I53701224298eb3abc6c2e261b1a7bcd51b8e6b43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fcd39b5ec013ebc5f0b6ba6a9abd7fcccdcfae1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fcd39b5ec013ebc5f0b6ba6a9abd7fcccdcfae1b"}]},"branch":"refs/heads/master"},"42fbe7d7302b9b9b03b94cf0876f76c2691c4bdf":{"kind":"REWORK","_number":5,"created":"2020-01-12 08:31:06.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/92/701692/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/701692/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701692/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701692/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701692/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/701692/5"}}},"commit":{"parents":[{"commit":"6f4854db3687bebf34a6a62ed1cd92e5257bc36c","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f4854db3687bebf34a6a62ed1cd92e5257bc36c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-10 06:19:45.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-01-12 08:30:55.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I53701224298eb3abc6c2e261b1a7bcd51b8e6b43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42fbe7d7302b9b9b03b94cf0876f76c2691c4bdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42fbe7d7302b9b9b03b94cf0876f76c2691c4bdf"}]},"branch":"refs/heads/master"},"88825ffa4716d9fde812eae9edaea49843552e69":{"kind":"REWORK","_number":6,"created":"2020-01-12 08:32:13.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/92/701692/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/701692/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701692/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701692/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/701692/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/701692/6"}}},"commit":{"parents":[{"commit":"6f4854db3687bebf34a6a62ed1cd92e5257bc36c","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6f4854db3687bebf34a6a62ed1cd92e5257bc36c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-10 06:19:45.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-01-12 08:32:01.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I53701224298eb3abc6c2e261b1a7bcd51b8e6b43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/88825ffa4716d9fde812eae9edaea49843552e69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/88825ffa4716d9fde812eae9edaea49843552e69"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
