)]}'
{"id":"openstack%2Frequirements~701712","triplet_id":"openstack%2Frequirements~master~Id1de2c4ee83d2093713f28ee2984204c50d68d58","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id1de2c4ee83d2093713f28ee2984204c50d68d58","subject":"update constraint for python-octaviaclient to new release 2.0.0","status":"MERGED","created":"2020-01-09 11:11:10.000000000","updated":"2020-01-09 20:20:25.000000000","submitted":"2020-01-09 20:16:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701712-1578601001320-0e1c537b","meta_rev_id":"8ff24a0a67a3263e7a3ea5d35edb84f34937f8ef","_number":701712,"virtual_id_number":701712,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-09 20:16:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-01-09 13:37:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-09 17:36:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-09 17:36:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-09 13:37:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-09 17:36:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-09 20:16:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1fae4c1d6ccb6debae293dff03a5925602e1f89","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-09 11:11:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5e3cd0632314317d5fa7e2422a86b89015b6726","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 12:40:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c16b202a3b4f463a89c1763983dd29e7 : SUCCESS in 4m 34s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7a09f925c2d74188a810b3cf5aa0d6a5 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7cd758eb01a41d990a4b92e41de535c : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/535334c3a87f44e685271a2d0e4fcb24 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0276ac6e59c9470b9bc9e99df25fac76 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/515c13fcba6643da8772e8f0f9183a81 : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/56596bb65330479288ab19506e93dce2 : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f87d86e8e52242beb6d08314667719ee : SUCCESS in 3m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4565731730444581acb11005dfff8e21 : SUCCESS in 7m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4c0b4babb9d24dc3acbfa69df53c125f : SUCCESS in 7m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1cc8123de177454b8007e05e28b680e1 : SUCCESS in 8m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f130324e6ff94648837bdf385b88ec1d : SUCCESS in 5m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1138ff39e6c14135b4918fcf8707a311 : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/90b77f2e65cc45b996f0d9a99650f4fb : SUCCESS in 7m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d39df622112445ecbcdd16563d06dce7 : SUCCESS in 13m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7e81e31473ec410fb0cdef4b232f25f0 : SUCCESS in 26m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ae3ac097f1e34145b92870d474399930 : SUCCESS in 14m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6d24b530c4a745eba6bb3095a7b7743f : SUCCESS in 12m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4f253c86b9284687882ef30a63f44967 : SUCCESS in 17m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ea50f3b804344942a8fa67434aef26d7 : SUCCESS in 6m 07s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/02099835e17349bcac69bdf92cd10a6e : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/da37b94a1e634af3b6f066938f77721d : SUCCESS in 6m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/81c8f91cd7fc48a886ae2b22103d57b3 : SUCCESS in 9m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/395d888eb2794c46a2069795398e3fc5 : SUCCESS in 9m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ceeed8271e54052aa475bced6cca14c : SUCCESS in 59m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/71dd6563b1df491b99af85982d4c6785 : SUCCESS in 1h 11m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/812b43ee6293482d9b2e867fe4deb415 : SUCCESS in 1h 24m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d7fe9b64f5464689ae6a85c638dba285 : SUCCESS in 58m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a5a1136c57f944c5f1f32cc406daad60011555ce","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-09 13:37:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5254656452ed2774818c668affcbebb9a3c2f4cd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-09 17:36:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f9a4effee7a14b93a7a33a10bde4522f47ff34a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 17:36:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9aa2023b8175926718e8e4112518c4ed6db6c697","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 20:16:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a65317bc2eb8a3cf4ade47cbc9f496b314ecab07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 20:16:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6a74f2d98ad45968ae5726633cc66fb : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d27b220a60c243e994408407565b911c : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/65a2604e55b44dba9dfa3d970da6a7fb : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d295d46f3f6d40fca062ec65e3d311b2 : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fcca775a08f046f8b80a776209bd24a2 : SUCCESS in 6m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad9d2ab4fc2f4f53ad12311ca811ff9e : SUCCESS in 6m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3d96219addb74d669b874bab456678b8 : SUCCESS in 5m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/802ad0fe4b1146ce8ae168e93e64e484 : SUCCESS in 8m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4aafe977393a4233a052d421e42ab128 : SUCCESS in 9m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7490db4e1bd54278909a1a0ae3dd439d : SUCCESS in 11m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a4ee3c07f1494b5cb8856b9561f9d45c : SUCCESS in 6m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2cc92fb3dbc14dcab0868fd71e9736cd : SUCCESS in 13m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e7a307be0e624143ad6fc414fad94a5a : SUCCESS in 7m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0aca1320ac4b487baae601f1e9886d95 : SUCCESS in 17m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b72bea5566724d9fa490e387d38d9dde : SUCCESS in 28m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fd8567664a3a4915ab7544ef6cbb8dbc : SUCCESS in 14m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fe3d9f6cfd5b4290a24eb9f116d5928e : SUCCESS in 16m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/020b257d39604af49505cb4c87ecc41f : SUCCESS in 17m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/91512995cb2f4bf0be9f4f6874e59cbf : SUCCESS in 5m 50s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/3a5825a8e37e40b2b04d349685976837 : SUCCESS in 7m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f4538ae18dfe4979ae0bc3fe8fd9cc70 : SUCCESS in 7m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4981bec8ae714891aff904288b370a16 : SUCCESS in 9m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/70cfcd85af5f49c9a7a46e24f2211eab : SUCCESS in 10m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f09df0441bd4e76a200e2b95f1536df : SUCCESS in 52m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e21138c91ab74abd9f5230cbefac4a32 : SUCCESS in 1h 03m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c9e12400217c4b41a2590dc31654567f : SUCCESS in 1h 49m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/929e370118ed452289c591606f6f1502 : SUCCESS in 1h 03m 42s","accounts_in_message":[],"_revision_number":1},{"id":"baa84b46aee35cc95646cbcf3c01501ff414ed01","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 20:20:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a7d62d7097e4ca795170496f00daa65 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f4cc1f712747497e11e0bbba27225b6804032a3e","revisions":{"f4cc1f712747497e11e0bbba27225b6804032a3e":{"kind":"REWORK","_number":1,"created":"2020-01-09 11:11:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/12/701712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/12/701712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/701712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/701712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/701712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/12/701712/1"}}},"commit":{"parents":[{"commit":"1a6430f8785b7f7d1f9b6fc7beb6694727c00a73","subject":"Merge \"update constraint for neutron-lib to new release 1.31.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a6430f8785b7f7d1f9b6fc7beb6694727c00a73"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 11:11:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 11:11:05.000000000","tz":0},"subject":"update constraint for python-octaviaclient to new release 2.0.0","message":"update constraint for python-octaviaclient to new release 2.0.0\n\nChange-Id: Id1de2c4ee83d2093713f28ee2984204c50d68d58\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta:release:Commit: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta:release:Change-Id: I5848635655db82775ad375ec986f8ffa4980ca0a\nmeta:release:Code-Review+1: Adam Harwell \u003cflux.adam@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4cc1f712747497e11e0bbba27225b6804032a3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4cc1f712747497e11e0bbba27225b6804032a3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
