)]}'
{"id":"openstack%2Frequirements~701787","triplet_id":"openstack%2Frequirements~stable%2Fstein~Ie4c1d93b7055cf2ef3a23150ea65453dd16a6718","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"Ie4c1d93b7055cf2ef3a23150ea65453dd16a6718","subject":"update constraint for oslo.messaging to new release 9.5.1","status":"MERGED","created":"2020-01-09 17:55:00.000000000","updated":"2020-01-10 00:16:06.000000000","submitted":"2020-01-10 00:16:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701787-1578615366816-0548d2f5","meta_rev_id":"226ed2aa892c3e59230305a3a224c18f030daa08","_number":701787,"virtual_id_number":701787,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-10 00:16:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-09 21:27:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-01-09 21:27:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-09 21:27:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-10 00:16:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8343572b3d527182bb757a4ad2f2b227b8d502f5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-09 17:55:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7daef0e6aa6ed333c7caf477e41a472f2cd5c9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 20:15:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/3cf1ad8c13fc4c8aa25613eafc297028 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d98605d6d50498ab700a191ecb1aa45 : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/379bc9f7089e42028036001dd880c975 : SUCCESS in 5m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7c0451991c8a4a189bc446979eb12b24 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25a92186816141bb931e8d65122f4ca4 : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/87bf8d8161fe476a8a357169ad5f0d0b : SUCCESS in 5m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c72adf1412b4542beed2e243161d8ce : SUCCESS in 5m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/18cb29aaf1924a43b4217004dd180e07 : SUCCESS in 8m 05s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/cfe32fb941ae45aa85eb6e0cd7d51b5a : SUCCESS in 13m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d82fc48e5ac2481b9acb8e96364e8022 : SUCCESS in 6m 37s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/cc28c5b1369e4eed8919990a6e9ce963 : SUCCESS in 9m 27s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/f3d376bf26b9451caf3d493d47af8152 : SUCCESS in 11m 03s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9763dc543c704e6fa65ef1f4d811b1a7 : SUCCESS in 5m 01s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/aef4af746a5349729ecc1a2ff88e9127 : SUCCESS in 7m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/eb5b91e69cbc4cffb233f4b42c486503 : SUCCESS in 13m 25s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/ca6488566bfb4942b9894256924e7e0c : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/de23f65e9d7b470a9205a0eb32ee507a : SUCCESS in 8m 30s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f34f6c318edb4de5ae6495ce2a067323 : SUCCESS in 20m 31s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/6873673e01b5462f9ff5fd1f89193c2a : SUCCESS in 33m 10s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f383b4a6c9d94ed7b657497247e0ac3c : SUCCESS in 25m 14s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/800ddf12e3b54bb99b77edbd836c6259 : SUCCESS in 24m 31s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ea3579a368ce46ef9ffaa9ab97e8392d : SUCCESS in 13m 16s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/bdac3e6e82a54c4683250e0770d87cfc : SUCCESS in 18m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8ec8c5d1bf54690b73dcd8d46da9452 : SUCCESS in 21m 22s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/ecbeebf9b6134738be7d6a300c13079a : SUCCESS in 5m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c7700ab9bf5e4597ba1f5b2709eca1d1 : SUCCESS in 8m 45s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/031fd0d7acae40049f8fc24136d02074 : SUCCESS in 10m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/229223954b454420b8491ecebe6b59fa : SUCCESS in 1h 15m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/37d2b7002d6840a89afaa6d693fbc1da : SUCCESS in 1h 23m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2d6f7003f9bb4c6ebfbcbbd3bc3b8d08 : SUCCESS in 51m 51s","accounts_in_message":[],"_revision_number":1},{"id":"97c0a6cef7180e10f16bea9c578769a7fbcd27c3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-09 21:27:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"500895b40db06471029251407717d1c985838e77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 21:29:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70b006b63f45b3bb2b133d475541c828b955b57e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 00:16:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3ded5eb7d76e24e8920e9006e1455624db57b441","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 00:16:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33d10107d5544b539b5c22bcd38143b3 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fffb00447a7f46ab8f75c9bc98a78aba : SUCCESS in 6m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8ab630c515d94589a281842678e476f6 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/642311327ca7474da7912dffa546611e : SUCCESS in 6m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a28e8cb7c1f48a9b2126f0fab13f8f7 : SUCCESS in 5m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ff4b2b0f7f14318952f0d51fe6d3e0e : SUCCESS in 5m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8697f8511f7e41c9bad2eaa075492a20 : SUCCESS in 9m 00s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/e9b5c76e3b974ce68fbfb40427a39ac4 : SUCCESS in 9m 41s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a04249ccd1ff4d979f9ea5dbacbe6a02 : SUCCESS in 10m 28s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/ccb9cea03e7a4b5c897d4feeb0ae592c : SUCCESS in 10m 00s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8daacafbb54f4e1c8ea52d3eabbba796 : SUCCESS in 7m 38s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/8f2663c2073e4d10a97a4031d86c763c : SUCCESS in 6m 20s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0f4505cb8f6243bbb2d2f90ceb49cbc3 : SUCCESS in 13m 11s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/48d1e4c447eb4a968202704f5614eea1 : SUCCESS in 12m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/27fca401b6f44ef1ba02397fb3394878 : SUCCESS in 9m 09s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/91dbcec498ea40a4987457da2ddf52f1 : SUCCESS in 24m 22s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/e156b4daa34e4e9f98d6b0eb96b1cd2d : SUCCESS in 31m 43s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/eaa64168bfda4f12833e0128556a992e : SUCCESS in 28m 04s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/6f24dcd62f6749a28eb543cbc9b2361e : SUCCESS in 23m 40s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1f690bd6c4fc4b45b3518b389c8e61d0 : SUCCESS in 15m 36s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/0475bd9c90a044898ef3e039421c24bc : SUCCESS in 16m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bbec3e574c274c869f9c6e76abff6f68 : SUCCESS in 18m 56s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/82592a4ca4f54372b0c6286e6dad7218 : SUCCESS in 7m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d948198d4fef4a56b185240339e7fc16 : SUCCESS in 11m 44s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/2a3ee9fd7e4441ac9615e4088f2fa794 : SUCCESS in 10m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c7eb9d52c48346aa8133edb23a73f302 : SUCCESS in 1h 09m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bcf4103900ae4eaf954d5c3f92a3a438 : SUCCESS in 1h 21m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c579ff30ba1d4de3877dcaac77eb338e : SUCCESS in 54m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"918654269d3d07f4b01ccf31b98a69496f0ce8ea","revisions":{"918654269d3d07f4b01ccf31b98a69496f0ce8ea":{"kind":"REWORK","_number":1,"created":"2020-01-09 17:55:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/701787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/701787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/701787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/701787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/701787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/701787/1"}}},"commit":{"parents":[{"commit":"0e55cd93abbf0b9972a2615478c367a24b58a9ae","subject":"Merge \"Add horizon to upper-constraints\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e55cd93abbf0b9972a2615478c367a24b58a9ae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 17:54:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 17:54:59.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 9.5.1","message":"update constraint for oslo.messaging to new release 9.5.1\n\nChange-Id: Ie4c1d93b7055cf2ef3a23150ea65453dd16a6718\nmeta:version: 9.5.1\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nmeta:release:Commit: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nmeta:release:Change-Id: Ib9a5e705d3a5164a3a859c3123543457c9331c99\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/918654269d3d07f4b01ccf31b98a69496f0ce8ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/918654269d3d07f4b01ccf31b98a69496f0ce8ea"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
