)]}'
{"id":"openstack%2Frequirements~701801","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I83393e38e9440b9cf0a17e2baef51f37e5a19700","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I83393e38e9440b9cf0a17e2baef51f37e5a19700","subject":"update constraint for oslo.config to new release 6.11.2","status":"MERGED","created":"2020-01-09 18:27:13.000000000","updated":"2020-01-10 01:15:55.000000000","submitted":"2020-01-10 01:15:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701801-1578618955969-d6071f12","meta_rev_id":"40ce57f1f879f1d0d5a5236641752b315586e8c0","_number":701801,"virtual_id_number":701801,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-10 01:15:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-09 21:27:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-01-09 21:27:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-09 21:27:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-10 01:15:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cd8b723cad3c54cc1d1c268574814e8a91c8a6b4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-09 18:27:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bde1a452055f10e9053e2c7dc449a385c7b0c3fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 21:21:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/dd0a3881e0274932a1d28a8f724e8c59 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8114e1a8d1442e6a457ac4c60fb38fe : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/78fd9b8d393d4d22a1cdf65550192fa4 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af0c19af652b4abb95b39653ebe4d048 : SUCCESS in 6m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7658746ed8fa43128006410bd0155ba2 : SUCCESS in 6m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e5f8d13302894947879c57ed937dfceb : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d6f0f264cdeb49d8aa6dd2f5970af1df : SUCCESS in 3m 55s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e2be718eea514345844643d26a95b25e : SUCCESS in 8m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8377dc3cc5bd48358108de8b2c7af519 : SUCCESS in 6m 39s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/59a530670fae4cd1b6d16a1bc374b906 : SUCCESS in 9m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7dcfbe64392b4955b111968207dca838 : SUCCESS in 14m 56s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/80c0448d5c944505b04d20ee69b1a1a7 : SUCCESS in 7m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5dd177fcd38d45d1b7e2f8202286905e : SUCCESS in 6m 06s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/490dca474b594ca29388ee40c41de4b7 : SUCCESS in 12m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4cd2a1585ce64b4f9af86efb14535f4f : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b905c7379eb74a048e60284d01036729 : SUCCESS in 6m 51s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f56e8fab768d44cb915de6cc6f16cfe3 : SUCCESS in 21m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/daded3cf3d9d40b6b09e393d25687091 : SUCCESS in 12m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c87418f34e424e8eb655ccaa957f76d4 : SUCCESS in 26m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b1e51ba7a53b42f5ae7de114237b16e3 : SUCCESS in 30m 25s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2a9c99f367b0416d8ecd2198c86e2a0e : SUCCESS in 14m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/965507499cf74f1093f114a2fd40cdfe : SUCCESS in 12m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e849e1ed84e34e8686965191caef447c : SUCCESS in 14m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/46baf6132bb841e89001313f25b944cf : SUCCESS in 18m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ce2eb1e3298a4b6b9c03f8d6d06af9c0 : SUCCESS in 6m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f3fc98c3e144450cb0c12149924736a7 : SUCCESS in 10m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ade60b70a1e34e53b3946e961592b214 : SUCCESS in 10m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f87343f1c9e6496e91bcfc4e4d34f6dc : SUCCESS in 1h 25m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8cbca01b5794416c92b3530b8ed6f8df : SUCCESS in 1h 48m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4bcf193e8a574dbaa1add49f3a0f0023 : SUCCESS in 1h 11m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1adee4aa5536f6f74409db0a5fb054c7f44e4206","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-09 21:27:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b6d1a2ef2fd5e39d8c2fff56d333285f394c2fb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 21:29:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db94f1221e39b021bb9461ac954e601a5604a1bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 01:15:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"11de6456f4481457ca8389561828d11f1c5d53f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 01:15:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3ece0ea0f754eccba7a2d967259cc4c : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/84241cfbc93b4d278330c98fa1e2ef0a : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da38e10961e44a6abbca7545b7b04792 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6bbc7680837e4319bdfbe24790425fd5 : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/429b9108eea449d8b9aa3709254a7780 : SUCCESS in 5m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ecbe735203db4e8597bbc70dc5f60c87 : SUCCESS in 5m 52s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5a33bd5c09404927b50f218576298895 : SUCCESS in 7m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a9700c5397b64009a6550c7432c2a6c7 : SUCCESS in 9m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5d89f092e60f4828b07ee1f53c1a82c1 : SUCCESS in 11m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/46d8a62876af4c63aba5a5aedc8fd096 : SUCCESS in 11m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/51bd9f27581f43339dba368ac9c192d4 : SUCCESS in 7m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0c68bff979574a74b749de1ff3252da8 : SUCCESS in 7m 38s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/eae68a0d292f44988c02d3b6c2bb6fa6 : SUCCESS in 12m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9a289a2a667043fa91f7dd453021c20c : SUCCESS in 10m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/97ffeb3d15154c1abb413e0530956032 : SUCCESS in 14m 13s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d84d62819d454d289473508c12ade502 : SUCCESS in 21m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/07b2baa63d6b441aa998c0105671ca1e : SUCCESS in 21m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b3d5ae91e2d342fb9c31250d4e9803e7 : SUCCESS in 22m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/66c2c2af7a5a48ca9603eddd516b9d78 : SUCCESS in 23m 32s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2ab05f465f2449caa46df6b25fa2cea4 : SUCCESS in 15m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ce4ccf95e9854ed7aae0073825d3be48 : SUCCESS in 15m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8d285f029d994edc8fbbc05602ee3d69 : SUCCESS in 15m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/210f781b7156424c93650917ca4775de : SUCCESS in 20m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a6ce369d47bf45d5bb79cba30e14a5dd : SUCCESS in 6m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bb5769fed8be450ebbc99b670d562f70 : SUCCESS in 17m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ed69eb44ec73482bbd3dc58d779ba7a9 : SUCCESS in 17m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3da3df6e6a5741e784cbc5c34d9cb2a6 : SUCCESS in 1h 09m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1f27fc9ac46c4716a213838a114f949f : SUCCESS in 1h 47m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ae85e459c0cf44cbac3bacaf5d8e9d80 : SUCCESS in 56m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ed9f1901a5ce0dd1915833297ada01dc3d982d9","revisions":{"6ed9f1901a5ce0dd1915833297ada01dc3d982d9":{"kind":"REWORK","_number":1,"created":"2020-01-09 18:27:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/701801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/701801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/701801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/701801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/701801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/701801/1"}}},"commit":{"parents":[{"commit":"a474e915974669621a86741f64c61be6abebd274","subject":"Merge \"update constraint for paunch to new release 5.3.1\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a474e915974669621a86741f64c61be6abebd274"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 18:27:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 18:27:11.000000000","tz":0},"subject":"update constraint for oslo.config to new release 6.11.2","message":"update constraint for oslo.config to new release 6.11.2\n\nChange-Id: I83393e38e9440b9cf0a17e2baef51f37e5a19700\nmeta:version: 6.11.2\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I2e365bb394d0b6e6a99d05ea4a039d3612a9e295\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ed9f1901a5ce0dd1915833297ada01dc3d982d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ed9f1901a5ce0dd1915833297ada01dc3d982d9"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
