)]}'
{"id":"openstack%2Frequirements~701802","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I39e3371ce613d2a5a1480b789d099d7772d66463","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I39e3371ce613d2a5a1480b789d099d7772d66463","subject":"update constraint for oslo.utils to new release 3.41.4","status":"MERGED","created":"2020-01-09 18:27:43.000000000","updated":"2020-01-10 01:15:59.000000000","submitted":"2020-01-10 01:15:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701802-1578618959377-af565c7f","meta_rev_id":"e7785790c02cf0b84a368e398cdf648145629773","_number":701802,"virtual_id_number":701802,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-10 01:15:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-09 21:29:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-01-09 21:29:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-09 21:29:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-10 01:15:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"75341364a0fa0f5f00a67c6f76515d29953cf295","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-09 18:27:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"008cbd8f394a772e9a2c13e6475133f723103f77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 21:29:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7106f95bb44a49ad9fa5706969fcbb23 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9909505128449fca2620909689bd84a : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e093a2ffca7c4ad2a162fdb4d1db0443 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02a3ee3625854eae8578c6c658a0e267 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7a8b166c6d544f48f602ba761cb23d1 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14d0785e90254f02b2de912085de1e8b : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b59e138a25b94379ac0de43001cf3e87 : SUCCESS in 4m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/770861d88ece4c3e9eabbd2de7abb277 : SUCCESS in 6m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/242b9e3a97e44218a2701275977aecea : SUCCESS in 7m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4da47bbb8c5a43bfaa0d4e39f08c4a89 : SUCCESS in 11m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/110314abdbba4d8c8351c42ec3c54f4c : SUCCESS in 9m 37s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/216cb4d7dac64c40b363ab8f9cbc02e6 : SUCCESS in 4m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8f29d28e5c1843bb9d9d62b1ebfe5d9e : SUCCESS in 5m 29s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ee4410d7b5c9479c9100c9202ab9a8cc : SUCCESS in 11m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3fc2034214504810aa895df6136df78f : SUCCESS in 9m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b40586af00794b47b2fbd73ad700e9e0 : SUCCESS in 9m 38s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/be86d551629547f091042b6419f66137 : SUCCESS in 24m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d88808df85c2415e883998b10934e534 : SUCCESS in 25m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/55a3dd6677b84cd583c6e2c1f9057cf4 : SUCCESS in 24m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4d6fc0d4508a4384b863ab995fe1cff4 : SUCCESS in 23m 41s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2652d38e70ad4f83821f9957c437ecbe : SUCCESS in 12m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1bcf269ae4b046e29ce5ad7a19d3c3a1 : SUCCESS in 13m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/72bc07ba45eb41668fb5ebec70940e54 : SUCCESS in 13m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/64c4b52fffbd40e88fd5394c603082bf : SUCCESS in 19m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e7441020787c4d6a8b0aa7e6dc09dcf3 : SUCCESS in 5m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f6104060f1b54cbfae07ad9379335645 : SUCCESS in 9m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/02ff207a1af941aea9adaf5f32dd6768 : SUCCESS in 9m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/abd1f032e78e4d49a7396a3ecabf1e58 : SUCCESS in 1h 15m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a6f90040e54244fb845552ab66479d52 : SUCCESS in 1h 32m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/25b3499459b9493eb5d2405f7b61cd7e : SUCCESS in 59m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d644dc6977d14052cc300e5be3de2de098e744cb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-09 21:29:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"276b6801beeb27c8402c20f6ce6ca2ae999f3efb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 21:30:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c8940e94d444395a122abd7a72c2611649c6b4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 01:15:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6d48b6988043d744499b908d9efad086e8a4cf14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 01:15:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/553a303f536b45dca23a406bf355bf35 : SUCCESS in 7m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c316e7b643942dd9874708d43ae5500 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f41356d65e6540ee8d6c67ae6102bfaf : SUCCESS in 6m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e7937f304404196bf515a229c813b0f : SUCCESS in 5m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/633e6b090a4d4b92b28f44b2e14cd3b5 : SUCCESS in 6m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/801be39e42834171a3fe8e86a8b6d81b : SUCCESS in 5m 14s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b2e4ac71f61b4b709d59f66d19b565c9 : SUCCESS in 9m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ef93bd90d1df444abdeecd2da1d71aad : SUCCESS in 11m 38s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/57dc73de84934bf19a2f34dcfcd4329f : SUCCESS in 11m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4f991c5a883f49bda40aa58964494654 : SUCCESS in 13m 09s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/729afe794f4b40cfba7986e66d7c7d25 : SUCCESS in 7m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3b3ca5212a0d4d28a87e4b82178262c2 : SUCCESS in 6m 49s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1fbace6593344303a0ba2c6d6c8bd551 : SUCCESS in 12m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7bc48009383e4e2cb3635827453a6c22 : SUCCESS in 12m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1d508132d0d24213814cc2da2f27f6e5 : SUCCESS in 9m 45s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/efaacbecab544aaab03e1d3daeee68d9 : SUCCESS in 16m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8bb8d7d2b06c4399b2ef10929cc3ec4d : SUCCESS in 14m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f100ab55e1624076b3dab5b3a3b45c19 : SUCCESS in 27m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8fc8fa7eafad4ab981ce967938416a4f : SUCCESS in 27m 02s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f7866a8149ab4961b141de6b9da6305b : SUCCESS in 16m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e88655a9846e40088e3df835d9c99b68 : SUCCESS in 15m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/388b768b357540bba85f5b94d6ab8798 : SUCCESS in 16m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/905b5f98378448b8a5963a207d742d75 : SUCCESS in 20m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/32dd4031954241cfb95252d850101772 : SUCCESS in 6m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fb260d09f9c2424ab62bdf3d210da864 : SUCCESS in 10m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9b9a1346ce0b4de5b5b5fcb14c5fb696 : SUCCESS in 10m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f800ffe31334bc598eeb931e6ca6797 : SUCCESS in 1h 10m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/50ca060f77924c83b7f117eee6ef560f : SUCCESS in 1h 35m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3856617753bc4a7f81d7fdf73157122a : SUCCESS in 58m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a2125b1f4711c375ac9c8b9e484a5499dd7fb88","revisions":{"3a2125b1f4711c375ac9c8b9e484a5499dd7fb88":{"kind":"REWORK","_number":1,"created":"2020-01-09 18:27:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/02/701802/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/701802/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/701802/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/701802/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/701802/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/701802/1"}}},"commit":{"parents":[{"commit":"a474e915974669621a86741f64c61be6abebd274","subject":"Merge \"update constraint for paunch to new release 5.3.1\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a474e915974669621a86741f64c61be6abebd274"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 18:27:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 18:27:41.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 3.41.4","message":"update constraint for oslo.utils to new release 3.41.4\n\nChange-Id: I39e3371ce613d2a5a1480b789d099d7772d66463\nmeta:version: 3.41.4\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I2e365bb394d0b6e6a99d05ea4a039d3612a9e295\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a2125b1f4711c375ac9c8b9e484a5499dd7fb88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a2125b1f4711c375ac9c8b9e484a5499dd7fb88"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
