)]}'
{"id":"openstack%2Frequirements~701813","triplet_id":"openstack%2Frequirements~stable%2Fstein~I039e53e0788c912d996c7e6ec6781049c8674440","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I039e53e0788c912d996c7e6ec6781049c8674440","subject":"update constraint for oslo.privsep to new release 1.32.2","status":"MERGED","created":"2020-01-09 19:06:00.000000000","updated":"2020-01-10 06:02:48.000000000","submitted":"2020-01-10 06:02:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701813-1578636168701-17819b5d","meta_rev_id":"1f75dba1810cc9043e05f8135f4962180102e45d","_number":701813,"virtual_id_number":701813,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-10 06:02:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-09 22:00:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-01-09 22:00:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-09 22:00:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-10 06:02:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f159c7a8ec3e4eb8dc5024dbea2d789f1a4f8fab","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-09 19:06:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76db51c2b62ab73ba7e33aa22e66d9c1707f446c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 21:38:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/3c86a667e5ba40dbab6b4c47447ed199 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70240f56982f46509d8eefb7cf1763e4 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2ef61061aa104d7199125f7b3687b70e : SUCCESS in 4m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dc3afac111c34671a38aeaf59c8439c0 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e7c13048a474c5cb1d34f3138c6cf8b : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dfeeda1aaa2a40edafc709e3ea7125aa : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8e09c808f1c3449ca4faac2819526b47 : SUCCESS in 4m 19s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/20ea23874c6f4214baa785f1843bbf69 : SUCCESS in 7m 14s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/b37be6a966ec47e48c9708f421c76a11 : SUCCESS in 9m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/028e85644ba844f4b0cb51173c22163b : SUCCESS in 7m 13s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3e36b3fcd236411bb14c7dcef22b6196 : SUCCESS in 10m 01s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/c04997646d534d49a0a1c868ce81e64e : SUCCESS in 10m 17s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/248c8761f9114c95b7aae5e40e3921de : SUCCESS in 5m 19s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/d18a91bfe20f45aab72021f3ea398745 : SUCCESS in 5m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/591d8b07efa44c5e8a3a43babc549047 : SUCCESS in 10m 02s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/f725c202b87b481dbdc048b4fb45d986 : SUCCESS in 9m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0bb0893d645b42bb82760bfef951a753 : SUCCESS in 8m 34s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7957853ab5c342c59256ae55c2480bdc : SUCCESS in 25m 03s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/9824c571ca1c47a8bfd6cd4e777af39c : SUCCESS in 28m 45s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/110d104dee014e3fb41c8a7f1ecb8b5e : SUCCESS in 21m 39s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/660c0c3fe2f84aae8bb7262db00198d5 : SUCCESS in 23m 35s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/73fc570a090142df89d2e6abe2abfac4 : SUCCESS in 15m 18s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/05b2982f24ef48c6817a2dcf9eb3ba45 : SUCCESS in 14m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/089959fec1bf47ed9ea87decc86dff1f : SUCCESS in 19m 15s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/23d0cebb0b494f96b747702e76a1ab85 : SUCCESS in 5m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b8b8a39a361b4009b342f341041a930f : SUCCESS in 10m 31s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/68c4539198784d9d86df5ecacf5ff64a : SUCCESS in 9m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/df1b87b422a044baa78c56493e371d8d : SUCCESS in 1h 23m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4057ae5eb5744d98b7155828574f1f20 : SUCCESS in 1h 22m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d8b0ca9773274e14ab44c8cb6d70e929 : SUCCESS in 1h 03m 26s","accounts_in_message":[],"_revision_number":1},{"id":"263cba1819a67c00f6d2225fa9c56c1145c9708c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-09 22:00:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"78e58aabd83dbb4e85c44077aec0dfbad2fe99e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 22:00:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b64874eb46c14b24a27256f9b91c812d18b5328","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 01:16:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/625c6b0330cb4de9bc0566db0d92fb5a : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9282b373595f4f1a8b9cec362752410f : SUCCESS in 5m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/53e1237940704bfda192d82a0a259589 : POST_FAILURE in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c35d5be26df340ce8dbfd8f8d23848d3 : SUCCESS in 5m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/56e9c07320834f2fb8b0de0d6ba4399f : SUCCESS in 5m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c55cc1f893294a6099244d5144060dfa : SUCCESS in 6m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8fb1feeeb22e48a79811fd7887d99376 : SUCCESS in 9m 06s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/5b160805ff2944229c9bf32db2b03aa5 : SUCCESS in 7m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/81d6950c3cf446918bcc9c1cc238058b : SUCCESS in 10m 11s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/ddf0c70e82014a7ba5404d186ebd736e : SUCCESS in 11m 00s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7c8ba7b8f9b14db2aec6a54b23e1f579 : SUCCESS in 8m 39s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/82dd88d1a571429a97ceebf35dc795f2 : SUCCESS in 7m 19s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b23bcf2acb374750819061a89c79fb14 : SUCCESS in 12m 20s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/e8c6e7aa02634d39bfc87bd919345a12 : SUCCESS in 12m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/91434fb496f04a828c3e98e6c48b2906 : SUCCESS in 9m 31s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c37c3260905c4fc98867560ef83475db : SUCCESS in 23m 18s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/82c54092602c47249c7639115218b6b6 : SUCCESS in 24m 45s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/89a78d2297904bc995a12ce559496d4c : SUCCESS in 21m 00s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/a830cb250a9f4201b4cd9508726d8611 : SUCCESS in 20m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ed4bcf93ce2943c0a479f0091e41ead4 : SUCCESS in 13m 34s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/71cbd76702314794a5e9f4ec85a676c3 : SUCCESS in 14m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dedd5548fea34b2e9452f7af77011c56 : SUCCESS in 18m 20s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/b8a2c27558504b359c287d6d1d124263 : SUCCESS in 7m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e80fee2f0bc04048a26b6cdfafab8cf2 : SUCCESS in 12m 41s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/297ea3c670ff44ef9207c07301864e89 : SUCCESS in 11m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b17f54af0a1844baa4fab4475be77f83 : SUCCESS in 1h 06m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/43dba45feda246b09234a6b3c2371da7 : SUCCESS in 1h 14m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c1c9d955c7054070b65263887d2a460d : SUCCESS in 52m 13s","accounts_in_message":[],"_revision_number":1},{"id":"374100cc901498dcc215fa2de1b4fe46ee6c821f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-10 02:46:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9008a4b6b336c9d41bc084908d09242172bc7414","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 04:20:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/884ccf2bc6c444b0a7a4212aa52f2da4 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f71196ebff60417d85af046f38757d71 : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c7015ebc4e2447dcbb46fd311e60a47f : SUCCESS in 4m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6ab4db2745e4479b93963ea6742e428c : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d42fbd65d710422e848bac4cbf571d16 : SUCCESS in 6m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/19d4e8073d2c40a6bf1e8717eafba1c2 : SUCCESS in 7m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/67a2c7702c68416a8e83bc850a329b98 : SUCCESS in 5m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fdfd1d50bca64ea084b15cd60415f88e : SUCCESS in 7m 06s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/97718bf197044aa2a31f4944f552a6df : SUCCESS in 9m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/474c1c56d9914afd87f419b9d89f372c : SUCCESS in 7m 40s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d001fb6d5cb34b8a965647b2062f1c41 : SUCCESS in 12m 08s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/cdda3b2b9856487887b23a1f315a04fa : SUCCESS in 8m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f2bd93f8632143bfaf63b0a43e8626a2 : SUCCESS in 7m 53s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/088532b94c8445e6afb635c4f1766a98 : SUCCESS in 4m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e5c679ab15bc4fcfb0f085e31859211b : SUCCESS in 10m 50s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/6e4023e371474068a6398b44821b0fa6 : SUCCESS in 8m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3f19134cc34747e989ef57ee5e9c3c3f : SUCCESS in 8m 48s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a96acdad3eb0445da7acd5a2fe2f45a0 : SUCCESS in 20m 03s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/aa243089aa3a43e899723615c8b74ea6 : SUCCESS in 22m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/be9af63824ec4217b2790e71af5a8846 : SUCCESS in 22m 54s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/2d8f1e2731134e44baca542579dd222a : SUCCESS in 21m 57s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5f773d21b7264964bd2384fddcdeae40 : SUCCESS in 13m 20s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/512b664db5854af6be4fbe6604ea57fd : SUCCESS in 12m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fd2726e7338648578199d3efc828fcdf : SUCCESS in 18m 55s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/5bfc91887bcb41bbbd3b0849a241d0cf : SUCCESS in 5m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/de16f85134274eb795c52db54efb4b41 : SUCCESS in 9m 28s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/646414d78eca4c71994af10d0cb783fa : SUCCESS in 8m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b0b1c21b895346989079de7eca3446a7 : SUCCESS in 1h 06m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/18dffdb1b8934c10bb9396683e909141 : SUCCESS in 1h 26m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7a2f42554a024bc6af667ec46fadfafb : SUCCESS in 54m 25s","accounts_in_message":[],"_revision_number":1},{"id":"06124dad1957647328986f2471eed2a672736f6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 04:20:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"997b07ac84305c0ab164aa25ff989cb3c8a36a3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 06:02:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0bba02f8d9a618af156d48dec71401761b0aa14f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 06:02:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/003a1f3d4f0a490db2a8dfd3f2c753de : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e21e61966d884185b370d653db25a048 : SUCCESS in 3m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a6f1f33c6c3145ca813f57247a97258e : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58e80c12469c4769833bd199bacf060d : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a0d29e9423ad4993b50fc0a727c166af : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c8b6c444c6204024a9754e70ec96a188 : SUCCESS in 3m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1e8983592f5046899a02dc5564402667 : SUCCESS in 6m 28s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/e9da11ee996b49b3befc89b5c77b41ef : SUCCESS in 6m 58s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f255ae50f3df476fb0da61e27dad7e9a : SUCCESS in 9m 17s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/d6c7d7ade71b40ae8ab8c3a67d237a93 : SUCCESS in 11m 19s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ebdaf444ddda4b03a5b00f740d284b83 : SUCCESS in 4m 24s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/642028643e75408a882635449290b424 : SUCCESS in 4m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e913bb14e08e46bfacb8c5e80bccdc8c : SUCCESS in 10m 13s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/8e1e428a53954f1da284901da6df4862 : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7733678d4d9f45ceaff8c67fa6844d46 : SUCCESS in 6m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8eaa2dd556a549bc8d82e4b16f7b64fb : SUCCESS in 20m 40s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/39692a4baaf64e1d91502f3711620dfa : SUCCESS in 27m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9d1f0991a1ea4ec79fcf1f12fca0ddcf : SUCCESS in 23m 26s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/d9180eb678d94a95b44c32a600b286fb : SUCCESS in 24m 51s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/678cf2ca578c427f900f9ffe3bf89009 : SUCCESS in 12m 02s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/e990fb0cea814d8595b7f6bdaa850b1d : SUCCESS in 15m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/20093d3130b047f5bdc9b34c969c228c : SUCCESS in 18m 00s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/e1c96f0612054da1b8781c6b7100ad4a : SUCCESS in 5m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/af1e1a832d18471bab7d33e562f0b7ce : SUCCESS in 8m 05s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/6b1b510bff2e4820afb44f86d1fdc1dd : SUCCESS in 8m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2912bd3c69a64a03a8492c15801afcab : SUCCESS in 1h 05m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5dd4215ad08f4e9bbcdbe5a1229140b1 : SUCCESS in 1h 35m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cf6c264aaa664600ba334f1874165a16 : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7e1d9f5a5162c109fd53f61ebad4d263be54fe0","revisions":{"d7e1d9f5a5162c109fd53f61ebad4d263be54fe0":{"kind":"REWORK","_number":1,"created":"2020-01-09 19:06:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/13/701813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/13/701813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/701813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/701813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/13/701813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/13/701813/1"}}},"commit":{"parents":[{"commit":"0e55cd93abbf0b9972a2615478c367a24b58a9ae","subject":"Merge \"Add horizon to upper-constraints\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e55cd93abbf0b9972a2615478c367a24b58a9ae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 19:05:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 19:05:58.000000000","tz":0},"subject":"update constraint for oslo.privsep to new release 1.32.2","message":"update constraint for oslo.privsep to new release 1.32.2\n\nChange-Id: I039e53e0788c912d996c7e6ec6781049c8674440\nmeta:version: 1.32.2\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I6b120321a39cf0ec7261d179df5c09ec505c9bd6\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d7e1d9f5a5162c109fd53f61ebad4d263be54fe0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d7e1d9f5a5162c109fd53f61ebad4d263be54fe0"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
