)]}'
{"id":"openstack%2Frequirements~701815","triplet_id":"openstack%2Frequirements~stable%2Fstein~I845e36ba6675ed88f38abef28e10c66fd9165315","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I845e36ba6675ed88f38abef28e10c66fd9165315","subject":"update constraint for oslo.rootwrap to new release 5.15.3","status":"MERGED","created":"2020-01-09 19:08:47.000000000","updated":"2020-01-10 02:49:01.000000000","submitted":"2020-01-10 02:49:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701815-1578624541280-f3637170","meta_rev_id":"e557ed3b37adbad64f8759838b1317d23d4be8d8","_number":701815,"virtual_id_number":701815,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-10 02:49:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-10 00:19:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-01-10 00:19:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-10 00:19:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-10 02:49:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"947ac870da0fa29b6bc89141583a1d8fedb03992","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-09 19:08:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c636d7ce9c543e7c7f3decad264c5744c0d418d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 22:20:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8428b6807f57436fbff78695af27670b : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49e5c28218f747a0b815e6fd12c461c1 : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6eb1e8c8413444b6825b35c2c4b4ba25 : SUCCESS in 3m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/43657b27e81541e2b3a9b42a67572604 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3038acc590e4424b2dc89d525c7c01b : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3b5cb91b4d464393b16eef4f89dffbb0 : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bc26781e06ab4228b3dfdbdd19b0c0cb : SUCCESS in 3m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e8eac5390f9140218551991fde9b5583 : SUCCESS in 7m 36s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/e465a79bb5c14f4da1c7320ce44800b3 : SUCCESS in 9m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fa7d3b0051a24e12aa1fdb199dc05f82 : SUCCESS in 7m 36s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3a18ce90482e401caace59c492fc0280 : SUCCESS in 11m 13s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/1a4df58e0fe5409094cb4d577d2a0a59 : SUCCESS in 10m 35s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/40628475cfd04d898a1ede0e2a4275e1 : SUCCESS in 5m 02s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/8f6536fd1ad442cab4cfe0ec2986f20b : SUCCESS in 5m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1aee18a0a6714427b80093e40143aec5 : SUCCESS in 12m 43s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/215189b33f65480da9e07e80e7a76d01 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/47f65cb38049438faf7f7e57540fbc8d : SUCCESS in 7m 49s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6b2406907975434fbf0e1df418bca6fe : SUCCESS in 22m 00s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/ffc9ac94a7d54abd926443bc0b6a3e74 : SUCCESS in 27m 30s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/bb3a0727a74e4db792c87d2a2ec1a453 : SUCCESS in 23m 45s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/8e85d1db47a64df99f22f13afcbc283f : SUCCESS in 32m 36s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b848ac073cf14c87b5a9906da6fd5644 : SUCCESS in 18m 18s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/02d75d19552d4c03b3e9f00e6d050fd5 : SUCCESS in 17m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a1b915f4198c405e8c0f0062c3b8e322 : SUCCESS in 17m 07s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/f2ae9a37b00d49e4adaf99d4cbfc23e9 : SUCCESS in 10m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9a1c65cafab2478fb267225b93a31761 : SUCCESS in 13m 56s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/ec01840f321f43d4859cb647ddba7a22 : SUCCESS in 12m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/28b9abe9331f44cca8d13563a8b37072 : SUCCESS in 1h 53m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3235f1f6d2a44d7492b9f740a4e41907 : SUCCESS in 1h 58m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b66dfd462c8a48edaeebfa82ce2634f2 : SUCCESS in 1h 12m 43s","accounts_in_message":[],"_revision_number":1},{"id":"80bf9abf52fba837fa19f2ae14df932642c76276","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-10 00:19:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cf4ffa6d6a40e015b7dfc023080cad37702a2373","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 00:19:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8838e47c02e51419d86c9836d36ace125a368cf2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 02:49:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5db28cd8019818d80fe159ff7e670e484136ba93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 02:49:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfabc75f34af43c78870324adedc165c : SUCCESS in 3m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5892b4d3c65c420db16f8ed3f4cd594f : SUCCESS in 4m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/67c15da40afc4844ac7244e8b677f387 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/810f00c854e740cfa146aec6e6e7a789 : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e110eb29dbd4e9a8ef47bf91bb8841f : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bc5dcf4dd400422594af27b1b43aa411 : SUCCESS in 3m 22s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1c0fc42dcd61434abf8bfefbec83abab : SUCCESS in 7m 29s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/6710bb7d25894fe09219ae94f5df2030 : SUCCESS in 8m 07s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e1b7333587234676b59856c6251b1f88 : SUCCESS in 8m 54s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/6242310311454e3f9cf81bfb191bed84 : SUCCESS in 9m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1e83e32c99a04071921f6c148386bc40 : SUCCESS in 5m 21s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/8cd88421ba2847eca9ac97ab48808b14 : SUCCESS in 4m 33s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/418642c3149f4acba52a136fb686e1b4 : SUCCESS in 11m 38s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/044559eab3344f95a877e0c905de4374 : SUCCESS in 9m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86bc50bbd3b2414cac504731c2128e06 : SUCCESS in 9m 02s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/41e7ab574ffa4606992233ccc03baa40 : SUCCESS in 22m 29s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/5dc32c47a3b54e0299cfcb33d7adcb29 : SUCCESS in 31m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1dd8805f73a54762aa442b1578923676 : SUCCESS in 21m 28s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/410ec32f48b1429abea57cee25523a94 : SUCCESS in 25m 51s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a26f66b3d5ed425e9bb59b29e821439d : SUCCESS in 11m 59s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/26cc201ae0244a6bb886eb51aa81ecd1 : SUCCESS in 13m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d5ccbd16030b483e8b06caaf1a2ac490 : SUCCESS in 19m 33s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/03b0c7c7563a46e7a6dfc0e45ceedce1 : SUCCESS in 5m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e8e396e0e9b34ce29011b9f51a5d0e80 : SUCCESS in 10m 14s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/3991e92a6621466c9b1bcc1b0f4129a4 : SUCCESS in 9m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/48909de01e3d4997a58462e45fa98b35 : SUCCESS in 1h 07m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c6a5811f7a0d4064a504c832136ffa1b : SUCCESS in 1h 33m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3d2f24682c604a5c8d1ae8fa166b1f31 : SUCCESS in 50m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6d3eac264e44785c8a23f2d8dd89e9457c99ecc6","revisions":{"6d3eac264e44785c8a23f2d8dd89e9457c99ecc6":{"kind":"REWORK","_number":1,"created":"2020-01-09 19:08:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/701815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/701815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/701815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/701815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/701815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/701815/1"}}},"commit":{"parents":[{"commit":"0e55cd93abbf0b9972a2615478c367a24b58a9ae","subject":"Merge \"Add horizon to upper-constraints\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e55cd93abbf0b9972a2615478c367a24b58a9ae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 19:08:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 19:08:45.000000000","tz":0},"subject":"update constraint for oslo.rootwrap to new release 5.15.3","message":"update constraint for oslo.rootwrap to new release 5.15.3\n\nChange-Id: I845e36ba6675ed88f38abef28e10c66fd9165315\nmeta:version: 5.15.3\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I6b120321a39cf0ec7261d179df5c09ec505c9bd6\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d3eac264e44785c8a23f2d8dd89e9457c99ecc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d3eac264e44785c8a23f2d8dd89e9457c99ecc6"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
