)]}'
{"id":"openstack%2Frequirements~701816","triplet_id":"openstack%2Frequirements~stable%2Fstein~I0853cdc803da65921909268baaee5bd0436ce678","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I0853cdc803da65921909268baaee5bd0436ce678","subject":"update constraint for oslo.config to new release 6.8.2","status":"MERGED","created":"2020-01-09 19:11:18.000000000","updated":"2020-01-10 01:16:02.000000000","submitted":"2020-01-10 01:16:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701816-1578618962478-69f755c7","meta_rev_id":"8d3c0b925214e93b4f0d705373edb45207f592dc","_number":701816,"virtual_id_number":701816,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-10 01:16:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-09 22:00:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-01-09 22:00:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-09 22:00:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-10 01:16:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2b3e23b9ed81704012d1fea88e12240c99b43058","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-09 19:11:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc2dd83bd060769bf809e2194fe1ed025bba1131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 21:50:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/307a35869fd74c1099765cf0665f6e77 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/159f105e4a594e9f85a946ea98d8fae4 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b2cab36eadac45bb9ff9b2db525221c4 : SUCCESS in 4m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c4fbef896d3142eda760ae2c42952cf2 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f5debcd95b648f89d6d894b2a7efb63 : SUCCESS in 5m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8d72bc4ead084764b7e1b3c31e48eb24 : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e7aa94cf37aa4d86be54c9b38ed2fcc7 : SUCCESS in 4m 59s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4906aba62f814b029820af652268e5bc : SUCCESS in 7m 57s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/ffb3446b7f82480e9a39ed32ebd67b20 : SUCCESS in 8m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4c15c6b582fa4e71837b2b952b550cac : SUCCESS in 7m 46s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3a0ab38021374410b0b038e3ab4e4f22 : SUCCESS in 12m 41s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/b9c55a3cca644a3e9eb988d3e09877b4 : SUCCESS in 15m 00s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/611db4743222454885189f5919931b15 : SUCCESS in 5m 57s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/fb5b9dbe25234b629cc4ca21bfd87123 : SUCCESS in 5m 52s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b918b7d9ebc2428c9c4b793f09512caf : SUCCESS in 10m 36s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/c3f5db2128ff4d1b8e443bcb565d099c : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f979982e83db4161b0c4ea95a78f50f8 : SUCCESS in 6m 46s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6f520624267a4b2d907416fbabb7c60a : SUCCESS in 23m 19s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/f514d763b0ba4db1a357c421041293aa : SUCCESS in 26m 29s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ea39f436587d4db88bb8e7a144a172ad : SUCCESS in 23m 12s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/b3e50a6d4a7a44f68a443dc18de06ae6 : SUCCESS in 24m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/da6d2537a68a45c9b9e263402ec16468 : SUCCESS in 20m 06s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/6a8063230e6d425fa85e333134d2a138 : SUCCESS in 13m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/067be71784be4b528c9bba2563024dd1 : SUCCESS in 20m 33s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/df38f2299a1b484b85bc8381093e0054 : SUCCESS in 5m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ee452b438b754973bf71c9ae893d0159 : SUCCESS in 11m 09s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/c7b92678fa3548c885e4110cc0df49c5 : SUCCESS in 10m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3c7855691ad46ed84990a8ae6cbd0f0 : SUCCESS in 1h 13m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2634ff103fce4eb5abe1c5e37cd5e30d : SUCCESS in 1h 23m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/493c766a414342f1b038e755e0cbce35 : SUCCESS in 56m 04s","accounts_in_message":[],"_revision_number":1},{"id":"af4288e440ab23512a0fd88b4773f63335d0c819","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-09 22:00:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7851e96957d2b3bc284156255ed90b0838532555","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 22:00:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27d0170351c00418053385a315135ddd5175ad3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 01:16:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6686c87328dbfa7f441a235c707b08836f74a16e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 01:16:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fb4c2d79eed4f0a8779a77ec37a82f5 : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d2836e8c22c4d06b0a4f8cd96c78297 : SUCCESS in 5m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a90c6098956945a6af0768b610269d36 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d2de3a332814014a384c113bfb0311c : SUCCESS in 5m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e63e6cf88bf04d8bb3a386c99889fcb7 : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1b48d8c15a244e6f89932c2e8274b5e9 : SUCCESS in 5m 18s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4aaea9be1c6948d796cdd8f1068a6f42 : SUCCESS in 7m 13s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/7956a13b15d24337aa498ee4a3c13a17 : SUCCESS in 8m 30s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/30a7b2bdf19345c781c7879affbfb543 : SUCCESS in 9m 56s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/087eb6e9aed641d4825db02d4bc71b2a : SUCCESS in 11m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/878ad11b506f423e85c83cb0c9b98318 : SUCCESS in 6m 04s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/e02ec77719474cdca5ebae677a025d90 : SUCCESS in 6m 19s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/cb37e9a17e6b418086c35972018aa91d : SUCCESS in 12m 27s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/c0f05401246e4bd3a9d0dd55a2188a35 : SUCCESS in 11m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fcd9989088844db99b37f171f0166693 : SUCCESS in 8m 46s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/13cb29628b954fc89191b62948d8812d : SUCCESS in 25m 15s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/07058bd901ae46d382eed439e981937b : SUCCESS in 29m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/fa000abf4d5c41e789110613709418e1 : SUCCESS in 20m 08s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/9e14e378b629435086f1dcf08913878c : SUCCESS in 24m 28s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/33cc13921f9948519cf6270ab945d031 : SUCCESS in 13m 38s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/4aeac53a2fd940f39d4406693b4b110a : SUCCESS in 13m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/722822b1cb0144a19293adb0e2678653 : SUCCESS in 20m 46s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/f96d29b32f81476abc6f821e2f084189 : SUCCESS in 6m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3bf4065fc4ab4135b8723c7984a9d12b : SUCCESS in 7m 55s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/6940662480b2442094ef18ec5f452e35 : SUCCESS in 9m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ed965e065c14e639ff7ae20f8bb62a4 : SUCCESS in 1h 11m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b779d780c035469a952be49d055d9a51 : SUCCESS in 1h 23m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/685c34265e96492da302c43703e80e52 : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d348c2b41e99cdc59b7ad0c6acc8382729f22b34","revisions":{"d348c2b41e99cdc59b7ad0c6acc8382729f22b34":{"kind":"REWORK","_number":1,"created":"2020-01-09 19:11:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/701816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/701816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/701816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/701816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/701816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/701816/1"}}},"commit":{"parents":[{"commit":"0e55cd93abbf0b9972a2615478c367a24b58a9ae","subject":"Merge \"Add horizon to upper-constraints\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e55cd93abbf0b9972a2615478c367a24b58a9ae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 19:11:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 19:11:15.000000000","tz":0},"subject":"update constraint for oslo.config to new release 6.8.2","message":"update constraint for oslo.config to new release 6.8.2\n\nChange-Id: I0853cdc803da65921909268baaee5bd0436ce678\nmeta:version: 6.8.2\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I6b120321a39cf0ec7261d179df5c09ec505c9bd6\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d348c2b41e99cdc59b7ad0c6acc8382729f22b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d348c2b41e99cdc59b7ad0c6acc8382729f22b34"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
