)]}'
{"id":"openstack%2Frequirements~701831","triplet_id":"openstack%2Frequirements~stable%2Frocky~Ib234738e38320158d1b5a8a633b9d36889927497","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"Ib234738e38320158d1b5a8a633b9d36889927497","subject":"update constraint for oslo.config to new release 6.4.2","status":"MERGED","created":"2020-01-09 20:39:33.000000000","updated":"2020-01-10 03:10:08.000000000","submitted":"2020-01-10 03:10:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701831-1578625808372-af565c7f","meta_rev_id":"3b4dd808b10f35f6217950dd4eafd1ed42f6a799","_number":701831,"virtual_id_number":701831,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-10 03:10:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-10 00:19:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-01-10 00:19:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-10 00:19:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-10 03:10:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"32f3aa53a1862e1237b9660f1c0c8116adcb7f93","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-09 20:39:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e271603a1d11509211d065a1d7f53499d034b1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 22:33:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/127c98776b004953bb53f8ff06417805 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e66990be04845d284e038815f32299c : SUCCESS in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0ecbb5fa25fb45a39cac5922af7d938b : SUCCESS in 7m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/22753262f94a41a4a0e7fa78a09bdac0 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/234aaaded2654c0abecd973148f3aab9 : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fbfe54169c9848e980813b3c92a43322 : SUCCESS in 3m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a2f39d0bfa5c41ed91c3050533d4671c : SUCCESS in 7m 51s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/bf12687867784a0c9c79e27970a2e43f : SUCCESS in 8m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0af2ae7e1c754899be34563c00cd7d9b : FAILURE in 8m 03s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/abc69bd216c74f749204e4106a86ef28 : SUCCESS in 10m 53s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/9a6db10b06384f1588e2b88fbf882c32 : SUCCESS in 9m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/124a9a3556314adda5ed87820088781e : SUCCESS in 6m 31s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/31dbea26645540738a5f348f6f6e8d1d : SUCCESS in 4m 48s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/765ebd92661241dbb5327cb7639b623c : SUCCESS in 10m 10s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/201121bb63f044c1a3a88af22150e19a : SUCCESS in 9m 56s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6c55b4274daa4ce0a2ef21047148add5 : SUCCESS in 17m 20s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/fd0a0743fac3464da0451bb8c53f9af3 : SUCCESS in 13m 26s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0433816b968a4ab0a9d004e09049727f : SUCCESS in 29m 15s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/03bb576b4c634be4a561611238e9988d : SUCCESS in 29m 36s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0256fdcf70334cd795f0edcc359bdf77 : SUCCESS in 13m 34s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/0e11de5fc2e64bbf8e4a41b335617136 : SUCCESS in 16m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a42ad2b7b6c48bea373c51ed206de99 : SUCCESS in 16m 54s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/bf306fd9c7be4f62914abc5b6bc7f864 : SUCCESS in 5m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9189d66d31934c6a823f0b50bb99acff : SUCCESS in 9m 28s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/652e1252644c48b8bcb0e8306c7f20f4 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9aa71e7717a4c8ebc272a280c6115eb : SUCCESS in 4m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/126a21dcd3e847469d1672d175fd5f33 : SUCCESS in 1h 08m 57s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/d2a59835669f448c84af07123132db27 : SUCCESS in 1h 36m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b07b951396e64f77a5c47c2843cf4d8d : SUCCESS in 56m 50s","accounts_in_message":[],"_revision_number":1},{"id":"cdbe8890b733bbe12fce9fb6283c54f9ea18f104","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-10 00:19:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4d317abb7673e6620d3590ba9ae6be1b8489cef6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 00:19:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc0769de8ae3bec64209047e8689fd318148adf2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 03:10:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7dc0aebeda66fb851a5469d74335fab12f7b8f52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 03:10:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5af39bda5c48443fa9e4301387091222 : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7528f0a5b7148ca859591e5db1f115a : SUCCESS in 4m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1b6b10600a794790b02daa2a3fee748e : SUCCESS in 4m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7b445df49d7d4305a81aae0eadec4fd4 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1e4ed7c8cc80479d87f50cf9a25ef7dc : SUCCESS in 3m 59s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e22b25d60d204cfdbb1c9b61cc8da108 : SUCCESS in 7m 20s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/6a390cc4414945c1b68f9953246628bd : SUCCESS in 7m 33s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/577d07ad9b4d4afbbf5631e32a5e4345 : SUCCESS in 9m 44s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/e3662237a675436798473951ddec77dc : SUCCESS in 9m 21s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/aa8f4047729f465084eeab65e37fcd02 : SUCCESS in 4m 58s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/aee4dfc584f1459592c3028cec577ebe : SUCCESS in 5m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2f029d95c05349248650bd0e0cbe50f1 : SUCCESS in 11m 02s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/0bf628731a834f4a9b6d8336c433828d : SUCCESS in 9m 38s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/3614cb183a9e4cb197b33c59c82700b1 : SUCCESS in 12m 54s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/182ba71f3a7745e2a77d9a89c1040456 : SUCCESS in 15m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/561e5da0957c4d67b4fa561466cd9770 : SUCCESS in 24m 04s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/8f0711ec1c6549de80e71f31c1304a45 : SUCCESS in 27m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/181d1104f62b4d78bb8490b5dc1d10ae : SUCCESS in 12m 39s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/77de64e0c503416681a185683a762211 : SUCCESS in 12m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2bad65900c714ade82da0a509aa4083f : SUCCESS in 17m 33s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/162607a14f5640a8a3c76ffa25b18034 : SUCCESS in 5m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/de2a0cdcabae4940bab2d3e8e0e8f9dc : SUCCESS in 8m 28s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/a4d8e5d3161749c586a01f7a85bbde84 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e3eeaf10e1740d29e314554be0aac77 : SUCCESS in 3m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/48fc58a99067469191701a18306a9591 : SUCCESS in 1h 05m 30s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/2f1f6178484947938c73058c23be16f4 : SUCCESS in 1h 26m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c1f4fc1b789c41c291f332508c49212c : SUCCESS in 55m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ef1a11877b9a56fdbd5fb88113c5c657699d1a8","revisions":{"8ef1a11877b9a56fdbd5fb88113c5c657699d1a8":{"kind":"REWORK","_number":1,"created":"2020-01-09 20:39:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/701831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/701831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/701831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/701831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/701831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/701831/1"}}},"commit":{"parents":[{"commit":"a361a52ebd3f25090cb8c42364f94a3007cdb723","subject":"Merge \"Add horizon to upper-constraints\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a361a52ebd3f25090cb8c42364f94a3007cdb723"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 20:39:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 20:39:31.000000000","tz":0},"subject":"update constraint for oslo.config to new release 6.4.2","message":"update constraint for oslo.config to new release 6.4.2\n\nChange-Id: Ib234738e38320158d1b5a8a633b9d36889927497\nmeta:version: 6.4.2\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I8206ac3edaa15f82a240335e8af321be7419ea5e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ef1a11877b9a56fdbd5fb88113c5c657699d1a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ef1a11877b9a56fdbd5fb88113c5c657699d1a8"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
