)]}'
{"id":"openstack%2Frequirements~701832","triplet_id":"openstack%2Frequirements~stable%2Frocky~Ibdbfcc8e76d9c6f12650a4e9c7677f2677cd14e1","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"Ibdbfcc8e76d9c6f12650a4e9c7677f2677cd14e1","subject":"update constraint for oslo.utils to new release 3.36.5","status":"MERGED","created":"2020-01-09 20:39:43.000000000","updated":"2020-01-10 03:10:07.000000000","submitted":"2020-01-10 03:10:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701832-1578625807378-7a9322a4","meta_rev_id":"3b5b26ce96f87f3f38fd54ba80fe52d52ea09bdb","_number":701832,"virtual_id_number":701832,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-10 03:10:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-10 00:19:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-01-10 00:19:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-10 00:19:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-10 03:10:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"24aa4697e23344e7923c538d83cd120056ce19b9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-09 20:39:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa88fa2166226867531bb04d499e1282ba2af8e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 23:01:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/979751b657ba4a269b835668edc263b7 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7a2a9aa511b44659d1e2452dde061ea : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a58057306db545b28805849245df55c5 : SUCCESS in 5m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0a8a76dd32964c3d9cbe376daa7163b3 : SUCCESS in 6m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/71cd264243964c38b81983d08fa23e78 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/254c552982814a0faec39e4b322ca330 : SUCCESS in 4m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1222a73ca3524fe5adb429b3e9199f27 : SUCCESS in 8m 05s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/e722b9b504e34e04a92df75d4fbb5799 : SUCCESS in 8m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2cdbc11b902449a99d57684a85ac2edc : FAILURE in 6m 19s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/efa4f935547a43cb96dbdddf0d660d65 : SUCCESS in 9m 25s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/f3e8d8d5c840471580b4800e6ca4c98c : SUCCESS in 10m 24s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2240dc37409044d6a15344f0358da768 : SUCCESS in 6m 32s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/839ade6eecb54317beeae10a34fad33e : SUCCESS in 6m 01s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3185693a28924f6685628de0f60204e2 : SUCCESS in 12m 17s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/760a66192d95460089e701f85d19ae35 : SUCCESS in 10m 18s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9fedb509d91246ca8acdbef9f2429af5 : SUCCESS in 21m 43s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/5d431cf9d6634efa9a7f1e80a3da2a33 : SUCCESS in 20m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6542270edf3b42db9657195cfc0098f3 : SUCCESS in 29m 11s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/62c66c9803a94d9e8f0187839c3a4bd1 : SUCCESS in 27m 28s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e623771d843847f0930381f9214bd388 : SUCCESS in 16m 02s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/e5b58a60cc5d44d9b27f0a370418e95a : SUCCESS in 15m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/14c76a0eaa4e4fcf9aa0baf889f520de : SUCCESS in 16m 56s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/8345128753c74953b41bf750cd8d9bba : SUCCESS in 6m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e7840cfe4fd047d7a9d14a448026b4fe : SUCCESS in 9m 24s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/c21581ee7dd84fbd955764be1a6ac1e9 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0e3914396b445a8b20e1e192eed2c61 : SUCCESS in 6m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/56e434f28fcd4ea29f5c482254e48e4d : SUCCESS in 1h 04m 55s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/25f2a494dc2949adb18f1729b853bc39 : SUCCESS in 1h 30m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c55483f06e2b46b3af856f658762d3c7 : SUCCESS in 51m 58s","accounts_in_message":[],"_revision_number":1},{"id":"bb46c332d3845237132073ec951c6710f955e6c2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-10 00:19:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3d040f1af120dc3387a9b75a8af55412889bb91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 00:19:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"600d5725a88f88a517d212590f3149fc98863b22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 03:10:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9745189560518f026900dc9db3ff5f42f4842f22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 03:10:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6553e70f0dd4cccb41cfcfdb5824ccf : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5205fe441bb04c78a0c4943a8b9d363c : SUCCESS in 3m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fb880b83ee064f9c89f8ef2c0e499973 : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/512b38d932e248fb9a88a5c0d3304442 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0c750ae2a3384b969ad1e6820dbd6507 : SUCCESS in 3m 17s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5e15d979fa0646f58420806e8f216eb7 : SUCCESS in 7m 35s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/388904bb31cb4e25a768976b0d2bd9c0 : SUCCESS in 7m 39s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/36f692803ff644809ec5980ab31bb894 : SUCCESS in 8m 59s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/cfcf3963dda849a6823b4b0e0333592a : SUCCESS in 8m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/515175b21a5b4e3d8d9c9c48a4b1bb03 : SUCCESS in 4m 47s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/3d40dea9e825424eafc2f0b2329b2d49 : SUCCESS in 5m 04s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1bd1076e5812431cb84f360f144f2b3c : SUCCESS in 9m 37s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/2a128fb5ca55474f92f4207488db4d9f : SUCCESS in 9m 25s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c39ab085ba064d3181a94dfe6efad889 : SUCCESS in 20m 47s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/5bbb71e4cd6f4fcca79a3d38f47fba5a : SUCCESS in 36m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2655457eadb64d98b58f9105684e88e6 : SUCCESS in 25m 06s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/08f5722de83047ba87599e173241ff87 : SUCCESS in 23m 04s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/eba0ef3bd7974b5c93b0f34870b9d311 : SUCCESS in 13m 34s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/839b45992c4d4fd093596184e56587cc : SUCCESS in 14m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/437b0510102249bbb34ca14f9b198a1b : SUCCESS in 18m 15s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/f9bb38db8b2f466ea50f9285bc6d0fe0 : SUCCESS in 5m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5adf4eb4bd6c4100b3e13881935078e7 : SUCCESS in 9m 13s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/505143d05d314fcfaf7f8ef8c14eed4e : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ad50ca1f52c4c88a3d351e3f43221e6 : SUCCESS in 3m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a3ff321d80da4ce3b4326a397654859a : SUCCESS in 1h 01m 49s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/ca72edc0d7724851b47c31112644c205 : SUCCESS in 1h 27m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5a28cad8d6944e2a91c19c22234f35ac : SUCCESS in 50m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ca47ddc57b373602eae2dbae5a6931f471c5f29","revisions":{"1ca47ddc57b373602eae2dbae5a6931f471c5f29":{"kind":"REWORK","_number":1,"created":"2020-01-09 20:39:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/701832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/701832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/701832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/701832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/701832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/701832/1"}}},"commit":{"parents":[{"commit":"a361a52ebd3f25090cb8c42364f94a3007cdb723","subject":"Merge \"Add horizon to upper-constraints\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a361a52ebd3f25090cb8c42364f94a3007cdb723"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 20:39:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-09 20:39:42.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 3.36.5","message":"update constraint for oslo.utils to new release 3.36.5\n\nChange-Id: Ibdbfcc8e76d9c6f12650a4e9c7677f2677cd14e1\nmeta:version: 3.36.5\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I8206ac3edaa15f82a240335e8af321be7419ea5e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ca47ddc57b373602eae2dbae5a6931f471c5f29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ca47ddc57b373602eae2dbae5a6931f471c5f29"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
