)]}'
{"id":"openstack%2Frequirements~702277","triplet_id":"openstack%2Frequirements~master~I63bfe896d0cdc84626db39febde7ad6577984174","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I63bfe896d0cdc84626db39febde7ad6577984174","subject":"update constraint for oslo.vmware to new release 2.35.0","status":"MERGED","created":"2020-01-13 18:18:19.000000000","updated":"2020-01-14 20:15:07.000000000","submitted":"2020-01-14 20:13:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702277-1579032815801-f3637170","meta_rev_id":"b24502455e01a6b506027ab1aec9f74c424a39d2","_number":702277,"virtual_id_number":702277,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-01-13 20:20:26.000000000","_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-14 20:13:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-01-14 02:32:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-14 16:30:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-14 16:30:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-13 20:20:26.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-01-14 02:32:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-14 16:30:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-14 20:13:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2af9ef6c5ed3f9dd410d89674ad349b215378ee0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-13 18:18:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"050869e3b54c59ffd4b93d46b6174fe7135bf4a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 20:10:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9652563558a4a578140e782f79e492a : SUCCESS in 4m 15s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/64e29f7dace44de69649e516c2778838 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fa5e4e890a04cd890d0f77fdffd146b : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53c7e3c4a43a47179a60a12ccfee99dc : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/79d94a006e7c4cb597054197dfadbf44 : SUCCESS in 4m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/99a926574f174c10a6840225760d9034 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e4c213f8880646debc271f7b2a283fdb : SUCCESS in 3m 50s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c97835734a374562a28e05fca280e8d5 : SUCCESS in 7m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4a5079b1bd914388889db4e28315c523 : SUCCESS in 8m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/92f5cecd450540a59bc202ab8f39e064 : SUCCESS in 9m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6d1cfd244cab4936a56d98c0356c8a75 : SUCCESS in 6m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dc74e6ff9dbb4e4ca9631ae34e015448 : SUCCESS in 11m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/164d44ec88d44abb9da4cf770cc2781b : SUCCESS in 6m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/06107bc0028444899fe825f4bd480a26 : SUCCESS in 15m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ed9a7c136d094787904439b79b9446a8 : SUCCESS in 29m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/944c73a05355442c95a07a3ee10de3ea : SUCCESS in 13m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/308ca0f17d004d9ea6703a46922400d6 : SUCCESS in 15m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/919160c97aed42069bdd4510bcbbb433 : SUCCESS in 17m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2d265eed9d7a4981988819af9acb6ad9 : SUCCESS in 6m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d184c1b49bb84aba84e14d245a7a68be : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5fcf52448bb143edbe3c911bb2cd8c2f : SUCCESS in 6m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/06b36bcb1f904acb986fc289babd20dc : SUCCESS in 9m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/95fc100dee62461ba71b352797b2538f : SUCCESS in 8m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cce0232c99304ad68ff4ecc765a986ca : SUCCESS in 1h 00m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0676ecc04a4949b7b3668b0c701b84f7 : SUCCESS in 1h 12m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7c660e5a20bd45be8e793365df42ca6a : FAILURE in 1h 39m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/55e097e7ebcf4097ae0c02237825b6ce : SUCCESS in 1h 25m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0e2d7d82385c34cfa03a48e470e0d9464df700ff","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-01-13 20:20:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5b692604172ed1044c5438ebfabadb6ad0b0b521","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 22:02:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca98f9fa2b5043628f85aa64414d032f : SUCCESS in 4m 09s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/36dec43ed0264c86a31d2ea730acf056 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a31e6bfc1f7c4b4c8273ca7e446ee430 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/742a5b297a904976a5e789a51b3579ed : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba24ea1f3ea446f3aeacbd2ae1ddda92 : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1bb1578aa54945fc8bb5eed2c0e9d2ea : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9613e7573f4442f88ce793ccb037df44 : SUCCESS in 6m 18s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ffff14a048e7479bb0cecec1d7988521 : SUCCESS in 8m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/832752010cdf495a9ca42810ff65c841 : SUCCESS in 7m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/258b39765c014c23a357ddc41dac36d5 : SUCCESS in 8m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9001744f2a914bc08d724bb9c49f2c38 : SUCCESS in 12m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a9307f23cf514686980314b83be7a3df : SUCCESS in 16m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cba14f0b424f450687ca9e1b98354222 : SUCCESS in 7m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/306da87e70ba40f3bc7b57e213d502c9 : SUCCESS in 20m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9c1a4136e4eb4de6b3e3aa58a85e6963 : SUCCESS in 23m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a5857ebcc71247afb510ae08ce1e7244 : SUCCESS in 11m 55s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/41099786adde430982aa8583a32828fc : SUCCESS in 13m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82f379a38eef475e86d0f92792c7e8b9 : SUCCESS in 20m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/af03efb98b7b43a0a14421ad61039c74 : SUCCESS in 4m 31s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a51a33cb46f64db48cafc8dbd18f455c : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c9f8c2cb717c4e218fae11dfe73f37cf : SUCCESS in 11m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/eb0fcf549f534e469342f585d50e1495 : SUCCESS in 9m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/50dcb57f01354f6ab8be7916efd5cafd : SUCCESS in 10m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80e7ddfef1ca42369d976ed6727e4aba : SUCCESS in 1h 03m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9f3cf5fbfbe84caea4d70d1fb33a307e : SUCCESS in 1h 06m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bf5909d1745d4af9a55b06aa746f55e2 : SUCCESS in 1h 36m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/235aa5456649455abea4bd74f55ea8e8 : SUCCESS in 1h 08m 52s","accounts_in_message":[],"_revision_number":1},{"id":"86eecfcaeacfbb82935240b9a5ece3f42554c6fe","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-14 02:32:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"79d3b53bf1977c74a0f068ff30cd4bf2ff3fef81","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-14 16:30:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f379176d163313bd8f95a68546a183d7663d87a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 16:31:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"015724fc2df3810001ed5882e7686a796d8eb050","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 20:13:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eddd1b1d34f34fe2a04acedd92070e22 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6bc1eb1baf54e788d047b641ec4093c : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30c6ddd4b1914ea3b2a815832a0c337a : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41652ea73fcb4f98a58f14fdf6d63f81 : SUCCESS in 5m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e82fd75081848108e211f26b21394ea : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bd4b1642bb35452f9536d2a6404793b4 : SUCCESS in 4m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b7f1f105e1ed468f889c53e0b9b064d5 : SUCCESS in 9m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d2b802e01956419b90b683760b1ca7f5 : SUCCESS in 13m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/68d4244c3b6e4018b9c1a7f172fb8033 : SUCCESS in 13m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4ca6fa4cb1aa44f48fe83ac35278fcce : SUCCESS in 7m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1feae2317e8c4690934c631ac40ee90e : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b4c638d863684b9c88239a9e5806cdd8 : SUCCESS in 8m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c6168613ab0042ed8897050cdd478ed2 : SUCCESS in 21m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bda6069905fd4d90b893b1bcba0a82ea : SUCCESS in 25m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ed5d7a2546204915a55aa647ad65b0a4 : SUCCESS in 14m 59s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bd57b61013c9428bb0849e64a6a2a187 : SUCCESS in 15m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1dd3636697e84cc3b5d636b4a292f4ee : SUCCESS in 21m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/659921215e1c4193b3c2669ceb220e06 : SUCCESS in 4m 43s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9b0ba0c4bd414db78fbba2fc8476d227 : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/25bac8d960684c978892d0bfff236365 : SUCCESS in 7m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a3ef0a534b38401d996f99f059d8a99a : SUCCESS in 11m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f48ad449c44e481799265bfe02f4286c : SUCCESS in 11m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27abfb81025c4d20bbf150c03b0c76fc : SUCCESS in 51m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1923812a2ff64facb19738bf199539bc : SUCCESS in 1h 18m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6dbb5b404a6a4735b2380414c1b3ea05 : SUCCESS in 1h 44m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/aebb8aba9665420aacdc163100e40a5c : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":1},{"id":"f356e46f8e044aaf71c4205544a42288962ec582","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 20:13:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1cd2377f18cdf7d185f8139156573f1a1e30e2b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 20:15:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b78ac9524d84f0d972f6408111e361b : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"42762396c85a97e128af4ec919e424a1492ae586","revisions":{"42762396c85a97e128af4ec919e424a1492ae586":{"kind":"REWORK","_number":1,"created":"2020-01-13 18:18:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/702277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/702277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/702277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/702277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/702277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/702277/1"}}},"commit":{"parents":[{"commit":"c55121348430c2fa33cf2285d96a8f9836ce9fec","subject":"Merge \"update constraint for os-client-config to new release 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c55121348430c2fa33cf2285d96a8f9836ce9fec"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-13 18:18:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-13 18:18:17.000000000","tz":0},"subject":"update constraint for oslo.vmware to new release 2.35.0","message":"update constraint for oslo.vmware to new release 2.35.0\n\nChange-Id: I63bfe896d0cdc84626db39febde7ad6577984174\nmeta:version: 2.35.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: Id64363a328208ca74703b15235a09c570da74c32\nmeta:release:Code-Review+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42762396c85a97e128af4ec919e424a1492ae586"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42762396c85a97e128af4ec919e424a1492ae586"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
