)]}'
{"id":"openstack%2Frequirements~702319","triplet_id":"openstack%2Frequirements~stable%2Fstein~Ifb805954ff36420004d552a690046d8ddf3bed5c","project":"openstack/requirements","branch":"stable/stein","topic":"constrain_python_memcached","hashtags":[],"change_id":"Ifb805954ff36420004d552a690046d8ddf3bed5c","subject":"demarcate python-memcached requirements","status":"MERGED","created":"2020-01-14 01:20:13.000000000","updated":"2020-01-21 19:37:32.000000000","submitted":"2020-01-21 19:37:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702319-1579635452493-cc57de1a","meta_rev_id":"172d640fc8cb19c2496312d6190faa7308b11b1e","_number":702319,"virtual_id_number":702319,"owner":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"date":"2020-01-21 16:37:08.000000000","_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-21 19:37:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-01-16 15:37:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-21 16:39:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2020-01-17 10:35:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-21 16:39:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-01-14 01:20:41.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-14 01:20:41.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2020-01-16 15:37:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-17 10:35:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-01-21 16:39:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-21 19:37:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73165714f6f76b0e91043bcb7b5fff085befe3d8","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-01-14 01:20:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21202d02472b26795ee3fa7c8827191ac419fa77","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-14 02:01:40.000000000","message":"Patch Set 1:\n\nWe should start with at least train since python 2 was still supported there.","accounts_in_message":[],"_revision_number":1},{"id":"1e4d001e3e6a96c148a6cc02ed11d7369f7109d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 02:44:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/72156d6651c74c95be6be56f4f8e96cb : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40b254112acb46e1928df860b66df75b : SUCCESS in 3m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b56a78f59999483c81b5b16933dd2901 : SUCCESS in 3m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3c7373999c6346f2a46fc15f3a04ec3c : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/203192f324af45249b33231a9199f704 : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/12e8e12fe933483686e7cf0f219b9c4b : SUCCESS in 3m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4aa685db2f534f08b15b112ca3f9dbe2 : SUCCESS in 1h 02m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5d8e6d751c9e49e6843412429292d57a : SUCCESS in 1h 19m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/28533b973c164c0e92a28d7f23ad7000 : SUCCESS in 56m 14s","accounts_in_message":[],"_revision_number":1},{"id":"16be956f4acaa1044a8b979c6c5e9ec53a7c618e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-14 03:43:08.000000000","message":"Patch Set 1:\n\nagreed, start at the most recently active branch that\u0027s applicable","accounts_in_message":[],"_revision_number":1},{"id":"6d209a7c7babafacccb22f82e957584b9153a76f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-14 03:43:33.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"88e330f0adfa73d380ef4d591930638c1a277688","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-14 15:29:13.000000000","message":"Topic changed from constraint_python_memcached to constrain_python_memcached","accounts_in_message":[],"_revision_number":1},{"id":"57d9aefbd03c31f4fa665def6c00d91ef816b652","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-16 15:37:17.000000000","message":"Patch Set 1: Code-Review+2\n\nThis has merged to stable/train now, so I think we\u0027re good to move ahead with this one.","accounts_in_message":[],"_revision_number":1},{"id":"842956910b7cda39e92fdcc1ba9fa3a0a0667cfb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-17 10:32:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThis patch don\u0027t need to be tagged like \"cherry-picked\"?","accounts_in_message":[],"_revision_number":1},{"id":"c50b7d946885b14ec2c7656b7d170648a1d17c83","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-17 10:35:07.000000000","message":"Patch Set 1: Code-Review+1\n\nremove -1 like asked on https://review.opendev.org/#/c/702462/1","accounts_in_message":[],"_revision_number":1},{"id":"f45d252926da12868bac709aac3acc361e036d4f","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-01-21 16:37:08.000000000","message":"Patch Set 1:\n\nMatthew, can you please take a look again?","accounts_in_message":[],"_revision_number":1},{"id":"b178215b133acc0e7ee75bba66cb9ff1f52fa1d5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-21 16:39:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2fd5f2417ac7ffc2be1ffe7be37fd4e42e5b7029","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 16:40:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85c3f08a29b1f3d3fcafda6ca2074f0978a4906a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 19:37:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"172d640fc8cb19c2496312d6190faa7308b11b1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 19:37:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/834a26c4af4c4bc995af5ebe365eec65 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/48d69ea245fd46d49e8297dd271e22bb : SUCCESS in 6m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3858b792082741b9936bf3e84212ad9f : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6116ba7bffbb4dfa9ebf5aadedc5d069 : SUCCESS in 6m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9667417580847afbe42cc0a41dfe389 : SUCCESS in 6m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d75780cba368499687c6bea3ccb3af7c : SUCCESS in 1h 04m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/03dc2f5025834b8dbc67559ecbcb87b1 : SUCCESS in 1h 31m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a5fd0f5647764a8dab53590862aa2351 : SUCCESS in 1h 10m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9e0f30f0fea60f3a020a4e5b27147dc9dc5de66","revisions":{"d9e0f30f0fea60f3a020a4e5b27147dc9dc5de66":{"kind":"REWORK","_number":1,"created":"2020-01-14 01:20:13.000000000","uploader":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"ref":"refs/changes/19/702319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/702319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/702319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/702319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/702319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/702319/1"}}},"commit":{"parents":[{"commit":"3e8716c54525ca50d4ac1d7abc6d6563e8d5ba77","subject":"Merge \"update constraint for oslo.privsep to new release 1.32.2\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e8716c54525ca50d4ac1d7abc6d6563e8d5ba77"}]}],"author":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2020-01-14 01:08:11.000000000","tz":-480},"committer":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2020-01-14 01:20:13.000000000","tz":-480},"subject":"demarcate python-memcached requirements","message":"demarcate python-memcached requirements\n\nTo support Python3, we need python-memcached 1.58 or above. However,\nprojects such as keystonemiddleware[1] have python-memcached 1.56\nin its lower-constraints.txt. For stable/stein, Python3 is now the\ndefult[2]. However, we generally avoid changing lower constraints in\nthe stable branches so we don\u0027t inadvertently break the downstream\nconsumers. Therefore, we need to separate python-memcached requirement\nfor the respective python versions.\n\n[1] https://github.com/openstack/keystonemiddleware/blob/stable/stein/lower-constraints.txt#L61\n[2] https://governance.openstack.org/tc/goals/selected/stein/python3-first.html\n\nChange-Id: Ifb805954ff36420004d552a690046d8ddf3bed5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d9e0f30f0fea60f3a020a4e5b27147dc9dc5de66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d9e0f30f0fea60f3a020a4e5b27147dc9dc5de66"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
