)]}'
{"id":"openstack%2Frequirements~702463","triplet_id":"openstack%2Frequirements~stable%2Fpike~Ifb805954ff36420004d552a690046d8ddf3bed5c","project":"openstack/requirements","branch":"stable/pike","topic":"constrain_python_memcached_stable_pike","hashtags":[],"change_id":"Ifb805954ff36420004d552a690046d8ddf3bed5c","subject":"demarcate python-memcached requirements","status":"MERGED","created":"2020-01-14 15:28:43.000000000","updated":"2020-02-16 11:38:34.000000000","submitted":"2020-02-16 11:38:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702463-1581853114194-b9790512","meta_rev_id":"b974a9336d40e7332092327c967de86b6269ab34","_number":702463,"virtual_id_number":702463,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-16 11:38:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-25 21:46:09.000000000","_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-16 06:29:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-16 06:29:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-24 09:54:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-01-25 21:46:09.000000000","updated_by":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"reviewer":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"state":"REVIEWER"},{"updated":"2020-02-16 06:29:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-02-16 11:38:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e88c52f6403981041d638ba27d46b5d5c638f700","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-14 15:28:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fee8a4ea480b02d2b1e00abd2bf40fb22fbe668","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-01-14 16:37:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"80d8e116d8e510651c20b13123e4cf8710b437de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 02:20:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/315626f1bb5e49e6997001f4d00b6953 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/527232678095487d962288ebe1716269 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b17356d9b4504c6591fad48d682dde70 : SUCCESS in 4m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/696a492843ce482e8bb399e1854ae1a9 : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d9ee9ebfa41f445eaf3a9d597b1be9b9 : SUCCESS in 5m 39s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/c8fd4f74d8da46df809af2c819aee4bc : FAILURE in 39m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a1369aab187943e2951236f7cf48fe47 : SUCCESS in 1h 10m 06s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/b278abdde79d460c8950e50bc7ae70db : SUCCESS in 1h 58m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cdbe9b07cd2541b5aaee779282dbc2a0 : SUCCESS in 1h 05m 41s","accounts_in_message":[],"_revision_number":1},{"id":"15812c0589ac4adb7fa92441a6a14af8e4503ade","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-15 04:39:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"88d09a2594e159c737dad8091562fa4a5ac7728c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 06:11:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/390dc5ea60d84c42912eccdcfd8671f4 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07af6f201d1b4ec2aaa5944342c93d2a : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4152a2ce6fa940fe849ec7d7c4607a71 : SUCCESS in 4m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5815fafcf33c4625b5e6ba8f754bb60d : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/792a9e8931634e9f8a6fe51a75adaaea : SUCCESS in 4m 10s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/e5bb86c5970e4e8c8ca0f36aa623624c : FAILURE in 28m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bebe3d99f55c4e3ca02e617faee22977 : SUCCESS in 54m 38s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/1e86fd97cbf24efda87c99e3061bd760 : SUCCESS in 1h 28m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c7903cd794444bad91774f360aaa2711 : SUCCESS in 53m 10s","accounts_in_message":[],"_revision_number":1},{"id":"68df8ccd9044a183250ac2b9066fb3a8828b8794","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-17 10:31:54.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThis patch don\u0027t need to be tagged like \"cherry-picked\"?","accounts_in_message":[],"_revision_number":1},{"id":"9d2baabab5d69e8bace6892ee6d89c1541e34786","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-18 06:48:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6492624eb746880434b4e50d396f9a1042d84337","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 08:13:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/bb9e620079b149868028a1d987813623 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b565a577d2404d0fa8994afb5cd889e1 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb0f242fe3e349b2b6df361168e0a905 : SUCCESS in 4m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/249fb4d9475e4f2b92dedb9c01f57e98 : SUCCESS in 6m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3de5bd408ed64b6fb0d94df1e7724037 : SUCCESS in 5m 47s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/7d17e9a5d1f849b8949ae406fc35a5a5 : FAILURE in 34m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b6276bbbe125451faa53381ce308d3c2 : SUCCESS in 1h 00m 53s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/25b00a725ae240feb6898f60481201a5 : FAILURE in 1h 19m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ff38625eaf32448888a219fae1e4301a : SUCCESS in 52m 35s","accounts_in_message":[],"_revision_number":2},{"id":"c4d54d4732a4750e291ac99433ada15e8558202f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-21 22:27:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8099f50dda792f0d5e84105030d9acf1efb10904","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 00:51:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/80c766e734ce4ba79f41237e978389d5 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/711669b663a1491aa595a09316dc01c7 : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/857a3a0c7df0466d86fe98a6acaa6814 : SUCCESS in 4m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8a94a450910d43eb8a5e14b8db29178b : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/101617a4f83945468fbe4b3f721bc946 : SUCCESS in 4m 47s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/20d0c6aa23934a46b5648014dd68d748 : FAILURE in 31m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/385e0cf07dfb4979bcf40f0ea51aa5b8 : SUCCESS in 1h 00m 20s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/facfc26580f74c1baca470cd61ab269a : SUCCESS in 1h 51m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/30cb90b9201b4c6693cad826213c3329 : SUCCESS in 1h 28m 41s","accounts_in_message":[],"_revision_number":2},{"id":"c163a3ed1abeb8b6e18fc556cbff28076ea8e025","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-01-22 16:39:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"494e67593ea7e0ee838e91d0401e5d14d236402b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 19:45:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/3922bdc4ee874633b144b6d6b58bd8ec : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0bc527412e4475f9d3b0ef04856096f : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9941c045cb3c4130836c4f26c2e32547 : SUCCESS in 6m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/050f8a098f9e429ba6c93674edfbbe0b : SUCCESS in 8m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a04f1074866b439d900bca976b6ff77c : SUCCESS in 6m 35s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0e106e7f6fa741578f634293c7b8110c : FAILURE in 33m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/66445b0b1b1043ae84025183cc1eab05 : SUCCESS in 1h 05m 35s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/156cd9c6edc24ab5af2690c0dbda96db : SUCCESS in 1h 31m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3356a4280d8b409a8e8aa67ddd6bb5d0 : SUCCESS in 57m 50s","accounts_in_message":[],"_revision_number":2},{"id":"d0ff631839b47df6c020f8523ebdec6e3e590433","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-01-23 01:49:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e315534eda2f52b1b412d985a224f1f010aec29c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 03:18:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ef15680021084553ac54a113d61ed2f0 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16f51b26bd284bfcb3153758d3eb4c23 : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9ac0c222aa96456288b25a2e553e5212 : SUCCESS in 3m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/39e1f494f7e44213ac7932bf5e2296b9 : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae96a2b2151945fcb90e52a8fc13e5f2 : SUCCESS in 3m 55s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0161a55693644faa8968b3ce5eb1e32c : SUCCESS in 54m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/42413612adce4f22b3921a5b7245c87e : FAILURE in 58m 56s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/72872a3583684f108cff9863aab9ba67 : SUCCESS in 1h 23m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6651c969b61241179e854cd597b212cb : SUCCESS in 52m 45s","accounts_in_message":[],"_revision_number":3},{"id":"3b56b7e158646725eb0251b7ce677a5221584ac2","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-01-23 15:43:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"894e74e3a0d0cda97c77a25e8d2923b15b7bb243","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 18:29:58.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/55a6d4c9812e4fde8a716d0d67e4e066 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe4e0b9c7d13410b9bb6d467b5b803d9 : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7c1f63f802f4e92a6735d6e63d8c6ca : SUCCESS in 5m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6389c3f6d5524f729c7a646a878ba495 : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0368d9fe1588439993fa09a4144eca08 : SUCCESS in 4m 51s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/60efc93eb5e649db83fcb3be8a61f16b : FAILURE in 28m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f6cbe9916c7944ef8e9c5b3efbabacd3 : SUCCESS in 1h 08m 51s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/e91aa824b1784222a22998becfc25e13 : SUCCESS in 1h 26m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f80d79ec56b3466c8f1fc3d7e4a6dacb : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":3},{"id":"c5ebc7272d75af0e700c58fee56a89d031f4345e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-24 09:54:52.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"57a0fcbc18f4e0ec2e90893cd578583816622c40","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-01-24 23:11:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7c66da1c051fd2328cdd4bb8d33bf3941af42e1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 00:34:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/926b3ff32f6e43bfb39b7a679d29801f : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebe92e558bda41318abe92976df8835a : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d85249a645449e3ba7745d048cb3ea8 : SUCCESS in 4m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cb31cb62dd224dac88e7612d2f42481c : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b64e78088dc2406da2963c5970aaf87c : FAILURE in 5m 02s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/d5f37172febd4e1faf174d6c0a107ab5 : FAILURE in 20m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa6a9fd9359142c39d0ca94d573c94a9 : FAILURE in 1h 01m 24s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/d83d3efd59e34e4c94881979014dbdbf : FAILURE in 19m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/84c81d12458e4b59852b9f3c70201a29 : FAILURE in 45m 52s","accounts_in_message":[],"_revision_number":4},{"id":"d844c0a54875aaa023a0536f7b32c29ebe6dc208","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-01-25 01:07:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"036d525e800710cf7ba8134433441cb655d022bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 02:41:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ae8864e2b43f4d88a06d787d252d7661 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a1569f824c940619657c4a4790ef593 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00ed3cd27f354f99bc184d5a5fefdf8b : SUCCESS in 4m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7ff69e06ab4e46cdadc43ca6a017cb9c : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/058e23fea10f4dc0b0dbf642d06d72a6 : SUCCESS in 4m 12s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/e00de63372224699a667c462589d5b73 : SUCCESS in 55m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b69d5b56b561429c9bfb979f1b15f988 : SUCCESS in 1h 04m 03s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/f0c25c0795e64b57a9511e71cf33d16c : FAILURE in 1h 28m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9c61f4cf709c4982a3f69b1a86da8e17 : SUCCESS in 1h 08m 24s","accounts_in_message":[],"_revision_number":5},{"id":"062254de5db3525962128ca101b560cfa6fc6f40","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-01-25 21:46:09.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"469ff9ad918e0d6f04af13dbd28812a3e534f5d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-25 23:20:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8fdda633359841599031370778dc7783 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6966761c56974e99b7970a70bf6e34e9 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd3ee3d050ed45e79d9076f981924f09 : SUCCESS in 4m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/800caa0c7502461e9be0805d7a9d0bd3 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/47bdc41b2d034708a16d5193b24031e3 : SUCCESS in 3m 39s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a850b19be5c441799d63717241037e9a : SUCCESS in 52m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d2c8a67876c84467b288634adfab0bed : SUCCESS in 1h 00m 32s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/05f87157472b4466a8fa7ad436e1880f : SUCCESS in 1h 28m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8683944b1b704398ac5e89ac82f9fadb : SUCCESS in 54m 16s","accounts_in_message":[],"_revision_number":5},{"id":"1b7145cd92e6819e136b5fb18b6cbf876150af1d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-25 23:22:23.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"87186b9a95cd5871a0d5fa56077b61490c99fc10","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-12 19:10:12.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"959831d2e1044f2f23185401beb085e42d24ca93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 19:10:33.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5fe3f7dd5cd717cb93ef6d19ea09bb7e67f0ae05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 21:58:44.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f392b249124470382cf47446f14a507 : SUCCESS in 7m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/77afc34606ee4984b4fe4c81c5840ca1 : SUCCESS in 10m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a96dacb299244ffda79296fd6004bb99 : SUCCESS in 10m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a1d134e60dd47ee82296d7911bb21a8 : SUCCESS in 9m 40s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/3bec5d808e4b4e9892d1ad7d0830443f : FAILURE in 13m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/40a4129634b140a3ac4880db2e0dd87c : FAILURE in 11m 42s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/dfc763ffb2f7429dbe3c6a61b8150a88 : FAILURE in 11m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7d75704db45641e88c89d7e3849b4055 : FAILURE in 10m 44s","accounts_in_message":[],"_revision_number":5},{"id":"743de7bc56899fd6e137e1205e33f8c32713e580","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-13 16:03:35.000000000","message":"Patch Set 5: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"287cea7a97fa5b71b00aab07a1133481f2367a24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 18:04:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/efcb38ad261f45d7bc5602e7503d6fc3 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af10edfacf0a452899e0273468d850b7 : SUCCESS in 6m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/963abbeae9c8496a90703fc5ed7d1c2e : SUCCESS in 8m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cdca7a77a16f411fb3abed9c7a961689 : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8eed19866eff4d3bbb0078b00a149a6d : SUCCESS in 5m 53s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/7e922e0c57824acabdda558d7ec135c8 : SUCCESS in 54m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/518aa93d32d74d4ea7855ba9b456b660 : SUCCESS in 59m 08s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/c490680b129c4d53b4a9433e79ff3dbd : SUCCESS in 1h 33m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/25e3e071dc6d48bfb772ff77739316cf : SUCCESS in 56m 45s","accounts_in_message":[],"_revision_number":5},{"id":"3b069c88b811c70d418b9b5f861b078e6b22ab5a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-02-16 06:29:16.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c21bd493fa5c2a0e1fa248f25c5e8fdc9c03f746","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 06:29:33.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4a8031bc138daab462934eea72697480ee027945","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 11:38:33.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29678469391848d1953b729ea04ad6e7 : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/905236c778ad4813874b5570f8f20080 : SUCCESS in 12m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9a18a752783341ff9b6ab2d428241fa0 : SUCCESS in 8m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f711d0636bf14a7aaa3d7aaf552888f3 : SUCCESS in 4m 47s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/d46297e6aacc4ab3a72b38caa34a79cf : SUCCESS in 58m 11s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/85447cb311dc44e6a8471d7869dbb267 : SUCCESS in 1h 07m 08s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/ad5fa92eb1b041358bd9548fdbaa005c : SUCCESS in 1h 42m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/266688a1f0df45cabd2a916683147196 : SUCCESS in 52m 08s","accounts_in_message":[],"_revision_number":5},{"id":"949a95c729ced38231c8799b530ab57cbcee64a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 11:38:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"36c442553e29181c69fcd730c3cb145e3822adb0","revisions":{"13c65d05ee188c3f955a0e0e2cb0803c3464e828":{"kind":"REWORK","_number":1,"created":"2020-01-14 15:28:43.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/63/702463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/702463/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/702463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/702463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/702463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/702463/1"}}},"commit":{"parents":[{"commit":"a50ddeafe4506ef4f85bc0d935525aaeb83121f1","subject":"Retire swauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a50ddeafe4506ef4f85bc0d935525aaeb83121f1"}]}],"author":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2020-01-14 01:08:11.000000000","tz":-480},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-01-14 15:28:34.000000000","tz":-360},"subject":"demarcate python-memcached requirements","message":"demarcate python-memcached requirements\n\nTo support Python3, we need python-memcached 1.58 or above. However,\nprojects such as keystonemiddleware[1] have python-memcached 1.56\nin its lower-constraints.txt. For stable/stein, Python3 is now the\ndefult[2]. However, we generally avoid changing lower constraints in\nthe stable branches so we don\u0027t inadvertently break the downstream\nconsumers. Therefore, we need to separate python-memcached requirement\nfor the respective python versions.\n\n[1] https://github.com/openstack/keystonemiddleware/blob/stable/stein/lower-constraints.txt#L61\n[2] https://governance.openstack.org/tc/goals/selected/stein/python3-first.html\n\nChange-Id: Ifb805954ff36420004d552a690046d8ddf3bed5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13c65d05ee188c3f955a0e0e2cb0803c3464e828"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13c65d05ee188c3f955a0e0e2cb0803c3464e828"}]},"branch":"refs/heads/stable/pike"},"a2ff7ae827c8a03708274e671a17f1a524858ac3":{"kind":"REWORK","_number":2,"created":"2020-01-18 06:48:23.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/63/702463/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/702463/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/702463/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/702463/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/702463/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/702463/2"}}},"commit":{"parents":[{"commit":"a50ddeafe4506ef4f85bc0d935525aaeb83121f1","subject":"Retire swauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a50ddeafe4506ef4f85bc0d935525aaeb83121f1"}]}],"author":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2020-01-14 01:08:11.000000000","tz":-480},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-01-18 06:48:18.000000000","tz":-360},"subject":"demarcate python-memcached requirements","message":"demarcate python-memcached requirements\n\nTo support Python3, we need python-memcached 1.58 or above. However,\nprojects such as keystonemiddleware[1] have python-memcached 1.56\nin its lower-constraints.txt. For stable/stein, Python3 is now the\ndefult[2]. However, we generally avoid changing lower constraints in\nthe stable branches so we don\u0027t inadvertently break the downstream\nconsumers. Therefore, we need to separate python-memcached requirement\nfor the respective python versions.\n\n[1] https://github.com/openstack/keystonemiddleware/blob/stable/stein/lower-constraints.txt#L61\n[2] https://governance.openstack.org/tc/goals/selected/stein/python3-first.html\n\nChange-Id: Ifb805954ff36420004d552a690046d8ddf3bed5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2ff7ae827c8a03708274e671a17f1a524858ac3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2ff7ae827c8a03708274e671a17f1a524858ac3"}]},"branch":"refs/heads/stable/pike"},"7d688564430b08cd14c8249cfedd2889351f29b2":{"kind":"REWORK","_number":3,"created":"2020-01-23 01:49:37.000000000","uploader":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"ref":"refs/changes/63/702463/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/702463/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/702463/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/702463/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/702463/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/702463/3"}}},"commit":{"parents":[{"commit":"a50ddeafe4506ef4f85bc0d935525aaeb83121f1","subject":"Retire swauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a50ddeafe4506ef4f85bc0d935525aaeb83121f1"}]}],"author":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2020-01-14 01:08:11.000000000","tz":-480},"committer":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2020-01-23 01:49:36.000000000","tz":-480},"subject":"demarcate python-memcached requirements","message":"demarcate python-memcached requirements\n\nTo support Python3, we need python-memcached 1.58 or above. However,\nprojects such as keystonemiddleware[1] have python-memcached 1.56\nin its lower-constraints.txt. For stable/stein, Python3 is now the\ndefult[2]. However, we generally avoid changing lower constraints in\nthe stable branches so we don\u0027t inadvertently break the downstream\nconsumers. Therefore, we need to separate python-memcached requirement\nfor the respective python versions.\n\n[1] https://github.com/openstack/keystonemiddleware/blob/stable/stein/lower-constraints.txt#L61\n[2] https://governance.openstack.org/tc/goals/selected/stein/python3-first.html\n\nNOTE: in addition to the original backport, this patch also cap\nthe gunicorn version to less than 20.0.0 as the later versions\nno longer supports Python2. Without this change, the\nrequirements-integration gate will fail.\n\nChange-Id: Ifb805954ff36420004d552a690046d8ddf3bed5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d688564430b08cd14c8249cfedd2889351f29b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d688564430b08cd14c8249cfedd2889351f29b2"}]},"branch":"refs/heads/stable/pike"},"e7028778274134283720bf7c4ea11b2ac6f14041":{"kind":"REWORK","_number":4,"created":"2020-01-24 23:11:16.000000000","uploader":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"ref":"refs/changes/63/702463/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/702463/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/702463/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/702463/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/702463/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/702463/4"}}},"commit":{"parents":[{"commit":"a50ddeafe4506ef4f85bc0d935525aaeb83121f1","subject":"Retire swauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a50ddeafe4506ef4f85bc0d935525aaeb83121f1"}]}],"author":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2020-01-14 01:08:11.000000000","tz":-480},"committer":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2020-01-24 23:08:30.000000000","tz":-480},"subject":"demarcate python-memcached requirements","message":"demarcate python-memcached requirements\n\nTo support Python3, we need python-memcached 1.58 or above. However,\nprojects such as keystonemiddleware[1] have python-memcached 1.56\nin its lower-constraints.txt. For stable/stein, Python3 is now the\ndefult[2]. However, we generally avoid changing lower constraints in\nthe stable branches so we don\u0027t inadvertently break the downstream\nconsumers. Therefore, we need to separate python-memcached requirement\nfor the respective python versions.\n\n[1] https://github.com/openstack/keystonemiddleware/blob/stable/stein/lower-constraints.txt#L61\n[2] https://governance.openstack.org/tc/goals/selected/stein/python3-first.html\n\nNOTE: in addition to the original backport, this patch also introduce the\nfollowing changes:\n\n1. cap gunicorn version to less than 20.0.0 as the later versions\nno longer support Python2.\n2. cap exabgp version to 4.1.0 as later versions have invalid\nmetadata file.\n3. cap libvirt-python to less than 6.0.0 as later versions no longer\nsupport Python2.\n\nChange-Id: Ifb805954ff36420004d552a690046d8ddf3bed5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7028778274134283720bf7c4ea11b2ac6f14041"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7028778274134283720bf7c4ea11b2ac6f14041"}]},"branch":"refs/heads/stable/pike"},"36c442553e29181c69fcd730c3cb145e3822adb0":{"kind":"REWORK","_number":5,"created":"2020-01-25 01:07:18.000000000","uploader":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"ref":"refs/changes/63/702463/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/702463/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/702463/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/702463/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/702463/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/702463/5"}}},"commit":{"parents":[{"commit":"a50ddeafe4506ef4f85bc0d935525aaeb83121f1","subject":"Retire swauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a50ddeafe4506ef4f85bc0d935525aaeb83121f1"}]}],"author":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2020-01-14 01:08:11.000000000","tz":-480},"committer":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2020-01-25 01:07:06.000000000","tz":-480},"subject":"demarcate python-memcached requirements","message":"demarcate python-memcached requirements\n\nTo support Python3, we need python-memcached 1.58 or above. However,\nprojects such as keystonemiddleware[1] have python-memcached 1.56\nin its lower-constraints.txt. For stable/stein, Python3 is now the\ndefult[2]. However, we generally avoid changing lower constraints in\nthe stable branches so we don\u0027t inadvertently break the downstream\nconsumers. Therefore, we need to separate python-memcached requirement\nfor the respective python versions.\n\n[1] https://github.com/openstack/keystonemiddleware/blob/stable/stein/lower-constraints.txt#L61\n[2] https://governance.openstack.org/tc/goals/selected/stein/python3-first.html\n\nNOTE: in addition to the original backport, this patch also introduce the\nfollowing changes:\n\n1. cap gunicorn version to less than 20.0.0 as the later versions\nno longer support Python2.\n2. cap exabgp version to 4.1.0 as later versions have invalid\nmetadata file.\n3. cap libvirt-python to less than 6.0.0 as later versions no longer\nsupport Python2.\n\nChange-Id: Ifb805954ff36420004d552a690046d8ddf3bed5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36c442553e29181c69fcd730c3cb145e3822adb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36c442553e29181c69fcd730c3cb145e3822adb0"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
