)]}'
{"id":"openstack%2Frequirements~702478","triplet_id":"openstack%2Frequirements~master~I0e7cbf1392247ab89c8d82c299980790f65d69bb","project":"openstack/requirements","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I0e7cbf1392247ab89c8d82c299980790f65d69bb","subject":"Remove requirements-tox-py27-check-uc job","status":"MERGED","created":"2020-01-14 16:40:34.000000000","updated":"2020-01-15 05:12:46.000000000","submitted":"2020-01-15 05:10:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702478-1579065009164-a981c286","meta_rev_id":"737f77e61e6e77f612cb3b442088c0f340701cc1","_number":702478,"virtual_id_number":702478,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-15 05:10:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-01-14 16:43:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-15 03:37:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-15 03:37:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-14 16:43:56.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-15 03:37:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-15 05:10:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d5ca85a67138145d58dbade516d5db18737c389","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-01-14 16:40:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df8b691b6a0cb769948ed522da6ca082b71fd5bb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-14 16:43:35.000000000","message":"Patch Set 1:\n\nI think we may need to keep the job definition around for swift maybe? Need to check if the job is used elsewhere.","accounts_in_message":[],"_revision_number":1},{"id":"03024d375dc9424e734c67e4404531583558b898","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-14 16:43:56.000000000","message":"Patch Set 1: Code-Review+2\n\nNope:\n\nhttp://codesearch.openstack.org/?q\u003drequirements-tox-py27-check-uc\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d","accounts_in_message":[],"_revision_number":1},{"id":"c2a085af0c9662a94e5c0a002a6f144a63a47ee9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-01-14 16:45:09.000000000","message":"Topic set to drop-py27-support","accounts_in_message":[],"_revision_number":1},{"id":"65879a23f65c5f85bb910f67759af8b120f513df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 01:21:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd653accf9a44323b07a79817b622f10 : SUCCESS in 5m 53s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/caf502485c0840f3a081483253abab83 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4278856652cb47e0aca53a171bfda50a : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98f7603c151f44d88b54e16b4175e3ec : SUCCESS in 6m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e7e78ad41e59471bad125d883e209470 : SUCCESS in 7m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50fd57ec13d742f591c24b7dad293ef2 : SUCCESS in 7m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df6a01207cc84260afe0167434688f41 : SUCCESS in 1h 03m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/40fd808a0eb3415e8623f3f9df5184fb : SUCCESS in 1h 16m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/18d88697927f43ee897a5b0723913651 : SUCCESS in 1h 44m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7b869f4c53ab402a9c75d024f2968d55 : SUCCESS in 1h 03m 29s","accounts_in_message":[],"_revision_number":1},{"id":"cbb488c1f901a002c3cdd0b4bbb7e55f6c3b8533","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-15 03:37:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n.doit","accounts_in_message":[],"_revision_number":1},{"id":"b6c838db68ef86458570dd2e04dd071c658edaca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 03:38:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f2553a22a3a6ae3b74c936dac7a511c4b0e97756","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 05:10:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"541ba07ee9228a8bb9de157bb6fbc3156b2431d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 05:10:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d776bcebd3744ac2b5a4d34d8178f8b6 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d59ab65564cd45b4b712b48d623d0cb0 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5c5332273af471482b0d1f01598f963 : SUCCESS in 6m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/601b25f5a4db4c268e7ce329e376e469 : SUCCESS in 7m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2aea796de4f4e7bb944cedf17e70c9f : SUCCESS in 4m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00ea7cd8c18d44b6a1a83049556e3d9a : SUCCESS in 58m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a4d001705bcf4b71a3b4bc036c0d04b4 : SUCCESS in 1h 06m 31s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9bac41a1d1754e4f96deb31b14644685 : SUCCESS in 1h 25m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f9a3d7ac42d84889b77f5c2913f1724c : SUCCESS in 56m 02s","accounts_in_message":[],"_revision_number":1},{"id":"737f77e61e6e77f612cb3b442088c0f340701cc1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 05:12:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82a62408d3794fb7973694b392f10604 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cbb15198c0a374d3feef9a79dba6f8c7435bcaaa","revisions":{"cbb15198c0a374d3feef9a79dba6f8c7435bcaaa":{"kind":"REWORK","_number":1,"created":"2020-01-14 16:40:34.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/78/702478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/702478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/702478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/702478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/702478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/702478/1"}}},"commit":{"parents":[{"commit":"c55121348430c2fa33cf2285d96a8f9836ce9fec","subject":"Merge \"update constraint for os-client-config to new release 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c55121348430c2fa33cf2285d96a8f9836ce9fec"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-01-14 16:37:21.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-01-14 16:37:21.000000000","tz":60},"subject":"Remove requirements-tox-py27-check-uc job","message":"Remove requirements-tox-py27-check-uc job\n\nAs neutron-lib became python 3 only, the py27 upper constraint check\njob fails. Let\u0027s get rid of it as it is not needed anymore.\n\nChange-Id: I0e7cbf1392247ab89c8d82c299980790f65d69bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cbb15198c0a374d3feef9a79dba6f8c7435bcaaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cbb15198c0a374d3feef9a79dba6f8c7435bcaaa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
