)]}'
{"id":"openstack%2Frequirements~702483","triplet_id":"openstack%2Frequirements~master~I93ddb63ee40c2711419dd56e8828cf631ba6012b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I93ddb63ee40c2711419dd56e8828cf631ba6012b","subject":"update constraint for osc-lib to new release 2.0.0","status":"MERGED","created":"2020-01-14 17:07:44.000000000","updated":"2020-01-16 00:59:52.000000000","submitted":"2020-01-16 00:58:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702483-1579136289378-f3637170","meta_rev_id":"821db250c24e630c11a1cc80d973cb2686d4e8f5","_number":702483,"virtual_id_number":702483,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-16 00:58:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-15 04:40:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-01-15 20:44:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-14 17:12:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2020-01-15 01:14:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-01-15 20:44:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-14 17:12:23.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-01-15 01:14:24.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-01-15 04:40:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-15 20:44:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-16 00:58:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87f9f6e5d22cc6954e52c6d30c97eaa70a00161f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-01-14 17:07:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c43beae6e2c60cf88953bd432c01af19549f377","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-14 17:12:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"64151a4b03159dd20cfc8aad6a39943ce439cba1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-01-15 01:14:24.000000000","message":"Patch Set 1: Code-Review+1\n\nYep, LGTM","accounts_in_message":[],"_revision_number":1},{"id":"633530572b08ef531202107e692f6ae8e0582d5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 03:41:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2072cc1aa454e53b66487fdd3e776f8 : SUCCESS in 4m 14s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/0861bad7ce0c4433ae3ed78c716e7452 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf310f3003514b32af344f847483c6a8 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ec4441e94a04ca097badf615b00d20d : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf93b154d889427295ae279a25e05695 : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a7c893691aa44a16865467fd8bd22085 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/65b18b9c0cd64af386904e7128f7a0ef : SUCCESS in 4m 47s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0ced7736fc7741c48d17abd2053227d1 : SUCCESS in 8m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ea271e91bbf0445cbd6c2808578990c2 : SUCCESS in 8m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fcb13ecc731a4f64bfd179e6d252e703 : SUCCESS in 9m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ce809c3a40ad47918467ae2ae40b741a : SUCCESS in 5m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4186ad22967e426fab1f964f151dc440 : SUCCESS in 12m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/10484465fdce407ea4bbe8ab7cc8c238 : SUCCESS in 7m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/956639e71e33492c9d5e37373c325f7d : SUCCESS in 20m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/99dc27d5667146c3b5a1e69626485f0c : SUCCESS in 27m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f8b5b1bcc64b411187c6ab235d612e72 : SUCCESS in 17m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6bf3f9b8f65a4176a18df6e6a50308d3 : SUCCESS in 19m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f240526d579a4d16891366bcdab48e27 : SUCCESS in 24m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/63785379f69a488f8b265ec35670c081 : SUCCESS in 7m 17s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d9a0db923c0a4ff1993c9bf46d3f82dc : SUCCESS in 9m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b6990679d82e45539f32281df0d45ffb : SUCCESS in 6m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/917d6d1c9c9e42f088eb4ed7599f6316 : SUCCESS in 9m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0e225bba2ea14074a2515e349566a3c8 : SUCCESS in 10m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/147c4e6271f64de4a2c2d0097ace72be : SUCCESS in 1h 17m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3bcf93559cb1411ea28b0562341844b3 : SUCCESS in 1h 02m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5be48c18383f497ab7c3932accc2dc52 : SUCCESS in 1h 55m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a4945a1881084562b5e9eba33820e958 : SUCCESS in 54m 52s","accounts_in_message":[],"_revision_number":1},{"id":"05f5b5a5aaca18570eb280ed0d9baa4bf11f94c4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-15 04:40:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ea2dad9609eab19a3f3831da45a4ed7114869826","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-15 20:44:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"13866f370afad1f5fda181a38ef286092883e9ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 20:45:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4aa48c218348e06d2c0ebfabef89275f54e8e17b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 00:58:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e7fb65197fb535416a8a469c0eb42baa272242c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 00:58:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d32d948da574169ab20bd5349a808b1 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1c50d55900b45c2b1e21ee9a9d6a88d : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f318f81011e4411981a8c9a802154bc6 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9118b9b9039e4cb784690f4310736a1b : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7b0b8b7123dc42729341c64037ab9498 : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ae1e4d0a6b664b569f4ffc25aa750b62 : SUCCESS in 5m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/58fe393ae26f4dbbb4af8053e751bf38 : SUCCESS in 7m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a727b52721804f7d9bdda585063dc10c : SUCCESS in 9m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bfd815901ca74771b4ee4bcb0ab2dbd2 : SUCCESS in 7m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/262c88163f434d58b908b7f3446babbe : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c8fe541313674449bc3ecc4b3175eb2d : SUCCESS in 7m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2c8ce7fd53b747809d106548c41d9d71 : SUCCESS in 14m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/84f52250255c42f2b8ec4d56d20ede2d : SUCCESS in 30m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/947e67f741d9468bb8cc2c28ddf42241 : SUCCESS in 12m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/42589ee7d9e1494e9d3a5d65b45fe12d : SUCCESS in 19m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/72396c35e44a46a18aa8cb397e752926 : SUCCESS in 18m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1c17b21a275a4bdcbefbd725a208113a : SUCCESS in 5m 40s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/580953d0bf1d4bd8be72bfd99fa1587a : SUCCESS in 7m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0d23954507f64423b74613168c242ee9 : SUCCESS in 11m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5fdb07f3a6a84dab9da8dfb1aff90e61 : SUCCESS in 12m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/be53c34ff2e847739d33a4cdf894177e : SUCCESS in 9m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e53dafc077c54e88bdc5984b1419b24c : SUCCESS in 1h 23m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bda210af4cf54bd4bc70d858c9d4462b : SUCCESS in 1h 20m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dac49f66f98e4ab8a71dd57c186f5e44 : SUCCESS in 1h 23m 26s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/da02087dadbb4b01b0288f873320a340 : SUCCESS in 1h 05m 39s","accounts_in_message":[],"_revision_number":1},{"id":"452acbd368e933b6b4035616005a028c89845d48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 00:59:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d852148891b241a0a017ac044be87520 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"457e4784c57a43e4286ed0609921a16ff70fc4e6","revisions":{"457e4784c57a43e4286ed0609921a16ff70fc4e6":{"kind":"REWORK","_number":1,"created":"2020-01-14 17:07:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/702483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/702483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/702483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/702483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/702483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/702483/1"}}},"commit":{"parents":[{"commit":"c55121348430c2fa33cf2285d96a8f9836ce9fec","subject":"Merge \"update constraint for os-client-config to new release 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c55121348430c2fa33cf2285d96a8f9836ce9fec"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-14 17:07:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-01-14 17:07:43.000000000","tz":0},"subject":"update constraint for osc-lib to new release 2.0.0","message":"update constraint for osc-lib to new release 2.0.0\n\nChange-Id: I93ddb63ee40c2711419dd56e8828cf631ba6012b\nmeta:version: 2.0.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ifd777ec06e7fa3e4a3247d66ce3c2f848ca8ba9b\nmeta:release:Code-Review+1: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Code-Review+1: Dean Troyer \u003cdtroyer@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/457e4784c57a43e4286ed0609921a16ff70fc4e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/457e4784c57a43e4286ed0609921a16ff70fc4e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
