)]}'
{"id":"openstack%2Frequirements~702773","triplet_id":"openstack%2Frequirements~master~I8f76b6b486993637228e6893c792cfa6ffada3bd","project":"openstack/requirements","branch":"master","topic":"trusty-removal","hashtags":[],"change_id":"I8f76b6b486993637228e6893c792cfa6ffada3bd","subject":"Remove trusty wheel build jobs","status":"MERGED","created":"2020-01-15 23:16:33.000000000","updated":"2020-01-16 16:25:06.000000000","submitted":"2020-01-16 16:20:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702773-1579191641149-d94f54cc","meta_rev_id":"97ba9094243e3692d99dc5d951ae2bde6a395fc4","_number":702773,"virtual_id_number":702773,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-16 16:20:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-01-16 13:03:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-15 23:31:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2020-01-16 07:21:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-01-16 13:03:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-15 23:31:56.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-01-16 07:21:50.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-01-16 13:03:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-16 16:20:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21d6bdcc7cf4f26bfd7b77362c3d1e44b8ab4ae7","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-01-15 23:16:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"819dc660e32a098a7b29d48dc17e11fc474f1319","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-01-15 23:31:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d11a19a30ed2b143d59373716ee5cc2d0f8a3d66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 00:41:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/105d342950444af6b1284e134e64c614 : SUCCESS in 3m 48s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/756941533ea74989b635fe6fc2e23c83 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76a82fdce7de4b2890811fa0047c5e39 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee5932ed47424c03a88f93625c445fbe : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/968db408a7994825a5be510b263a015c : SUCCESS in 5m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/011f767e959a4a168b6d3c21bbe2c4bb : SUCCESS in 4m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b305e59451440a1a391b81d6e6a0887 : SUCCESS in 1h 09m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c47eba4c448451ba3b9144b5348a2e7 : SUCCESS in 1h 18m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c7eae16434a04477acc9bebc1f97fbd4 : FAILURE in 1h 19m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d9b00c1e47fa4f519aaaba6c5d034340 : SUCCESS in 1h 09m 56s","accounts_in_message":[],"_revision_number":1},{"id":"65ffb9df868b841b7a799859b8a82dba19e3a59a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-16 07:21:50.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a7245fee0b3fbc0318dd2cf401e4070e6bd0cb28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 09:06:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a3e0572119d45d1941b2c088eb5db23 : SUCCESS in 4m 55s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/d9e1591149bf4734b6168c729ccbd54a : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4993a7816d214ac8ad75d3580ae0a10b : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08375f0327e84fe2a82668a8d83f8137 : SUCCESS in 6m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26ad63bdfabf4ff6959544ea39c3c4a9 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fbb9faa4101a4e3294d7d17bd84b29e9 : SUCCESS in 5m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e0683f095fe42dd960434ecd0ae53ce : SUCCESS in 55m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4a3e8e0f60a43388f5fcefa8e4b8a95 : FAILURE in 1h 39m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/24911d66a2734a2490d98e1d5d6c8d09 : SUCCESS in 1h 30m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/93e0f361401e48d9a8770f72e8e2f5f4 : SUCCESS in 1h 00m 57s","accounts_in_message":[],"_revision_number":1},{"id":"00db4fab62f6c6fcba21f5b163fe9b985ebcd480","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-16 13:03:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fc754b8e9118749ac72d848ffce11394aa5ccabf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 14:48:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2646814fa57d444bbd685d778105d8ce : SUCCESS in 4m 04s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/632b4b55a05943e5800c6a152f3082c2 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c074b84e2b646cbbc0f71d90f8a1308 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d3aa2503cf048aab4bae36887acd4f6 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6bb87e1ccabf4fc6b2c23ba41ad47729 : SUCCESS in 4m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3d39987c6dfd4ace855b0564b40cee73 : SUCCESS in 3m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/510a6cd784f541538ef1c3d394df70c7 : SUCCESS in 53m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/93878b64169b46dc9acd6471d7f68d27 : SUCCESS in 1h 17m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/da23b8c954c94e028bdbb645b883399e : SUCCESS in 1h 42m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5974ebc221f949c3b204a3c07d430f1a : SUCCESS in 59m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b1aa1fd6d1eadee6ccdc9b3b5152d257621d17d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 14:49:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b4dad2ba69b3aaf8893b4bcf49a166ad142cedb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 16:20:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/984451e0198945da9d3d4ff8b2337dfe : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c15bb118468648de898cf7ce16a1b0ad : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4b2140a53eb4eab8a5961dafbe5c2b4 : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71cda3f3457f4887ab0bc81c5071bdc3 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2b81ea9da4834df5964c40ac7afa4fc5 : SUCCESS in 6m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e20e6d9858b8423aa557e4d71f240752 : SUCCESS in 1h 03m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/95aa2759ec2d4e5a850defd55b76b155 : SUCCESS in 1h 09m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ad1ed714ff0c4df094cec7af3481ff58 : SUCCESS in 1h 26m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/dc4d4182a48e41b2aceff96d574d5c86 : SUCCESS in 1h 00m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b77d9fbf224f0f58d927925fe28d93e90ef8e202","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 16:20:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"156eacbd5caeeb64ee3b4dfc140523e02c61039e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-16 16:23:36.000000000","message":"Patch Set 1:\n\nthis needs backporting to all active branches...","accounts_in_message":[],"_revision_number":1},{"id":"7f2962a73d12b27af2a75f0a8845b6253c70b7dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 16:25:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fb6e401a68046fa8b54b3513f3ae183 : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ba28d145c535fc1ce6efb18e45da9269cf9138a","revisions":{"5ba28d145c535fc1ce6efb18e45da9269cf9138a":{"kind":"REWORK","_number":1,"created":"2020-01-15 23:16:33.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/73/702773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/702773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/702773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/702773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/702773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/702773/1"}}},"commit":{"parents":[{"commit":"00993edc039414571aed6cfb08c87ef3adaac2b6","subject":"Merge \"update constraint for neutron-lib to new release 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00993edc039414571aed6cfb08c87ef3adaac2b6"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-15 23:13:44.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-15 23:13:44.000000000","tz":660},"subject":"Remove trusty wheel build jobs","message":"Remove trusty wheel build jobs\n\nIn prepartion for Trusty node removal, stop testing wheel builds in\nrequirements.\n\nChange-Id: I8f76b6b486993637228e6893c792cfa6ffada3bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ba28d145c535fc1ce6efb18e45da9269cf9138a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ba28d145c535fc1ce6efb18e45da9269cf9138a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
