)]}'
{"id":"openstack%2Frequirements~702874","triplet_id":"openstack%2Frequirements~master~I706e73505666e27dc409182db317f05284e6fe85","project":"openstack/requirements","branch":"master","topic":"fix_warning_config_parser","hashtags":[],"change_id":"I706e73505666e27dc409182db317f05284e6fe85","subject":"Fix the warning to read the configuration file.","status":"MERGED","created":"2020-01-16 14:28:34.000000000","updated":"2020-01-16 18:31:29.000000000","submitted":"2020-01-16 18:29:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702874-1579199353053-5f1f21a0","meta_rev_id":"b78cba1391f56ac1dcbbec941d442967c88bd7b6","_number":702874,"virtual_id_number":702874,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-16 18:29:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-01-16 14:41:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-16 16:43:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-01-16 17:18:54.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-16 16:43:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-16 14:41:37.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-16 16:43:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-16 17:18:54.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-01-16 18:29:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"afe673b808dc878eaac7357d1f6a2a166b924a81","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-01-16 14:28:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"645658ca30c1cff5b62f26ca0b6939da6a367cc4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-16 14:41:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"988ce65385190343445333a8631168fce9f04e6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 16:36:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0db29a329404f49a85bd87fcca7685c : SUCCESS in 4m 14s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/cfcb1d02b2684adab5566653dcbf730d : FAILURE in 4m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad091b9a42fb45aeaa29559f52e50727 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/615eacf8b4074518a2f4e4364037e115 : SUCCESS in 7m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ebad60f3ae0e4ab5a1d8c85dfd298eaf : SUCCESS in 5m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85f14084eecc45e1a8dcee6ecac92fea : SUCCESS in 4m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6db31096e68145cc801d86d2c512aa36 : SUCCESS in 54m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7bc5e5371fb6414b8adca711f9adf4da : SUCCESS in 1h 13m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e04d1ec094fd4007bc3b6e2d28c67887 : SUCCESS in 1h 38m 57s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f9c4276f3dc04fbcb6e594fcb95b7cc3 : SUCCESS in 1h 05m 56s","accounts_in_message":[],"_revision_number":1},{"id":"ff01611b44f92edf1adfc120378702544141c1c6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-16 16:43:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0f8898f5c6ab3f0ecfe85e3bd533e3f4c2fca45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 16:44:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1464f844fa3b7f881f5a75328cd0bb59ab7fc047","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-16 17:18:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"93aad8755dafd183405a126b102b14c9e70800a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 18:29:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd95e8e82a7e447a9e22f5e8aaeebf50 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c4ca5e6818d466cb27b5271f889ee8c : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/208150d1752b4c02898f79240b7eda2d : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a677be00354a40baa02473bf4d29d80f : SUCCESS in 5m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/00b0b507455e4e8a9748783f01407baf : SUCCESS in 6m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94d264c08a9549fcb78ff57d1053e964 : SUCCESS in 1h 04m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d73d8bac16c44acaa08582f79e2c1e92 : SUCCESS in 1h 03m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6544059bcb8245289b752e5393e9930c : SUCCESS in 1h 30m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/976c38f75b3d41a4aa26b0adcf52b2e8 : SUCCESS in 1h 03m 50s","accounts_in_message":[],"_revision_number":1},{"id":"667c2afe7c58d3f87990e579be468b708660710a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 18:29:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"18437c37764902eca97759eca901d40ac391c08e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 18:31:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87dd03d4de804b469722cdc1cab35e6a : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4da6e47b8f7e2912de062d9e70f4fb3c06090b32","revisions":{"4da6e47b8f7e2912de062d9e70f4fb3c06090b32":{"kind":"REWORK","_number":1,"created":"2020-01-16 14:28:34.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/74/702874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/702874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/702874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/702874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/702874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/702874/1"}}},"commit":{"parents":[{"commit":"47a3b857cb7d0244025683a009e899cca964a989","subject":"Merge \"update constraint for os-traits to new release 2.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47a3b857cb7d0244025683a009e899cca964a989"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-01-16 14:28:22.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2020-01-16 14:28:22.000000000","tz":60},"subject":"Fix the warning to read the configuration file.","message":"Fix the warning to read the configuration file.\n\nThe method readfp is deprecated and will be removed in future versions.\nUse read_file method instead.\n\nChange-Id: I706e73505666e27dc409182db317f05284e6fe85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4da6e47b8f7e2912de062d9e70f4fb3c06090b32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4da6e47b8f7e2912de062d9e70f4fb3c06090b32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
