)]}'
{"id":"openstack%2Frequirements~702893","triplet_id":"openstack%2Frequirements~master~Ia5b6d73817fe531c4f660c8afff4de3ec8cecf7f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia5b6d73817fe531c4f660c8afff4de3ec8cecf7f","subject":"Raise python-rsdclient upper constraint to 1.0.2","status":"MERGED","created":"2020-01-16 15:25:50.000000000","updated":"2020-01-16 22:02:52.000000000","submitted":"2020-01-16 22:01:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702893-1579212062838-fe3cfc39","meta_rev_id":"0ed73023bbbd2d4c0e8f4fc7509741a8b566d07d","_number":702893,"virtual_id_number":702893,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-16 22:01:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-16 20:24:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-01-16 16:38:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-01-16 20:24:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-16 16:38:04.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-01-16 20:24:38.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-16 22:01:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b064464bab895ff75b701a83c1f11cc141e2e61f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-16 15:25:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0315893a85a5957ce393c6e988dab7a2416c3956","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-16 16:38:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b36b5681fd234a8914b0515a033afd3dfa80e796","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 17:13:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36ac6efaf7b34f0bb7588752b3576a87 : SUCCESS in 4m 27s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/14a24ba0b2514a0d90986e28f5457bf9 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b321e810b484590b0786bfaef182166 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29a62290aff3439fb64ccfc2ad02bfaf : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f14a4deecabb4083b20d8a290eb6eff0 : SUCCESS in 6m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7815fda26664b50bed94df0f6e07edf : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9e58c08e9a4b467ab143a844a66237ae : SUCCESS in 5m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4bc83b36a5ed450ab14e72d0731f17df : SUCCESS in 7m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f37c0adc41c744d383431cb19b97e55b : SUCCESS in 10m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a2f222bda20f43ee8b3a6717d121c34e : SUCCESS in 5m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/93e4942d76dd4f5cb2dd793b8e5b49eb : SUCCESS in 9m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7bbdd8a7137d4b1793962e43f189a8d4 : SUCCESS in 8m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/55e32507bdcd4cdb903e210bcd5a4161 : SUCCESS in 13m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dd384cfb8a8c4a89bdfb473e133ce334 : SUCCESS in 25m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3906d17a31024c3db410772722cc9d5f : SUCCESS in 14m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/83b2f8b61fdd4befb644a5a74ab5050b : SUCCESS in 16m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f3a211e17ef3477d86503d14dd8e4d2f : SUCCESS in 22m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9cc10966ec464bcca284e02420043fe4 : SUCCESS in 6m 46s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/173ac63162374778ac768596a9dbab9e : SUCCESS in 7m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e1206b87efaf421789624f848f4891f7 : SUCCESS in 8m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0243bc416c514630b30c2cff1b04b66d : SUCCESS in 11m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5343c72ef083468b83362e17e210c8f2 : SUCCESS in 14m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fd308a653fe41608885189fc762a005 : SUCCESS in 1h 08m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/81bbde1bd4f44352bf03dc73aac990c1 : SUCCESS in 1h 20m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/33f9dc650d7249ffa0986ea8c62fa8cc : SUCCESS in 1h 41m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1337883e13734ae5857e6f244bc905c2 : SUCCESS in 55m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5f8a0d50ab88c3e90352342b857663824e4b40c8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-16 20:24:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cb764eb7833da71e9c5a258db236debba14ca6ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 20:24:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"03d493688673b2853148b94b0e275bdb4a57328f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 22:01:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c019d4f8580448eeb681eea44973a073 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8dd50b9dae94a8c9b0db1ccb7c536a3 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d00b3aa756f486cae52afd442ad2c37 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13eb8d27f37d4338ad599ac062379639 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca3cbfa5327b462a837d059f781f0b8e : SUCCESS in 5m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/44b8dad5c6f642bf88dde9be4aac5912 : SUCCESS in 4m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/47450f00771b42fbae4f6cdc759c0271 : SUCCESS in 8m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8ba0a04cb4df469c866b60c73c78f7dd : SUCCESS in 8m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4fc3e8173b1949a080ec53671fca379e : SUCCESS in 4m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a2ccd960cfa946eb93d5992a040736f7 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/70c91af582f24a12807af61358b221d0 : SUCCESS in 6m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/092662dd0e46433ea669c4a2cafc5755 : SUCCESS in 14m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1afcfa05346543bbb2253acf8d4151dd : SUCCESS in 25m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0cdaaa62d64c429a9fcb6f145bc30d82 : SUCCESS in 12m 24s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b7407470e5e144fda61f5cc287a94b97 : SUCCESS in 13m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9e7870d0ac1a4030ab9a10f99540e1f2 : SUCCESS in 16m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/08b18deaa6ce4d5fbe325c033f31e8f3 : SUCCESS in 4m 19s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/79e28984fb1c4c18b2356abbd5b8886a : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad1cf4aefb3941a5ad8cee5c5c6c886a : SUCCESS in 6m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c4fee128aed8412f84778083d4c3a5da : SUCCESS in 9m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e375f9c95e59479b919c7303cb6f8c0b : SUCCESS in 9m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb04eb8faa4c4612848eb1b6e050b244 : SUCCESS in 58m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a75ec3dc6d774d3184578c12f65bf0c5 : SUCCESS in 1h 09m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b32cc3a6a1404e17a44cbb13a20581d8 : SUCCESS in 1h 29m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8d4dd9cb751c4c65b094c0266563c1b9 : SUCCESS in 56m 00s","accounts_in_message":[],"_revision_number":1},{"id":"2ae01f4b0a255e45c7f4bc1bb331b1d7264c4aa4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 22:01:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fe13495eadca31829470287bd182c3fb153f9262","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 22:02:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91837b6b7fef4307ab35f2361020566c : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"95cda74696f9ee443361dcf72df1df576e4e636f","revisions":{"95cda74696f9ee443361dcf72df1df576e4e636f":{"kind":"REWORK","_number":1,"created":"2020-01-16 15:25:50.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/93/702893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/702893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/702893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/702893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/702893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/702893/1"}}},"commit":{"parents":[{"commit":"1a6430f8785b7f7d1f9b6fc7beb6694727c00a73","subject":"Merge \"update constraint for neutron-lib to new release 1.31.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a6430f8785b7f7d1f9b6fc7beb6694727c00a73"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-01-16 15:24:54.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-01-16 15:25:42.000000000","tz":-360},"subject":"Raise python-rsdclient upper constraint to 1.0.2","message":"Raise python-rsdclient upper constraint to 1.0.2\n\nThis contains an important fix to not import a private module from cliff\nthat has been changed in the latest release.\n\nChange-Id: Ia5b6d73817fe531c4f660c8afff4de3ec8cecf7f\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/95cda74696f9ee443361dcf72df1df576e4e636f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/95cda74696f9ee443361dcf72df1df576e4e636f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
