)]}'
{"id":"openstack%2Frequirements~702915","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I8f76b6b486993637228e6893c792cfa6ffada3bd","project":"openstack/requirements","branch":"stable/queens","topic":"trusty-removal","hashtags":[],"change_id":"I8f76b6b486993637228e6893c792cfa6ffada3bd","subject":"Remove trusty wheel build jobs","status":"MERGED","created":"2020-01-16 16:26:26.000000000","updated":"2020-01-18 01:49:58.000000000","submitted":"2020-01-18 01:49:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702915-1579312198246-a981c286","meta_rev_id":"ec86f4ef2c09161338b94800e28c8b87fb6f2fe5","_number":702915,"virtual_id_number":702915,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-18 01:49:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-01-17 15:57:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-01-17 15:38:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-01-17 15:57:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-16 16:26:26.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-01-17 15:38:47.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2020-01-17 15:57:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-01-18 01:49:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"30b447d5708fe67c76bb7cca6898fe818a1fbc7a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-16 16:26:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c40c53e2652fd466bc0a19aac525aa3dc52fae81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 19:12:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/ccc89e0e1bbf428a845d6102e67f77e8 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec1f1b2f548041259398103053882697 : FAILURE in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9b44ab1f3cec43a2a37e88567579090e : SUCCESS in 4m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2a7af1552bfc4192b3b6f3cabe31146c : SUCCESS in 5m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c70cab24795645b2854c1f452054e15c : SUCCESS in 5m 23s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/8e8af00da58042d7b33688a5e85274a7 : FAILURE in 30m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a3945ecf3514fd3af75f4c29fb698dd : SUCCESS in 1h 14m 08s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/73e469059012477d90cbe88f9feb70a4 : SUCCESS in 1h 30m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/20302eac56a948cfb01e6fa43eda5a61 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":1},{"id":"1acdb7f60d0df2ad855f2bb579a8d017340a965c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-16 19:49:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d096147cbb571d228ef9a904829cdd162b724a2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 21:16:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/5273d5762f3b4af585599703d6a80383 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44b51d8f64b54843be236d640aeba3ac : FAILURE in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7cc34e4ba2e94ad8bbb05e70409283bc : SUCCESS in 5m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/46e96ad0dfe8466da9ddca8be82c1dee : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c4260e7f54d3444784bb028d0ea41cca : SUCCESS in 5m 38s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/0953bf21627748d6980b3978ab4af59d : FAILURE in 31m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d2f940d213a84f1a951d26c986d4064d : SUCCESS in 59m 58s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/6f125f74ba4d4214995ccc8c1176ab83 : SUCCESS in 1h 22m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0985646194504e92a8090ced8d002cff : SUCCESS in 52m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9385f25e1251bd5ea936086a9ac1f182c0751e2d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-17 01:16:18.000000000","message":"Patch Set 1:\n\nrecheck, probably legit","accounts_in_message":[],"_revision_number":1},{"id":"0cb14ea69b26cc929e8d80e3dea0694c851bf830","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 02:41:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/72f30c76d7914ed5977ed867638f01c3 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4b34eef6d094479810a4fc4a6fd8242 : FAILURE in 3m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6f6394bc3304ef4bafc252d9bf58f3d : SUCCESS in 3m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d62793e24e924955a0ba183a8d4e33fa : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/05cb553cf8aa490fa48c2f3d05b809be : SUCCESS in 3m 19s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/1538a43b6f194a0ab1bec66a7820d0bc : FAILURE in 31m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/beae7a4fbab048399439a492d19a7912 : SUCCESS in 1h 05m 46s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/562a2461bd9642eabbca937af20f9100 : SUCCESS in 1h 19m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e812a373844443ec9bc400a826428362 : SUCCESS in 52m 11s","accounts_in_message":[],"_revision_number":1},{"id":"75a3d4c27941e3915c05d82ae0eaa045091f2eac","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-17 07:52:21.000000000","message":"Patch Set 1:\n\nFixes for testsuite are in https://review.opendev.org/703023","accounts_in_message":[],"_revision_number":1},{"id":"29340fd922562ce9d8082beba4c9802cc094ae76","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-17 09:32:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4033253d449e809e57fdd73f18cc334ffadb52bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 12:22:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/85290fe5e913440ebd7412c69a75e2bf : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65104441ebe44be7b0faa151c0fdf027 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e01372b05de457f8dc655182134aab1 : SUCCESS in 6m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/04085c62d79946ac8900480f941b0c6c : SUCCESS in 6m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/59ca2757842044c8bb9f70fe7dfe15fe : SUCCESS in 4m 45s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/8703ed67f1a9453da5439b972fb41a1e : SUCCESS in 1h 26m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cc91aa58ce7a49bfa2c0ce2d2932140c : SUCCESS in 1h 26m 14s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/cf9d21d6d2094087a49d0af85f6c8614 : SUCCESS in 1h 28m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/332c19215f954434a92938b42afd8c09 : SUCCESS in 58m 24s","accounts_in_message":[],"_revision_number":2},{"id":"c90d7d34ee017248c322044b17039b781c138421","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2020-01-17 15:38:47.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks good to me.\n\nThanks","accounts_in_message":[],"_revision_number":2},{"id":"8b8301f01881f75988a5e25089618e2ffe42639e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-01-17 15:57:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3a5a8c3b278ef336692c272a5b4ee9cefdfc084e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 15:59:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"34224a482187d801b56c1c2780403360a10c55a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 20:09:15.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6e16421c5787462831300395fd507e4d84f2a543","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 01:49:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c2ba234651cd599495a82f8699b9775b4bb69cac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 01:49:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51b5bc23f52d4d3f840d64804b74e313 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0b6cc02d7bbb4202bb8199b5882e25c6 : SUCCESS in 5m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c2b954a242864ed28fa7cb8233f78369 : SUCCESS in 5m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5700f1bfc11c40fa927203d58cd2a8bc : SUCCESS in 4m 22s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/9b923a1f979a40c5a532a484ba2b8ead : SUCCESS in 55m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9fede6971e04594a6e6f701a20d3439 : SUCCESS in 1h 05m 39s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/28ce945fd71a4afbb831b868ed7ec51e : SUCCESS in 1h 18m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/20b9d6dbb77e4f54bf40588d652f1832 : SUCCESS in 1h 02m 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f7b45b47bfc5f9b635eb5fa6613482c8c253b091","revisions":{"afa32fc7aa07343000b5a381ac61020b436bc369":{"kind":"REWORK","_number":1,"created":"2020-01-16 16:26:26.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/15/702915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/702915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/702915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/702915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/702915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/702915/1"}}},"commit":{"parents":[{"commit":"8811182c31cea08757ac70d8491fbb632c05dac8","subject":"Merge \"update constraint for ldappool to new release 2.2.1\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8811182c31cea08757ac70d8491fbb632c05dac8"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-15 23:13:44.000000000","tz":660},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-16 16:26:09.000000000","tz":60},"subject":"Remove trusty wheel build jobs","message":"Remove trusty wheel build jobs\n\nIn prepartion for Trusty node removal, stop testing wheel builds in\nrequirements.\n\nChange-Id: I8f76b6b486993637228e6893c792cfa6ffada3bd\n(cherry picked from commit 5ba28d145c535fc1ce6efb18e45da9269cf9138a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/afa32fc7aa07343000b5a381ac61020b436bc369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/afa32fc7aa07343000b5a381ac61020b436bc369"}]},"branch":"refs/heads/stable/queens"},"f7b45b47bfc5f9b635eb5fa6613482c8c253b091":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-17 09:32:48.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/15/702915/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/702915/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/702915/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/702915/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/702915/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/702915/2"}}},"commit":{"parents":[{"commit":"bb70fa451f366db180202cb6a778d6208713a2cc","subject":"Fix testsuite: pep8 and requirements-integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb70fa451f366db180202cb6a778d6208713a2cc"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-01-15 23:13:44.000000000","tz":660},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-01-17 09:32:48.000000000","tz":0},"subject":"Remove trusty wheel build jobs","message":"Remove trusty wheel build jobs\n\nIn prepartion for Trusty node removal, stop testing wheel builds in\nrequirements.\n\nChange-Id: I8f76b6b486993637228e6893c792cfa6ffada3bd\n(cherry picked from commit 5ba28d145c535fc1ce6efb18e45da9269cf9138a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7b45b47bfc5f9b635eb5fa6613482c8c253b091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7b45b47bfc5f9b635eb5fa6613482c8c253b091"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
